)]}'
{"id":"openstack%2Foctavia~776909","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I736816247131715f5c385b4680614ec3218a2ad7","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I736816247131715f5c385b4680614ec3218a2ad7","subject":"Fix pools going into ERROR when updating the pool","status":"MERGED","created":"2021-02-22 08:48:08.000000000","updated":"2021-03-05 13:06:47.000000000","submitted":"2021-03-05 13:06:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776909","meta_rev_id":"c0fdb10156a40d0604cbab7e4b76b43d5384ee9e","_number":776909,"virtual_id_number":776909,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-05 09:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-04 00:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-05 09:26:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 08:48:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-22 11:19:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 09:26:43.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"534eca4625f710da871b6d0f04c291fa578f2922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:48:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6babb6a5ce1d1a5ff88b2368af398a1d67b66bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 11:19:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad7a94afafa84b318e8443c8ca0fe61e : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/282f4e57fc774ed28eee12c8fbd1bd9c : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04346ef293444253b73100df3c1dea0d : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34631e5398754305a1c84211581679f0 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/149469685b804eaf89cb6fb0b97e8f5f : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cdbd93e63de43f2b1e9be3a848d7bf0 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357c7d6066f54c6d9b7dd38afb1b987f : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38515d32f4e4441d8833a7d689e92128 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1802738110a04d3380250e2d89118ca8 : SUCCESS in 8m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/628e58352dc241f3a86a8d4081efc0c9 : SUCCESS in 10m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c8cf228dc1234f8dbbd89d75c850a90c : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d73edcfd877541db913626c0f897be7b : SUCCESS in 12m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5cd9ffd404b548098e0807fd26966d0d : SUCCESS in 1h 10m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ce1aaec90f8f45b6b319d9dd05502fc6 : SUCCESS in 1h 19m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11cbf4cd14f94e6a9e23ee26ee44afa0 : SUCCESS in 1h 49m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9f818f0319fb44a1bdb845d60be50558 : SUCCESS in 1h 51m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d0fe86fd84324032af39b1da0bc5b2b0 : FAILURE in 10m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/16d36bf0cd3f4595a962cee57a25a79e : FAILURE in 47m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a17d976e4de4d369f6606a16814a9ff : FAILURE in 54m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0887c15ef95841c98906503d22395e01 : SUCCESS in 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bbc30bd3f3e642438f1e10a52823e161 : SUCCESS in 1h 54m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16d71a6a52b94abeb17615273c50ed8c : SUCCESS in 37m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/eca1643a73f445e7a914970821640dd0 : SUCCESS in 54m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8cb2d4102c5646a7b5b83d69c7279d4e : FAILURE in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"696fec96aa04bbd022519aa06e3478a1638800bb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:20:56.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4d3a1730bce6aab0c99a5083baf8cdbc14f38e8c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:26:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ff6093a120a880adf98bc3954d17804857a2610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 09:27:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8bb49b1f7f59e7087378b88f65c022c89808455","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:06:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8092995b94b4f6dae48c59d5032224f : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24fe689b43b4a39b08242d600f6d8e0 : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/787f7653d92d45b9983cd1631319148e : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/414f5359a29a4d37a7d505704a781886 : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f6acc7dd7a74f419dd27b3181f4c98a : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b397dbf94540d98b836286755c7c22 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f13a877874ec40efb52cf514919fe292 : SUCCESS in 5m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8cba073ae3e74f0ebb67e4e01d651a9c : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27666578932c40df98f5220a23d704de : SUCCESS in 10m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45498caf01fb4c1599c390b625a1e918 : SUCCESS in 1h 15m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4fb7de19f4a74d56b60c4155e1828370 : SUCCESS in 1h 19m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c1f7bedea714adbab17fb267f6c1d4a : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c0fdb10156a40d0604cbab7e4b76b43d5384ee9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:06:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a001c6aef3f7ad3f079ac9e44243320b7cff89b2","revisions":{"a001c6aef3f7ad3f079ac9e44243320b7cff89b2":{"kind":"REWORK","_number":1,"created":"2021-02-22 08:48:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/09/776909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/776909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/776909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/776909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/776909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/776909/1"}}},"commit":{"parents":[{"commit":"131c69608dcb3c621f2dde12d4c4b0ae1c87bf45","subject":"Fix amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/131c69608dcb3c621f2dde12d4c4b0ae1c87bf45"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:52:32.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 08:47:58.000000000","tz":60},"subject":"Fix pools going into ERROR when updating the pool","message":"Fix pools going into ERROR when updating the pool\n\nThere was a bug that would cause a pool to go into ERROR if you attempted\nto update the CRL or client certificate on the pool.\n\nConflicts:\n    octavia/api/drivers/amphora_driver/v2/driver.py\n\nChange-Id: I736816247131715f5c385b4680614ec3218a2ad7\nStory: 2008295\nTask: 41180\n(cherry picked from commit 370aa4e61c421311b06d986aafdf123cd6737fd8)\n(cherry picked from commit 24e3a4d93a5c0ae6dcc880b35bd847ba0393e040)\n(cherry picked from commit ebe234eb252ffd4a051eab6bf05e587430d8432e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a001c6aef3f7ad3f079ac9e44243320b7cff89b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a001c6aef3f7ad3f079ac9e44243320b7cff89b2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
