)]}'
{"id":"openstack%2Foctavia~776912","triplet_id":"openstack%2Foctavia~stable%2Ftrain~If2841720009c2e402127e2e0080efdd56b68f6c9","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"If2841720009c2e402127e2e0080efdd56b68f6c9","subject":"Configure rsyslog on Octavia service nodes in devstack","status":"MERGED","created":"2021-02-22 08:51:42.000000000","updated":"2021-03-05 13:14:05.000000000","submitted":"2021-03-05 13:06:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776912","meta_rev_id":"069071d0c0880cb03ceece297867b83d2623adb6","_number":776912,"virtual_id_number":776912,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-01 09:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-04 00:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-05 09:26:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 08:51:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-02-22 12:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 09:49:05.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-04 00:42:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a8655013c9edac50dbbf933b5b95ba68bef60567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7d8ea3d37ee765fa96ebce15aa393f121ad8319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:16:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37aca02838fd4cf69bd0efd063684ef2 : SUCCESS in 10m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd347be4210f4e8792dfd1e640d01998 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ce1e86efe64a22a68859bdbf9fc8ea : SUCCESS in 9m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62f884dd75f44ed397af98679e138d7e : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/988c346e92fb40eebe1837e9f5515d43 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c251ae91781469fabda53b9ddb2a2ef : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c871206efc84ed9b7bea29f321081be : SUCCESS in 11m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4ae5e336f50a4f19b90eee8310ea2e78 : SUCCESS in 8m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3fef0abaaa31436bbcb62aa802911071 : SUCCESS in 8m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/05c13216d7604a7da412971d5f458676 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8164cd7e3ee147e992cb6db0887b85c4 : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99c1d05d4cd749ffaf1e95c573ac9c79 : SUCCESS in 1h 15m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e338ed58b44949109175a5ef591c39db : SUCCESS in 1h 26m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb5ab893b465488eb598546b992cad42 : SUCCESS in 2h 13m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/781eed88332848c6a5a8a2c9d6348685 : SUCCESS in 1h 56m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/66aa8ba5c444437996cfaac7ee9e8e1e : FAILURE in 10m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/173a0076de594f3ca16d63202f63fccb : FAILURE in 49m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9719dc166a34e52a805becbdce95243 : FAILURE in 47m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/99892ab7d2b34bbe93f0c156fc4d51b8 : SUCCESS in 45m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/de9c4950031641ccb6976d8e45278aab : SUCCESS in 1h 40m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8466a3efccd8417b9f509f0951df2a62 : SUCCESS in 44m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fe77d46e42854f31ba8c0da9727a1235 : SUCCESS in 41m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d18228ccfa9b460395389eae42d40003 : FAILURE in 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5af05567ae8ee8e88d191c9b68e62524109f308","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:49:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"404bf099b1d41fad163272bd4d41a072940b871c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:42:00.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"5e285e7f39cda9d76bf2ac12985d8b8bfc54e267","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:26:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c427f747ce4c65f6385025408711fef02118379c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 09:27:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"318d435e7f9f4f9c09dadd3f239ea270f1d97f3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:06:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c3080b509ae4077a37934a15284e0d7 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7cfe6d0f984496abfeb62f68f07b12 : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e73477159754f31989a87960cf90732 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d10284a28712495dadc93bf183da6003 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9968c8140570444d9beec6ead46647e8 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72717477710c4d31b4753f68122b91da : SUCCESS in 12m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b7e38167eb4b4d48bbc9f0cc81749010 : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3655a3e7a48f402d8c39efc48fa72164 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74dfc4d598294103878181f1cb2acaf1 : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5488d93016c44fcc8f548f7d9144c010 : SUCCESS in 1h 33m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8095135351aa461493db40fca43496a9 : SUCCESS in 1h 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ec7d0bc6066d84c50b1dfbd0d1c86f1cc8ffa76b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:06:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"069071d0c0880cb03ceece297867b83d2623adb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:14:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6188c383de264d2585ec67a0cbe9efe5 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"726d322f0a6edd52ff281a320a5c0505f3e97792","revisions":{"726d322f0a6edd52ff281a320a5c0505f3e97792":{"kind":"REWORK","_number":1,"created":"2021-02-22 08:51:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/12/776912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/776912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/776912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/776912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/776912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/776912/1"}}},"commit":{"parents":[{"commit":"131c69608dcb3c621f2dde12d4c4b0ae1c87bf45","subject":"Fix amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/131c69608dcb3c621f2dde12d4c4b0ae1c87bf45"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 07:45:33.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 08:51:24.000000000","tz":60},"subject":"Configure rsyslog on Octavia service nodes in devstack","message":"Configure rsyslog on Octavia service nodes in devstack\n\n{admin,tenant}_log_targets options are configured with\nMGMT_PORT_IP in devstack, which contains the IP address\nof the local management interface. In multinode setup,\nit means that the second node should run a rsyslog\nservice to receive logs from amphorae that have been\nspawned by its worker.\n\nChange-Id: If2841720009c2e402127e2e0080efdd56b68f6c9\n(cherry picked from commit f45092a87673b54d578d8df058917e325d9b77b8)\n(cherry picked from commit 0e593f4d2645f74e780b6da288f2fc36f75dca8e)\n(cherry picked from commit e9bc5d9f71c54fc05f93c6359ee1275e97ea4760)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/726d322f0a6edd52ff281a320a5c0505f3e97792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/726d322f0a6edd52ff281a320a5c0505f3e97792"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
