)]}'
{"id":"openstack%2Foctavia~776919","triplet_id":"openstack%2Foctavia~master~I196e13996ce4bce93d9a8eed87f6c3cb8ef3cea1","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I196e13996ce4bce93d9a8eed87f6c3cb8ef3cea1","subject":"Fix pep8 errors","status":"MERGED","created":"2021-02-22 10:56:37.000000000","updated":"2021-02-24 12:03:21.000000000","submitted":"2021-02-24 12:00:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776919","meta_rev_id":"f4ba983019f2372e04f44df6067aab0e04626779","_number":776919,"virtual_id_number":776919,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 12:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-24 09:26:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-02-23 07:56:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-02-22 17:57:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-02-24 09:26:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-2,"date":"2021-02-23 07:56:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 11:58:52.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-02-22 13:30:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 15:22:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-22 16:48:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-24 09:26:36.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"e51d9a543ca5bbb832a05f02d6a27525e86fc141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 10:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4a77094dfd85f50000e7c4b1b7230b9b0d9e928","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 10:57:02.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"32d8a580167f45c2c01000938be4782e561b28f4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 11:58:52.000000000","message":"Patch Set 1: Code-Review+2\n\nSeeing pep8 and other CI jobs passing, LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"0d80b392d6409f56d84db3fb0e24f99e76411bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 13:30:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dac148bab8654681951972796100adb8 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f94f8aecc6c54c9f94567827415fa40c : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8daba0b7fe4a45949138337434de0845 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/166c61cf07b64be7a3b41a073f024fb8 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a3bfe3792f549c3aec47b1ef40ab7f2 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaf89480aa0542689204608e4a8df094 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe883dabce24de9b1de9b9cd790946d : SUCCESS in 12m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9db1b5171658407caffdaeddcaf87172 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9950f398066e4c1e9951bd617cd65d8e : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8cf5f1231d534bb59dcd5da8202ab7f9 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc1d37450b2d45e9a36c13ac7356e8eb : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/929f56ad277a4ccfad41e8927326b576 : SUCCESS in 1h 17m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c2d0b9eb9504c3996af36597205ab91 : SUCCESS in 1h 21m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1258470994e24877b0c2dbface86740d : SUCCESS in 44m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2344edc737ef4f4bba851dfc6d9a0781 : SUCCESS in 41m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82308ec8068e49b5bf1b16867ea7f34c : SUCCESS in 50m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/42d36a75826a4a038f406dbf4bd9bc85 : SUCCESS in 55m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73ec56ec89764afebb7eb6b7b55410b2 : SUCCESS in 1h 57m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7499ea0997034f0192c3b0fc5d9f41f6 : FAILURE in 2h 01m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6a705332bfe641bf9b9300439172db87 : SUCCESS in 2h 05m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9f1a06227974dde8b32d9f721a736e3 : SUCCESS in 2h 22m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/918aab467f1e4d3fa56d1a26c9973775 : SUCCESS in 2h 11m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/740cbd4e966240ce91657fccaab76b4b : FAILURE in 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec3f6786dc31e52ffc890f3241251a5dd8909a31","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 15:22:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e283e2e796e0f55993118f77fbd94e03b24faa5a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-22 16:48:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"313d4c2ab9c6a27c68e428f167d2007153be0699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 16:52:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d197df22c284f6eadb04c2f03ac9c946fbb63557","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 16:55:13.000000000","message":"Patch Set 2:\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5db2fff1b279de4ba3b10d18aa58d33ecc17524e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 17:57:31.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Greg!","accounts_in_message":[],"_revision_number":2},{"id":"86576d9513903016c6e3b604ea05f3c2ab3cb575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 19:04:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bed5c8346dd349c19831ba939cec2342 : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8f47bf32fff4d3cb724f3487cf2ea5a : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abcbb64c93154f3b878cf47309384c73 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94b0dddf191f46cfb60506d906f0325f : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/359692e337504c99a1e187d07749ee05 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59f85b640c654ee19aee091b80414a21 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cffa94f8834d8bbd8d6b070644ffb2 : SUCCESS in 13m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/637065f037d34abbaa3adc68f780f6a8 : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e56c90f94ef743ef85206b684a22d7f9 : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/835251d7500342139c5b8577a9513901 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c080f631a064f48907344181b3bfefc : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd0fec3d2e7640b7a3cb339f97931f41 : SUCCESS in 1h 20m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62933f4fb65946b4ac7fa1e940e96337 : SUCCESS in 2h 00m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/932b555cf760407697c92a25a8ffe7b2 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6b2683fa7c0a45dc89690f394aa586d3 : SUCCESS in 38m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97a3d8488bab437aabe29ba18bc879e8 : SUCCESS in 41m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b70c720232d24d72b29f1ad9bf07e868 : SUCCESS in 58m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/685ad4e212084e8da2335424993212c4 : SUCCESS in 1h 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff543a758c5b4ce3aca23f0708e2bde4 : FAILURE in 1h 57m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b40cf30d4628418aa4aeb2d0d9b08fd2 : SUCCESS in 1h 50m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3ccd70c52b1e464aa38a876dae8d8695 : SUCCESS in 1h 37m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/437b03f77f524785afe3ff8bd7e7b64c : SUCCESS in 1h 41m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/44015fe1a7494354ac7173160bc3d4c0 : FAILURE in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a48d7bc7f1290571142e8eb148a32dedb6d69ebf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-23 07:56:00.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2\n\nBC-2 because stable/victoria branch will take a different approach: https://review.opendev.org/c/openstack/octavia/+/776926","accounts_in_message":[],"_revision_number":2},{"id":"dcbf1a0c4acb82eac9353ac5647c16b1e494382a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-24 09:26:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8562158c0b9806f86eed743dabf69d23b25ed15e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:27:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93d83d2257ae065d6f1fd3769a78494f633b671d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:00:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/964e24f6b26c42ff9475f83a441f707b : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6e9f80f9154a538a7d56533aa4bf59 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/565b6ea5e8c347cd8428ab4d4e3f8a82 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/345ac71d39c246d59163071ae40e53ee : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb97d527cba49c6b9f1cb809964bdba : SUCCESS in 17m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1263b34037c84707b33df144f2eef415 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93aa6bc0f45245f2aa2db7bfe4076221 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b7e6a526c96346919309a8e1cade36ef : SUCCESS in 1h 00m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3950e6fd9bd4ed8b2de2d94c6c5f89c : SUCCESS in 2h 22m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6cc3fccddaff422fb027488ce48c9c95 : SUCCESS in 41m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3cc27ef6fe674b3eb6227584d4ad3f48 : SUCCESS in 40m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3b30ce7531442e0883a824da85ac607 : SUCCESS in 42m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6735ba3a725044aa878f135304577418 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3f6548f6a3b428860c8d13c670061029f1548bf1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:00:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f4ba983019f2372e04f44df6067aab0e04626779","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:03:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8d47f55dcd4e31886d82489ab7df1d : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c37738ba6f44686d88d9da6d1010e1b8e6af2dc5","revisions":{"f3c378e7b535a6e082e981271d8712c00d17e43f":{"kind":"REWORK","_number":1,"created":"2021-02-22 10:56:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/776919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/776919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/776919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/776919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/776919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/776919/1"}}},"commit":{"parents":[{"commit":"ce2e0d74258bec3a5ed4983462acee234c67864b","subject":"Add test coverage for SCTP health checker script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2e0d74258bec3a5ed4983462acee234c67864b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-22 10:51:44.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-22 10:54:27.000000000","tz":60},"subject":"Fix pep8 errors","message":"Fix pep8 errors\n\npylint 2.7.0 added additional checks for inconsistent-return-statements:\n- inconsistent-return-statements message is now emitted if one of\n  try/except statement is not returning explicitly while the other do.\n\nChange-Id: I196e13996ce4bce93d9a8eed87f6c3cb8ef3cea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3c378e7b535a6e082e981271d8712c00d17e43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3c378e7b535a6e082e981271d8712c00d17e43f"}]},"branch":"refs/heads/master"},"c37738ba6f44686d88d9da6d1010e1b8e6af2dc5":{"kind":"REWORK","_number":2,"created":"2021-02-22 16:52:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/776919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/776919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/776919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/776919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/776919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/776919/2"}}},"commit":{"parents":[{"commit":"ce2e0d74258bec3a5ed4983462acee234c67864b","subject":"Add test coverage for SCTP health checker script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2e0d74258bec3a5ed4983462acee234c67864b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-22 10:51:44.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-22 16:52:27.000000000","tz":60},"subject":"Fix pep8 errors","message":"Fix pep8 errors\n\npylint 2.7.0 added additional checks for inconsistent-return-statements:\n- inconsistent-return-statements message is now emitted if one of\n  try/except statement is not returning explicitly while the other do.\n\nChange-Id: I196e13996ce4bce93d9a8eed87f6c3cb8ef3cea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c37738ba6f44686d88d9da6d1010e1b8e6af2dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c37738ba6f44686d88d9da6d1010e1b8e6af2dc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
