)]}'
{"id":"openstack%2Foctavia~776926","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I23319b1d8d3de0509b20d4a37bff93fc2155d47f","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I23319b1d8d3de0509b20d4a37bff93fc2155d47f","subject":"Pin pylint version","status":"MERGED","created":"2021-02-22 12:15:56.000000000","updated":"2021-03-05 03:33:37.000000000","submitted":"2021-03-05 03:32:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776926","meta_rev_id":"f7ef41dc946af8701072e514c514c63d0124cbad","_number":776926,"virtual_id_number":776926,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 03:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2021-02-22 15:32:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-02-22 15:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-03-03 16:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-03-03 16:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 12:44:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-02-22 14:47:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 15:32:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-03 14:09:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"11763acd241893be4b8560e452411a7abcbba93b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 12:15:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"889faa255ffe2baaebed531313009688e89f0464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 14:47:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f34ea23203384cdba224a1a1cea73967 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2177ec9cc1e45e0a79b17528d2ad8d3 : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3d28a5728d64b50baaa6f1cf9cacb29 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a3b43505d914168bb000b31cc8d6be5 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/379494a1f4a040c7baaabc723e01cfa7 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea7c8130bdf34498a713bbf01a012d43 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e66f2d024874edfbc7c08cf009d0b01 : SUCCESS in 14m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f9b62a68eb74a54914e10d2ae00c4f5 : SUCCESS in 5m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8a6ea29116524379bcb359292f7f1e67 : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fe6534490404435d8e495319f895e7b9 : SUCCESS in 10m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f5b94ab9a80644f98ad3ab4b9d53d737 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/976e8c3c066b47009d12a2a32514a1ac : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b3fd590438f3499db1a85f921407a37f : SUCCESS in 1h 12m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d4aebdace94480d89d9263c0cd19859 : SUCCESS in 1h 32m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e951ea6d8aa748e88bc80553a850e2d0 : SUCCESS in 46m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/912441ca7f54476f821172eeb0480141 : SUCCESS in 1h 08m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d694a85326a451eb513f9fc41b91646 : SUCCESS in 49m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/201299c404d74ade9a8773a12561dd5e : SUCCESS in 1h 59m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cada9f9588074bff9a600e7ab52e923c : FAILURE in 2h 13m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/20cd71e9c9974fd48da548cffa4cf514 : SUCCESS in 1h 36m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ae3c25fcc5a47e694a8e6888477beb9 : SUCCESS in 1h 50m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/86ebf0b70ce04c2185e40bb5e0eb5f95 : SUCCESS in 2h 24m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/62fe2bb59b1f4cc1a80aa8c41fcf9f91 : SUCCESS in 1h 38m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33b82563fe7468d2041ef5b2f011496de17bb26d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 15:11:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"198987b51614f754c7c429ad208953e86fad2b3d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 15:32:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9bfce84f9230ffb2f211df9bb49e416e44f7fb6b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-03 14:09:33.000000000","message":"Patch Set 1:\n\nLooks like this is blocking the stable/victoria gate","accounts_in_message":[],"_revision_number":1},{"id":"add2a879a10539853b48596ea5cdf4aebc80edb1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-03 16:07:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"35e6d8c477b85c3cb462f40b6f36051efceb1f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:07:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88b55d64db196394f1971b77adb1a859d18b6840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:43:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/272b578a58334e0296b39657017db63f : SUCCESS in 2m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337b03b0833a4c4183d2f09f29060fe9 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4df8fc196cc4c9cb52c0108257de46a : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef93520960f14e18959d1d077fc1d264 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dbd2ddd6bc344749327cb2887c1b74b : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef90db08189451db5cc0fcf04da547f : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5425b1ac71574dafb61062c728df0ff6 : SUCCESS in 6m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/12552de079f1418ab3ea440de597e829 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/655df03da4aa4a54adac6f61f772cc13 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/35c8e2f56cbb484f9adba74245478fb3 : CANCELED in 32m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb70b7944c104571a215e9846909c147 : CANCELED in 32m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19a81f6975a54cc6b96a51cb9e848b74 : CANCELED in 32m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0df9de915ac84fa7bd6d2e0955aa9fbc : CANCELED in 32m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da9652e1e28c4db5abedbc451c44d8f2 : CANCELED in 33m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea47f76a19fe4ab9bfa0a24f4cc0e13e : FAILURE in 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"36456fd19748ff680604bba0bb133a11e3623e9a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-03 19:08:20.000000000","message":"Patch Set 1:\n\nrecheck\n\nLooks like a dib-build mount issue?","accounts_in_message":[],"_revision_number":1},{"id":"d1b94b1e75a690cd944e3e9afd5f4eb3d3d719b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:26:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7942251be69948c1906d83e50c72fbeb : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70c1919baaa94417b884d809ad0fff56 : SUCCESS in 12m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b394b5b2e69f438986b030ad3c275566 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d30be239ba4d42a00e6f12fedf1617 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41c806c04d614117950fc207a75d7a68 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cc86b8c09704931966ff74a2b8582ba : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a98b9beca7974d90b69f07f102dc64ee : SUCCESS in 15m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f00e2545c8043b9bdb855adffe56a34 : SUCCESS in 6m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/31e1ac02e5ce433c8a4d239134071b5b : SUCCESS in 6m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/082848d0990743ea8afda0e9d5ae5e32 : SUCCESS in 11m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b0cf424a4dea449e9003a5396bb3c5a9 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e7457f94b0f4ad59c1df817fc69452d : SUCCESS in 10m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b7c3479705849f4a6c2c73364d900d5 : SUCCESS in 1h 30m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4749f04f6734067af2d0a100623334d : SUCCESS in 1h 39m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c729470d77d54cda9809b7a114d66083 : SUCCESS in 58m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/249f7ef57eb445709ca29d4f14d35a51 : SUCCESS in 40m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/597c182f84544c8f875ff931326f86bc : SUCCESS in 38m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c38e605d7e5d4aacbd2eda2dfe49d834 : SUCCESS in 2h 05m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca4ea853d4bf4604b2141ebd52bbeb15 : FAILURE in 2h 02m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/80e7f74c8c3c449b95a941b93b49e63c : SUCCESS in 1h 45m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb0377ffd0f44d42ad122f0ff71d0755 : SUCCESS in 2h 08m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bb17ab41d29b452b86d444807747bf16 : FAILURE in 1h 52m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/105604a9a2b34fda831a19a53a40b083 : SUCCESS in 2h 05m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e34dfc35b541461a67b28f8681346e5cabb8a0ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:27:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c270ec87329f5e836076f484e7503a94f4b25382","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 22:17:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81a0da8b4a064ed5aefc990e5d4b466e : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c24e946cfce4aeb810d73220b13d477 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8d9a30605d49d8aaa47dccd4bac616 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc65cd6fd3e14c8d999d9ccd59bc896c : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1c026b5e71a4fb2acb6ffa298e1bf9d : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4636e8c3905d4f8b8514466da4115061 : SUCCESS in 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dd6e1807fb74f56be308285a550b9ff : SUCCESS in 7m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ffe333943fc44423a79566eeb8545f5d : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4d615237cb4490dad3732ae5b6efe12 : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a675cb1a0f344dad8d751c058afa428c : CANCELED in 42m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7df8870eb1845108b696bde2e479ff6 : CANCELED in 44m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f611b445efa140b9bfb36d8b469ffc64 : CANCELED in 44m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/81bf1a41f6f0460fbc6ae595df3dbf90 : CANCELED in 42m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ab702b3c4eb47aa896bd366d81616d6 : CANCELED in 42m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f30fe31b3f184735b8f6ac462a464d07 : FAILURE in 42m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5e6f8380c2f14d1e0234c12de653245bbe4233f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-04 20:04:41.000000000","message":"Patch Set 1:\n\nrecheck\n\ngrenade fix merged","accounts_in_message":[],"_revision_number":1},{"id":"c14932e745de3a032c553baa7ea992dff88bcd3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 22:39:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/35a3bdd70e884238a4868f6051190125 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78b84d8b5c6346e9ba2ace0d563f985a : FAILURE in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd6e53312dca484a8480281b65d364c1 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d75dd43d21441319c790f96dab5542f : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44174345e5bd46378843cfdb0f7605c0 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a3770f6160a48f480b07c764a26f496 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e4f0cc053f40f7844a199e9786e0e0 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d58d1ac3fe14e148f76f2d9c1472da2 : SUCCESS in 7m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2f3df38c23fb44be9fc848f291e4f38f : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/35f107dcc89a47cfba2cae6a9bebb88b : SUCCESS in 7m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f64f866fec3d4765bedd0f6118d0fb11 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/759e92bb6b1942029e87581d7f7c833c : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/05304de5bdb74e0eb5d689bce9b46c57 : SUCCESS in 1h 23m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c095284ad84849908f750d9ae17ecf49 : SUCCESS in 1h 40m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/378b5b63cc6f45549e3094ebaf4e3ab5 : SUCCESS in 54m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8b144df75ced4d2591381ab865ec1f52 : SUCCESS in 43m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11ef0bf0da044fa18eef843a323f8418 : SUCCESS in 31m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8118025a5e12441b96c2eaaf569742df : SUCCESS in 1h 46m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4452531aa4434aa8a6bcdf2bce4accc5 : TIMED_OUT in 2h 32m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be9d93e35ed34d0d8712af8559f6f69d : SUCCESS in 2h 06m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/62df9004e4c941d6a5c9169ce192efcb : SUCCESS in 2h 15m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6ab3880f240a4cd2b598c2d1fed4d325 : SUCCESS in 2h 25m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0a273aad7ac14130b4fcd174f2f84e6e : SUCCESS in 2h 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55d5cbc4a903c591336d99824bd862af1627cf7d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 23:21:51.000000000","message":"Patch Set 1:\n\nrecheck\nbogus socket issue","accounts_in_message":[],"_revision_number":1},{"id":"fccf27f5271aa8c84c618b3afa914bdac243e3a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 01:39:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/33641afef12d4211bf3b1e51b9c3530a : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0bb01cfe8d94c71b577b25796f08fa2 : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccd05f403fa84e09bb723f19ccde1b04 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec904966d7c4855889a442246a4eea3 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27174a6c782746e280bdeb6ea6b60025 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b33fa5a5ca2a452aae3354128a98845a : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdad9dccf4d84405a245442e24a5b7e5 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64a4fd82c4e8424b985f0b209e3ebd81 : SUCCESS in 4m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/12ad87d9f85749098ba5713fc46f06e9 : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/494da0d9253f4856b5757bb9322326ea : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9f83a64e2f994c37967db9635ab78dec : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/545aba7ca3e045b6ab7996d55c519980 : SUCCESS in 8m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/604e44a650a44442a12f5457d495bdd4 : SUCCESS in 1h 23m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22090832cb144773a24a1aca3e0119ea : SUCCESS in 1h 49m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad0e4423d44d4bb09a175d4f566e259a : SUCCESS in 55m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d9c045082a504833bd9a0ee9b6b54596 : SUCCESS in 54m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0403237c0440486cb0956a953c6fbd6e : SUCCESS in 50m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/008d18233b144a248ed25c5f5950a3c4 : SUCCESS in 1h 53m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2365a9cad53a41fe91c801e76ff2d114 : FAILURE in 2h 15m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b6e920ecafb46eea4a97bb814dfa008 : SUCCESS in 1h 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2f211ef0df90455faffdd219854719f3 : SUCCESS in 2h 00m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/824f26ca3c9b405f94b82ff2845d28c4 : SUCCESS in 2h 06m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/236d6919e16048df9a4dec83a2cc9ed0 : SUCCESS in 1h 44m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0002f8bafd01690b5a83087cd5a7a072323f53d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 01:40:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4b63855ca10f58bf10ae1ce3fbc2a859ab0acd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 03:32:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0776087a6f142ee8833e4ed430f9cca : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6f82da925a2487db1f99c75e8f8f593 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f05274fcb08f42e5be61929346506788 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d12d7cc963e44a32bd47995adb68b00b : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7f499e2614f42babd46b3ac0c4f5b87 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3072518d857645d985c5fec5f8b03fca : SUCCESS in 13m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c68433936dc7449ab02c3c6b3be945f1 : SUCCESS in 5m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b1d37763c45442b0b15224d7fbc261b1 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d1635c1e8344a828ba054af19691272 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3949d9b6f417454aa3a7ce3757eb6415 : SUCCESS in 1h 20m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/909f14cff68343349899931ae4ae935f : SUCCESS in 1h 46m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f353ea10357b48eebefcd939d83ae29a : SUCCESS in 1h 00m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/09cb9d4a13a3471a8d1170ef83cc3f38 : SUCCESS in 39m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d3be26af5d545c4b8bc87dfe1f1deab : SUCCESS in 1h 01m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb7dfa0cbff845599796bf79b01b82cc : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"91810d529ed837082964b780f8218496545637e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 03:32:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f7ef41dc946af8701072e514c514c63d0124cbad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 03:33:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1bf1247fb74caab6120a755aca22e9 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/435780feea414b52803864d7172496ae : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fc8f22373437a2901984b986defb09f2764dbe5","revisions":{"0fc8f22373437a2901984b986defb09f2764dbe5":{"kind":"REWORK","_number":1,"created":"2021-02-22 12:15:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/26/776926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/776926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/776926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/776926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/776926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/776926/1"}}},"commit":{"parents":[{"commit":"b5c6c884a4a4de48eea7ea08ba865f08fbfcf972","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5c6c884a4a4de48eea7ea08ba865f08fbfcf972"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 12:08:22.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 12:14:41.000000000","tz":60},"subject":"Pin pylint version","message":"Pin pylint version\n\nPylint 2.7 added new checks that make pep8 fail. Following what we\u0027ve\ndone in previous stable branches, it is better to set an upper\nconstraint version to pylint requirement than addressing additional\nchecks as future pylint versions are released.\n\nChange-Id: I23319b1d8d3de0509b20d4a37bff93fc2155d47f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fc8f22373437a2901984b986defb09f2764dbe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fc8f22373437a2901984b986defb09f2764dbe5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
