)]}'
{"id":"openstack%2Foctavia~777974","triplet_id":"openstack%2Foctavia~master~I11083f3917b0f3694d7d7a7f87a47ce612fccfd8","project":"openstack/octavia","branch":"master","topic":"task/41953","hashtags":[],"change_id":"I11083f3917b0f3694d7d7a7f87a47ce612fccfd8","subject":"Defer import of octavia.api.config","status":"ABANDONED","created":"2021-03-01 10:28:59.000000000","updated":"2021-03-03 01:20:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e01d9b1c4215237982ef1b75460138f6687d9d7","_number":777974,"virtual_id_number":777974,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-02 18:59:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2021-03-02 16:44:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 15:21:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 16:37:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2021-03-02 16:44:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"74aa2b44f32b75ee992003213b2132afb55e3367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-01 10:28:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b45cfee6695ebc9a5ec414d9404b493bb8fb0e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-01 10:32:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a22065afee6860892553c3d3299b59e135e05f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-01 13:06:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"58e980007ea9dc7bdaba92c6ffb41025c23b6131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:21:50.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/424d21a8b75e46fc94ec761d9da5d8b5 : SUCCESS in 11m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/000e74ab103447ed8df11ca065101ff3 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944174800c7b45a38ce40a03bb1064fb : FAILURE in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19033c195d04415bb145b8ba426859b3 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3a34ccad1154b3a979aa847c0025356 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bb2eab78de14b0a8edbbf0a3c4f5353 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e340ca02ea4531a64875e3442bfd81 : SUCCESS in 11m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30772b7b6b234d078af33b85ea61cad1 : SUCCESS in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d590d7a71c1246adad07deec8d688e82 : SUCCESS in 9m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f05123eac724ae5b58df2caa3f9012d : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e769894dee44384bcf0cb5477a83611 : SUCCESS in 9m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f123bae80f614378ae1ab415494df343 : SUCCESS in 1h 22m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4258ff323b94ec5b05bfb11e9b0581d : SUCCESS in 1h 43m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc5b5c9f190b4785a14b58795951c738 : SUCCESS in 44m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bc49fd90e6de472ebea69993ad249435 : SUCCESS in 40m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2479e86172e418e982e06be57dfcf95 : SUCCESS in 45m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9bd0c3b99eb74bdcaffb4e939f7afb65 : SUCCESS in 59m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/68b92a87ffe94844b4577e554289f9aa : SUCCESS in 1h 56m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/558ab12959ed405cbae63353275fb94d : FAILURE in 2h 07m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/744f1586fe1a4ea1b0c7cf4fbc7c5b16 : SUCCESS in 1h 44m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3500b78acc24bdc9d14d24db973d690 : SUCCESS in 2h 11m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/81560e176a7f49c4b79155a361f2ccb4 : SUCCESS in 1h 56m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fef08c21d3444c67ac6a002f96868731 : FAILURE in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a4a732c87b3de52979a16006691c1b1d210e6fae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-01 23:42:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8ffd1d8ae22d04afaa53a1ffe48eab3eb3a9cbff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 01:53:03.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/904032254cf64202b3bbb80d2fe57c21 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b7b1e7cd5dc4602b2ac1333c7108f42 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34e73caa59b4124862399759b15c52c : FAILURE in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b8b5393e5c8400aa0257f2c9305c883 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5472b3d0f3984d8aa6b5299c5402afc8 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cda1386d61384c15834d661e818f8db9 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e219b2ba9f9e493089ef19e4fc0ab4eb : SUCCESS in 12m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3a2fd9bf75e94ebea4fde5ee0b3f8de1 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c70f29dac8a349aeab91a5a145d1b3df : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5a973be41b8b4b0f8d6bdef94b0960a2 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad8898eb9a4d4be896b48a957d675418 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b91d9398c9a347aaa0eab9411884841f : SUCCESS in 1h 51m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c1f62f37b5442e089264671c7a04975 : SUCCESS in 1h 45m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91a12519c125453897cf46ff370fbd3a : SUCCESS in 52m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/20aa40e2986543679493ab57777cb780 : SUCCESS in 44m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c94370f12474ffcbdb6b32dc3fd917f : SUCCESS in 47m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d30bb38743343ccb7b329135d53d338 : SUCCESS in 57m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ec7a9abd11543a08a2f6f4fbb074e3d : FAILURE in 2h 07m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74abded54ae941eea4ea8d3acc5d9057 : FAILURE in 2h 02m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/66c75dc0409b4f3682c53126122fcc77 : SUCCESS in 2h 05m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba3d67699e624bc88532302ea4c909c6 : SUCCESS in 1h 51m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/66e8dfd9558443249ac18353265b78c6 : SUCCESS in 1h 39m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/683a8dc677204dc3876dc231587048aa : FAILURE in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2198b66b617c5651129118a09fb010c64660168a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-02 02:13:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2e92153ee2b1d5b30965fd6933ed4bc78e84eac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:17:55.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4c30a17228c46d8a27910284bc6c620 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/365b9b8ca9894fc5ba58db04b5dbbd7e : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b371375fbe9e43a5894f11aa08755f1e : FAILURE in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53a0c367f3d945fe918bdd107dcddc46 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8bcbdbe3f70469689f5dc93330016b5 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ed5c8ca4aaf43ebb867a8e1720b4d91 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436071db1647463a994c20ed0fac3571 : SUCCESS in 11m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/281a6fd9963d4cbf8973bd3f9b99e41f : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5ad35d8b2b9049c8ae3d5e1a9072a243 : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1d75657c52814556aeee1c19777dc399 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3a4b350d2e34767867148d263a8a64c : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93cec48c56f94a43bf2a137c9288674a : SUCCESS in 1h 23m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab60f3281a6447fca072be295a4b4f5a : SUCCESS in 1h 45m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd7527bfeb864262b532859fc3302c73 : SUCCESS in 39m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4f92ba6c010c48b6ba5db254aaf39ea4 : SUCCESS in 37m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ce44cb4656f4f81946ccc83edff5a54 : SUCCESS in 43m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c6cdf9c94a4496cb52134fc244ee4e1 : SUCCESS in 51m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/076d4a48c7644f4bbcdbb7e4b4bcb38d : SUCCESS in 2h 01m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3af96f87a7ee4be09ed99b9dc7071b6c : FAILURE in 1h 55m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9fe3e6e0f5ff4149a4b7e828eab5f6a9 : SUCCESS in 1h 52m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/238db980474442f58b4e5512cf7eaf92 : SUCCESS in 1h 46m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/eae558e0c79b45c68c8c6ba62be0d1e8 : SUCCESS in 1h 49m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1e64fc29018544468d3437625ad878f8 : FAILURE in 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dde8e021c98f281631c746161897e576516c4bd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-02 16:15:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"76b8f1fb517e217bfc0885c0749a74b6c707035c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-02 16:20:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9bab9c886af43c1f580bac35aa2482fc8c49f7b0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-02 16:37:53.000000000","message":"Patch Set 7:\n\nI commented on the bug report, can you include the version of Octavia you are running?\n\nThere was an Oslo/PBR change that caused our Octavia configuration file to not be loaded at all. This has been fixed, but may not be present in your version of Octavia.","accounts_in_message":[],"_revision_number":7},{"id":"499cde5eb601202b607ea0658d734dc248f15362","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-02 16:44:20.000000000","message":"Patch Set 7: Code-Review-1\n\nI am asking because the configuration is always \"looked up\" in Octavia to support mutable configurations (plus it is a best practice). We don\u0027t cache configuration settings like some other projects do.","accounts_in_message":[],"_revision_number":7},{"id":"8cdff6dff5262ea98d52daaa2c82df4b235c8b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:59:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a89e2f9bbac54794946fc32b706494b1 : SUCCESS in 11m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8244c4fd67284d8794ef2cf01ea74627 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931c822462fe423aab57e62d01579851 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/348022558f2b45b2b2c75d2193f1dcb8 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d5e65247af94159926e8adcef606bd4 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abfc6d4444a44b44a3c60403af413ae8 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49407e2cd82c429c8dc8a250657f16f9 : SUCCESS in 12m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1344f4fca35a4a5eb563aeb1ca37f1ce : SUCCESS in 9m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62842a79c2ab41ae9d41aefb3ea31fc0 : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea0b3969b2724f4c939a7bd6659eaaf7 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a8534b4c81ff4813925b823bb9004775 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd5b969983ce4321971f1d89a63af546 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/daa8d4c6398a4e3a95c80ec08515cbc4 : SUCCESS in 1h 47m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2488725789344313bffad98dce077a3e : SUCCESS in 36m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7273caaa55e1467b94281f545f9e928e : SUCCESS in 42m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3761571527854dfcb4603fa5d09e18b6 : SUCCESS in 40m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d075313fdc254c39b042121a07f19b00 : SUCCESS in 59m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/89dec82dd08445d99fc4cde1fbab555d : SUCCESS in 1h 44m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bbb3285190b541c1ac1e08ac71fe7fca : FAILURE in 2h 16m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9bb5d81af2624d05a2025fcde2ffa14b : SUCCESS in 2h 30m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4c266318a47247eb86109a64608258a5 : SUCCESS in 2h 20m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8a39103f3ca248358e6b4940aee247f1 : SUCCESS in 1h 42m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/53d08c9ec1714722bbdb560f516a1e46 : FAILURE in 37m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"46b1f6a9387915f81b000eecee33576c3ffcb253","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-02 22:46:43.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e I commented on the bug report, can you include the version of Octavia you are running?\n\u003e \n\u003e There was an Oslo/PBR change that caused our Octavia configuration file to not be loaded at all. This has been fixed, but may not be present in your version of Octavia.\n\nWe are seeing the issue in stable/train (5.0.3).","accounts_in_message":[],"_revision_number":7},{"id":"a777728cf263a4cc144115275c92a8a7f323d019","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-03 00:52:31.000000000","message":"Patch Set 7:\n\nI have looked into this issue deeper and I think Greg\u0027s existing patch for this[1] is the best, consistent, solution. We should not be setting defaults in the WSME types if we can avoid it.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/777938","accounts_in_message":[],"_revision_number":7},{"id":"de360641577535b1aba39c476cc520eda7760dc9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-03 01:19:57.000000000","message":"Patch Set 7:\n\nThank you Michael, for your feedback.\nI\u0027ll abandon this change then.","accounts_in_message":[],"_revision_number":7},{"id":"1e01d9b1c4215237982ef1b75460138f6687d9d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-03 01:20:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e56998c16be8d86dd315482179f02e39301436d2","revisions":{"34cf3b83aca7303713198cd808ef6cb8a23ee822":{"kind":"REWORK","_number":1,"created":"2021-03-01 10:28:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/777974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/777974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/777974/1"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:22:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:28:56.000000000","tz":540},"subject":"Defer import of octavia.api.config","message":"Defer import of octavia.api.config\n\nWhen octavia.api.config is imported, it also triggers import of\noctavia.api.v2.types.* which were included by octavia.db.models.\nBecause this import is called in a very early stage before config files\nare loaded, some modules are initialized with the default config value\ninstead of the actual value written in octavia.conf.\n\nThis change defers import of the octavia.api.config modules so that\nit is included only after the config file is actually loaded.\n\nStory 2008666\nTask 41953\n\nChange-Id: I11083f3917b0f3694d7d7a7f87a47ce612fccfd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34cf3b83aca7303713198cd808ef6cb8a23ee822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34cf3b83aca7303713198cd808ef6cb8a23ee822"}]},"branch":"refs/heads/master"},"44b9e0a91fc80ea1ecb80f1cc5c7476506379896":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-01 10:32:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/777974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/777974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/777974/2"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:22:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:32:22.000000000","tz":540},"subject":"Defer import of octavia.api.config","message":"Defer import of octavia.api.config\n\nWhen octavia.api.config is imported, it also triggers import of\noctavia.api.v2.types.* which were included by octavia.db.models.\nBecause this import is called in a very early stage before config files\nare loaded, some modules are initialized with the default config value\ninstead of the actual value written in octavia.conf.\n\nThis change defers import of the octavia.api.config modules so that\nit is included only after config files are actually loaded.\n\nStory 2008666\nTask 41953\n\nChange-Id: I11083f3917b0f3694d7d7a7f87a47ce612fccfd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44b9e0a91fc80ea1ecb80f1cc5c7476506379896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44b9e0a91fc80ea1ecb80f1cc5c7476506379896"}]},"branch":"refs/heads/master"},"31e35c9c72562fe863eccd6456abd8fb0a9e924c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-01 13:06:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/777974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/777974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/777974/3"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:22:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 13:06:43.000000000","tz":0},"subject":"Defer import of octavia.api.config","message":"Defer import of octavia.api.config\n\nWhen octavia.api.config is imported, it also triggers import of\noctavia.api.v2.types.* which are included by octavia.db.models.\nBecause this import is called in a very early stage before config files\nare loaded, some modules are initialized with the default config value\ninstead of the actual value written in octavia.conf.\n\nThis change defers import of the octavia.api.config modules so that\nit is included only after config files are actually loaded.\n\nStory 2008666\nTask 41953\n\nChange-Id: I11083f3917b0f3694d7d7a7f87a47ce612fccfd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31e35c9c72562fe863eccd6456abd8fb0a9e924c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31e35c9c72562fe863eccd6456abd8fb0a9e924c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c0882ea11587b6d0957ae4f8e5e02859c9f53abb":{"kind":"REWORK","_number":4,"created":"2021-03-01 23:42:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/777974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/777974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/777974/4"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:22:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 23:42:40.000000000","tz":540},"subject":"Defer import of octavia.api.config","message":"Defer import of octavia.api.config\n\nWhen octavia.api.config is imported, it also triggers import of\noctavia.api.v2.types.* which were included by octavia.db.models.\nBecause this import is called in a very early stage before config files\nare loaded, some modules are initialized with the default config value\ninstead of the actual value written in octavia.conf.\n\nThis change defers import of the octavia.api.config modules so that\nit is included only after config files are actually loaded.\n\nStory 2008666\nTask 41953\n\nChange-Id: I11083f3917b0f3694d7d7a7f87a47ce612fccfd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0882ea11587b6d0957ae4f8e5e02859c9f53abb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0882ea11587b6d0957ae4f8e5e02859c9f53abb"}]},"branch":"refs/heads/master"},"2f67d43ecc33cbc52d242bd9a77ad328185c6b3b":{"kind":"REWORK","_number":5,"created":"2021-03-02 02:13:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/777974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/777974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/777974/5"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:22:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-02 02:13:12.000000000","tz":540},"subject":"Defer import of octavia.api.config","message":"Defer import of octavia.api.config\n\nWhen octavia.api.config is imported, it also triggers import of\noctavia.api.v2.types.* which were included by octavia.db.models.\nBecause this import is called in a very early stage before config files\nare loaded, some modules are initialized with the default config value\ninstead of the actual value written in octavia.conf.\n\nThis change defers import of the octavia.api.config modules so that\nit is included only after config files are actually loaded.\n\nStory 2008666\nTask 41953\n\nChange-Id: I11083f3917b0f3694d7d7a7f87a47ce612fccfd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f67d43ecc33cbc52d242bd9a77ad328185c6b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f67d43ecc33cbc52d242bd9a77ad328185c6b3b"}]},"branch":"refs/heads/master"},"71e3ba998f5dd22d6a7ba94468ed3ba1f8e96eee":{"kind":"REWORK","_number":6,"created":"2021-03-02 16:15:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/777974/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/777974/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/777974/6"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:22:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-02 16:15:43.000000000","tz":540},"subject":"Defer import of octavia.api.config","message":"Defer import of octavia.api.config\n\nWhen octavia.api.config is imported, it also triggers import of\noctavia.api.v2.types.* which were included by octavia.db.models.\nBecause this import is called in a very early stage before config files\nare loaded, some modules are initialized with the default config value\ninstead of the actual value written in octavia.conf.\n\nThis change defers import of the octavia.api.config modules so that\nit is included only after config files are actually loaded.\n\nStory 2008666\nTask 41953\n\nChange-Id: I11083f3917b0f3694d7d7a7f87a47ce612fccfd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71e3ba998f5dd22d6a7ba94468ed3ba1f8e96eee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71e3ba998f5dd22d6a7ba94468ed3ba1f8e96eee"}]},"branch":"refs/heads/master"},"e56998c16be8d86dd315482179f02e39301436d2":{"kind":"REWORK","_number":7,"created":"2021-03-02 16:20:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/777974/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/777974/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/777974/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/777974/7"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-01 10:22:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-02 16:20:08.000000000","tz":540},"subject":"Defer import of octavia.api.config","message":"Defer import of octavia.api.config\n\nWhen octavia.api.config is imported, it also triggers import of\noctavia.api.v2.types.* which were included by octavia.db.models.\nBecause this import is called in a very early stage before config files\nare loaded, some modules are initialized with the default config value\ninstead of the actual value written in octavia.conf.\n\nThis change defers import of the octavia.api.config modules so that\nit is included only after config files are actually loaded.\n\nStory 2008666\nTask 41953\n\nChange-Id: I11083f3917b0f3694d7d7a7f87a47ce612fccfd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e56998c16be8d86dd315482179f02e39301436d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e56998c16be8d86dd315482179f02e39301436d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
