)]}'
{"id":"openstack%2Foctavia~778063","triplet_id":"openstack%2Foctavia~master~I435c40a5493828bce24c2850182f0ddc8e5bdd1f","project":"openstack/octavia","branch":"master","topic":"fix-az-lb-failover","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-10-26 22:53:13.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-10-26 22:53:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I435c40a5493828bce24c2850182f0ddc8e5bdd1f","subject":"Fix failover of az-specific loadbalancers","status":"MERGED","created":"2021-03-01 20:35:42.000000000","updated":"2021-10-26 22:58:35.000000000","submitted":"2021-10-26 22:53:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778063-fix-az-lb-failover","meta_rev_id":"516487e79241a93d70d5474d6c2b8097b1ccbf99","_number":778063,"virtual_id_number":778063,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 22:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-10-26 09:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-10-26 20:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-10-26 20:39:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-10-26 09:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-10-26 20:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 21:55:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-01 23:24:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 08:17:05.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-10-26 09:42:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-10-26 20:39:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"565ed6e439ca009aac99936a2c12a2292fa83c09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-03-01 20:35:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"231a4074d0408e43b05040b66bbe31a45f5c03f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 23:24:11.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b4f68aed10e47ed8abcf9ee378901d1 : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c2c56666a79428ab7c396da093ab66d : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f1c4c912a741108a66aafea4de6831 : FAILURE in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/880c54017b6e47da8dd1a10e869bb6c2 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc0f5d191a5b4eafa2f67685615c0e5f : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d14f2224dfb540fdbdadcc051cc4a85a : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d05a46876a2411baabf5f475437d6a5 : SUCCESS in 11m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/239e88e0bdb742a589b4243cdbf619ec : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c242d065972944a984d80d9d0d6ce67b : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0cbc4e76a55445de8df4ddded280c79e : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6b63e8364a945bfa70ddd9acc1e192b : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e83b533004d74ff9a1f44f0cc13a4385 : SUCCESS in 1h 14m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32d84cd2e44e4b1c87dd6b53780c71de : SUCCESS in 2h 01m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/398af1c9dde847a39b4ca3f0da850d65 : SUCCESS in 56m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6d0ea448555540b3a7e3959bf5eaab5a : SUCCESS in 41m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8528ed79e2074462a4b19bf23214e016 : SUCCESS in 52m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/556f49f3bc0c4641a3fbc9389fbb80d5 : SUCCESS in 59m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8296e6dff1b41559d5168bfe675a8af : SUCCESS in 2h 05m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/313ea4ac89134651ab402592e10af713 : FAILURE in 2h 28m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/56a3a59377ea41ac808e567ca1f19a6b : SUCCESS in 2h 07m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3fe73650239749a1b00d9706dd21d826 : SUCCESS in 1h 58m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3386213d2fc44bb498a88860eb622cdd : TIMED_OUT in 2h 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6b72772a6cc948ed9b9b6c2f4b1b160d : FAILURE in 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdf2a7349739e21e64044cfab9c0a3315131534d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-03-02 01:02:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6add5ff74ca4ec89614e58a0d940aab83458c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 03:30:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0ecd3a582a64f45842b4abd032b665d : SUCCESS in 12m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07f05bbd30c7412f9a0c508cdd6c201d : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43c5f0fbb5c4284b50bdba2b0bc6b56 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b4e3799516426fbd28ffaaca66998c : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31a44a80456448ed991d948b8b953168 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea530f511d634bc2b9d46f08ab8ce941 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c045736d6bb04b829a550aeb4b48b6f1 : SUCCESS in 12m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3940cdbfa1594d0f8de10d2bacf3cca9 : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d6bd5c302eb949888cbe263e6b39d100 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/30eda98cc74e490fb2274b47c76502f9 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ecb854512e094703a7d1017110eff58e : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd07893834d941e29781e8d07c98c56a : SUCCESS in 1h 20m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bb70ee2649c4aba9498a8757cdf1b12 : SUCCESS in 2h 11m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d6432f56df84d20a53bfd25d97bc009 : SUCCESS in 51m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a157382a6a5f4c5dbfb9ac59e633e9b9 : SUCCESS in 48m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/160789e03c9a453c87dd8243da5c1a30 : SUCCESS in 38m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad6b2023bb2f4f1cad6e89d914739c26 : SUCCESS in 57m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec5b2d7262384a63890f49694dba06f9 : SUCCESS in 1h 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c70bf74e7004eeb8f92853b1652b892 : FAILURE in 2h 19m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f83085b4eb9d4190beb57df97b685c83 : SUCCESS in 1h 40m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2ee89a9920de4bcfb808a854332fdedf : SUCCESS in 1h 58m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2395dafb2205407fa0a3359cb872c9a2 : SUCCESS in 2h 25m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2d6b2a4b99e940f0bc9ddc8a907029b2 : FAILURE in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba2cf6d1fa66ae7258a19ab602d7a7e219eb0e4c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-10 08:17:05.000000000","message":"Patch Set 2: Code-Review+1\n\nGood catch! Can you also update test_calculate_amphora_delta to cover this case?","accounts_in_message":[],"_revision_number":2},{"id":"028ef5f364e71df342285904535755c41a5f343b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 09:40:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"983137b47c674dc7f313b51df720716974935daf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 09:42:20.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"862d5e086921374b1886290d43ca4d0e6ea2007e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 11:54:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b78ad64a0d7a42aabebd9e475ccf2878 : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d5661059ead46e1a4cdcb98dc2900c8 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bf1423f75c46d1a1698003b9f41beb : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b80e730d0196491cb7074c912664622a : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64045dd71bf844db84ed3bd68b4cd8b8 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f2d29cff5684cbd8a9ee63bf87c4832 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb54f1d8f884ffd9033a83e538b5d82 : SUCCESS in 15m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1966237e9b304f218b76f454a0dcae6b : SUCCESS in 7m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2c860027bef345eda7b5b269243c6dfb : SUCCESS in 10m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7308b91a396947d8bb55a58db235673c : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/937d91888081489994fbfc6bc565a57c : SUCCESS in 9m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e2d863164ae4470892ad0cfe7e0b7a4 : SUCCESS in 1h 59m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1693491a3df64d7eacede53ead50d700 : SUCCESS in 1h 24m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6a2d0dbdc1e48dca9fa0b0c8249ce50 : SUCCESS in 38m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a654fda89b6a4a019b332d4ba2d7fb9c : SUCCESS in 48m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc44a6aa10cd42dd8ec28ecfdaa38680 : SUCCESS in 35m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e0a79e2742e4024b5a02c4f45d30cc5 : FAILURE in 1h 40m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f01639e0090b4b5f8f1f719c2b45c1e2 : SUCCESS in 1h 53m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1a6449c980154d07a94a82c72e5574ea : SUCCESS in 1h 30m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0b9a2428894142e4baf3aa8d675f1500 : SUCCESS in 1h 42m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ef7c2c6f7f114bbdbf31927996d38015 : SUCCESS in 1h 27m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/eea787d5416140fcb6a22749b884a2ee : FAILURE in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9ab0fd7658626413dd5a3766f265b244e5414d2a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-26 20:39:46.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c3110ed57b8abacb40e8a959b76526aa51d998c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 20:40:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"757c055147769fb8afc4aee930a4415c0ae56fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 22:53:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1f22d8e60cd4368a5dec740f2734ad4 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99c5ac7f4ed45098196ac7555c426b6 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f27422e3b53a49b286bd9dfb482dbfac : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/047e7fa90ad94cfa8e6d820f6c2f2a5b : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05f031248904de49a85035650eee761 : SUCCESS in 13m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/282799c67c204596a08888eec3422102 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0202e5141ed043e58feb2ee12d177a97 : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/14af3a789a9442c9a0468b428a1dd919 : SUCCESS in 2h 09m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1520c00c2b5947c6b8994f67775a9c6f : SUCCESS in 1h 31m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e49215f0856d47cabee45785cdbbb02e : SUCCESS in 33m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2efcb609b0da4e80acc3601850f271f7 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":3},{"id":"bc872a8055e78ac11f476934a1ac4a110313422f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 22:53:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"516487e79241a93d70d5474d6c2b8097b1ccbf99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 22:58:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0ffa0f646a4536880308ec716ef5a5 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"81134bdfec79c8019c26e7846d31a2b2952965a0","revisions":{"324ba7dbc9fcd7615639d41d6610b5234814fe87":{"kind":"REWORK","_number":1,"created":"2021-03-01 20:35:42.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/63/778063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/778063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/778063/1"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:24:06.000000000","tz":-300},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:33:53.000000000","tz":-300},"subject":"Fix failover of az-specific loadbalancers","message":"Fix failover of az-specific loadbalancers\n\nThere is a bug exists in CalculateAmphoraDelta class\nIf there is no such a key as constants.MANAGEMENT_NETWORK\nin availability_zone dictionary, method get will return None\nIt means that management_nets list will become [None], not\nCONF.controller_worker.amp_boot_network_list as it supposes to be\nIt breaks the following failover flow\n\nStory: 2008669\nTask: 41957\nChange-Id: I435c40a5493828bce24c2850182f0ddc8e5bdd1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/324ba7dbc9fcd7615639d41d6610b5234814fe87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/324ba7dbc9fcd7615639d41d6610b5234814fe87"}]},"branch":"refs/heads/master"},"5aece52d0eadad9260474a264153228821a82984":{"kind":"REWORK","_number":2,"created":"2021-03-02 01:02:39.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/63/778063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/778063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/778063/2"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:24:06.000000000","tz":-300},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-02 01:02:17.000000000","tz":-300},"subject":"Fix failover of az-specific loadbalancers","message":"Fix failover of az-specific loadbalancers\n\nThere is a bug exists in CalculateAmphoraDelta class\nIf there is no such a key as constants.MANAGEMENT_NETWORK\nin availability_zone dictionary, method get will return None\nIt means that management_nets list will become [None], not\nCONF.controller_worker.amp_boot_network_list as it supposes to be\nIt breaks the following failover flow\n\nStory: 2008669\nTask: 41957\nChange-Id: I435c40a5493828bce24c2850182f0ddc8e5bdd1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5aece52d0eadad9260474a264153228821a82984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5aece52d0eadad9260474a264153228821a82984"}]},"branch":"refs/heads/master"},"81134bdfec79c8019c26e7846d31a2b2952965a0":{"kind":"REWORK","_number":3,"created":"2021-10-26 09:40:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/63/778063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/778063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/778063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/778063/3"}}},"commit":{"parents":[{"commit":"0b84631948bd3c5eb25a0921d5fbad657036304f","subject":"Merge \"Increase TCP buffer maximum and MTU discovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b84631948bd3c5eb25a0921d5fbad657036304f"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:24:06.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-26 09:39:54.000000000","tz":120},"subject":"Fix failover of az-specific loadbalancers","message":"Fix failover of az-specific loadbalancers\n\nThere is a bug exists in CalculateAmphoraDelta class\nIf there is no such a key as constants.MANAGEMENT_NETWORK\nin availability_zone dictionary, method get will return None\nIt means that management_nets list will become [None], not\nCONF.controller_worker.amp_boot_network_list as it supposes to be\nIt breaks the following failover flow\n\nStory: 2008669\nTask: 41957\nChange-Id: I435c40a5493828bce24c2850182f0ddc8e5bdd1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81134bdfec79c8019c26e7846d31a2b2952965a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81134bdfec79c8019c26e7846d31a2b2952965a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
