)]}'
{"id":"openstack%2Foctavia~778132","triplet_id":"openstack%2Foctavia~master~Ic63929d8864e5285baf70dd85e6362988bf2863f","project":"openstack/octavia","branch":"master","topic":"task/41491","hashtags":[],"change_id":"Ic63929d8864e5285baf70dd85e6362988bf2863f","subject":"Fix weighted round-robin about UDP and SCTP listener","status":"MERGED","created":"2021-03-02 05:09:19.000000000","updated":"2021-04-01 17:21:49.000000000","submitted":"2021-04-01 17:19:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"778132-task/41491","meta_rev_id":"ed5771527278d88636a581c30794588c955787f9","_number":778132,"virtual_id_number":778132,"owner":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 17:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-04-01 13:13:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-29 20:40:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-31 15:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-01 13:11:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-04-01 13:13:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-04-01 13:13:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-31 15:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 07:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 15:28:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-29 20:40:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-04-01 13:11:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-04-01 13:13:33.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"9c4780df69da0482f4762019e078a1d3e6d9a471","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-03-02 05:09:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be55244757ff327b46abdac04e42de21bf52a61c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 07:16:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a43cb95deaee4c75af9ba3aaf3360a70 : FAILURE in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/066d67364aa84d17b89df2a852cdd96d : FAILURE in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c65045de2148e28d31d8d6f9999d4b : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c12b098d474c4d0b8b8d7f3bba4fa2ae : FAILURE in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95e7f6819ce943a4a8989d1532041dec : FAILURE in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0cb275780e24b1fb52a7cd674b49184 : FAILURE in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d11a1de8e84c7d9a254ca0dc8ab8ab : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8fe10b6759c418d91407e91c7489e77 : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/23d9f711d67044f28f8eb03e2240745f : FAILURE in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/207cc846766b4be1b2aec5def71c9eb6 : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/70d9f4513eb240c8a82d05bfc317ae8e : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24e61ad0060e4420b90c8f138666adc3 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6dc8d58a53e44e68a7b8eb04208cbc5b : SUCCESS in 1h 21m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa994fffa52647c99ac5e986293f2c1f : SUCCESS in 1h 44m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5454394ab09548d0a54a47f71cd1a2f4 : SUCCESS in 45m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b1c2358050d64405bd4dc9c9ad8c9b80 : SUCCESS in 38m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a311e0ba30524d7587ebe488d6fda9c2 : SUCCESS in 38m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/11f91095231e48b3bede6e6b23a7685b : SUCCESS in 57m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c9e24246e60848f68c360432fbf43edc : SUCCESS in 2h 05m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b356ed8ea61641cea7439eff520d31ef : FAILURE in 1h 49m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e13ad008df354732af672d54aacef22e : SUCCESS in 1h 34m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1bc79f496c9b4f9bb4893690998c880f : SUCCESS in 1h 53m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f9df2d5cbf804ea8ae9a27bf6b438809 : SUCCESS in 1h 50m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1571b2b95c884614bfc82d48bb1b1fec : FAILURE in 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5e8891e49e4f7a2caab910647c2a5061bb3d491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-03-02 08:11:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c950337d06978962000a0ea78a35cda04668228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:38:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4343b9322e0409587992772ce679269 : SUCCESS in 12m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3810e061d4dc4fb5861edb0f1cd6fac4 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a95a479808ad48708a4749622bfbb3bf : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ea471561b044f809108892cb26d7118 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f97e59419a146929aab1cfe735b89dd : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23c155f6814b432cac197675b9b68786 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab94dd41ad448fe8905ca913ed6ca73 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9241285c0c37496f951f89805b32dd4e : SUCCESS in 5m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a58ab3705f3a4fa6a5072029b0056010 : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a822cda463614272a8a05f2653236dac : SUCCESS in 9m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/61a8d37964e94e98baae9fe45e819920 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e97638101d74a37bee15eb8c7776e9f : SUCCESS in 9m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/879fcb0dba644b4faf83d32143b48d62 : SUCCESS in 1h 17m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12ba4ad23e5d44ba96f4bc0c5c688097 : SUCCESS in 2h 11m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f98074cbed9c40369d7a7e2e9cc9b8bd : SUCCESS in 45m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e6bf00c899b34313ab332dca27f465ea : SUCCESS in 42m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62ddc578834349dd9024c9c95360e32b : SUCCESS in 42m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f3fdb48a26a241a39a411eeb035d9ba7 : SUCCESS in 1h 02m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c81958ec4d7346bf8b78bd99be390799 : SUCCESS in 2h 00m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72ad62b537ba46bc8d9dae40714d2a4e : FAILURE in 2h 22m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6fc8082dcd4995990ae3f6c28ed9a6 : SUCCESS in 2h 06m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f1542e25e14541a6b07627302a9be6dd : SUCCESS in 2h 05m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c09a7a7d67c74eb8aaac3e467849ab89 : SUCCESS in 1h 52m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/54e1d9b7919045c8a4365c78959f458b : FAILURE in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eddadfc891f7576b5e041eab68e41f42be8a0b1b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 15:28:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"03bdc93355d28cae1fdb72f4066830433a189cca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-03-03 02:48:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"31996e631fbee737689d8729870d6cc2c5b8c658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 05:23:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec4e476fe38f41cb9cc47c942c323187 : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9479d1e535f47c989bb6fad3c2660de : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e497a38a2bc14fb0a406bce586c1f70d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eee7a3cceba74684a2345215a7c4d6e1 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/290c92af8da144f5a0b9f6a8d0721ca5 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d839c29768047d785f26489d52255e4 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4978fb65b74e8ebe699a4e7b56f48a : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c14281a818ed49249d5919e16c79cf14 : SUCCESS in 10m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/114c0bd86624476d8da295e7e6dbfe74 : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/350aab3387fc4a2c9d5b6b53bb7bc476 : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7c1dedc4d08b4093a5fdae6688874f14 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/585e6ebb0d4d4929b9f1621158bd91ca : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1daaeedf43bd483983247dc413dbf964 : SUCCESS in 1h 22m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdd99146a1ad493d8e075f45a2e8e470 : SUCCESS in 1h 56m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eebd0b451af74b879c016ff614e4ef42 : SUCCESS in 1h 04m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7a9d8e8587d844b6a0be3ebd8a63cd9e : SUCCESS in 42m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17c2dc201c6944b39754a16538408370 : SUCCESS in 1h 02m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ec784565f5014de9ba229bb94d4dc37d : SUCCESS in 1h 00m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a0a1f925432405cb5da6a39c50b1624 : SUCCESS in 2h 14m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f4b47408ff1422b8ae463a3cf31450f : FAILURE in 1h 57m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e761feaab02a4069b5774399365a5ff5 : SUCCESS in 2h 10m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/491e611b344c46a298ced77bbd6ddfef : SUCCESS in 1h 57m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/031b04227bce4cde969146c778fe463b : FAILURE in 2h 22m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c58ee7b740fc43668ad200d761a7e30d : FAILURE in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1506f18aed1e5e3a458ae5a86e24c44838a36d4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-29 20:40:16.000000000","message":"Patch Set 3: Code-Review+1\n\nI think this looks correct, adding Greg.","accounts_in_message":[],"_revision_number":3},{"id":"d0b5295dab66643ddacc6497df1257c483a9c74d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 15:48:43.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\nLGTM, Thanks for the fix","accounts_in_message":[],"_revision_number":3},{"id":"d0832fee3d39951e430df220fd8b88133a0ae3cc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 16:32:31.000000000","message":"Patch Set 3:\n\nFYI I verified the fix in my devstack env.","accounts_in_message":[],"_revision_number":3},{"id":"cbad73bbba7bef106e20f64a02baf8ccb87af82d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-04-01 13:11:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0730947bbc6cda1bade81be59baf87058a3ba662","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-01 13:13:33.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ec78929537a9f3a566963cf7ec7373246cce6fe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:14:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1ef8226cab62a140b9d67869a5bd008e34c97096","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:42:02.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4e4b9a24bc276843463a81e1fd71140fe74239b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:19:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/835b75873c0e49c9b59793b321e238e5 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d91a43c6a643b28ea1fbf3501e200c : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db7f854dabac4eb5a3c54bdc76c01de3 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0236f6b48a93477d8062b9b727f4d668 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3be6a10a3bf2461faa059dff8f7a8a1d : SUCCESS in 14m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c5e2106b744c2ab9c853f1acf80b3c : SUCCESS in 7m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/67730eafee8f4699b96f49df8a6cf54b : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50b126123fbb4f3f95680ef3a6ad7017 : SUCCESS in 10m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4431fae6cd8e483d99f55c50984798db : SUCCESS in 1h 32m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef7221857fcc4d32ae0a556c57490c8f : SUCCESS in 1h 24m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/250b0f28ff6c461c943ca35f74ccf3e5 : SUCCESS in 46m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2c99ae1ef394437faf790ff995d232ff : SUCCESS in 41m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/489fd2a3166c40e88f24f6e297b7a57f : SUCCESS in 43m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c19b9ce53a764ab88f38f9ff60da3064 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":3},{"id":"ab231129c666b0c23b56e30c3f7307f480361d21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:19:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ed5771527278d88636a581c30794588c955787f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:21:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65db11221e414094820cca3447afc585 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8013aa9ca926497cb3519d4329397d3d : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5352a10f62f1a4763aeee81e2e638a9896b6ce79","revisions":{"5e5e3951b052b51aeb35ae1664ed695f02cda935":{"kind":"REWORK","_number":1,"created":"2021-03-02 05:09:19.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/32/778132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/778132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/778132/1"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 05:09:06.000000000","tz":480},"subject":"Fix weighted round-robin udp listener","message":"Fix weighted round-robin udp listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work with weight for udp listeners,\nit should be \u0027lb_algo wrr\u0027.\n\nhttps://keepalived.readthedocs.io/en/latest/configuration_synopsis.html\n\nStory 2008462\nTask 41491\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e5e3951b052b51aeb35ae1664ed695f02cda935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e5e3951b052b51aeb35ae1664ed695f02cda935"}]},"branch":"refs/heads/master"},"062bd169570ce184c960141a67b1faf246ffd737":{"kind":"REWORK","_number":2,"created":"2021-03-02 08:11:20.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/32/778132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/778132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/778132/2"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 08:11:02.000000000","tz":480},"subject":"Fix weighted round-robin udp listener","message":"Fix weighted round-robin udp listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work with weight for udp listeners,\nit should be \u0027lb_algo wrr\u0027.\n\nhttps://keepalived.readthedocs.io/en/latest/configuration_synopsis.html\n\nStory 2008462\nTask 41491\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/062bd169570ce184c960141a67b1faf246ffd737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/062bd169570ce184c960141a67b1faf246ffd737"}]},"branch":"refs/heads/master"},"5352a10f62f1a4763aeee81e2e638a9896b6ce79":{"kind":"REWORK","_number":3,"created":"2021-03-03 02:48:40.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/32/778132/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/778132/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/778132/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/778132/3"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-03 02:48:28.000000000","tz":480},"subject":"Fix weighted round-robin about UDP and SCTP listener","message":"Fix weighted round-robin about UDP and SCTP listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work correctly with weight\nfor UDP and SCTP listeners, it should be \u0027lb_algo wrr\u0027.\n\n\u0027wrr\u0027 is superset of \u0027rr\u0027 algorithm, as round-robin is\na specical instance of the weighted round-robin scheduling,\nin which all the weights are equal. [1]\n\nAlgorithm in HAProxy is set to \u0027roundrobin\u0027, which also support\nweights, but in keepalived must be set to \u0027wrr\u0027 to work with\nthe weighted round-robin scheduling, as it is different to \u0027rr\u0027.\n\n[1] https://www.keepalived.org/doc/scheduling_algorithms.html\n\nStory 2008462\nTask 41491\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5352a10f62f1a4763aeee81e2e638a9896b6ce79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5352a10f62f1a4763aeee81e2e638a9896b6ce79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
