)]}'
{"id":"openstack%2Foctavia~778377","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88","subject":"Fix overriding default listener timeout values in config file","status":"MERGED","created":"2021-03-03 07:47:46.000000000","updated":"2021-03-05 13:15:23.000000000","submitted":"2021-03-05 13:09:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778377","meta_rev_id":"4502414dbd80d445c60a25d02873e22c3a57c4e5","_number":778377,"virtual_id_number":778377,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-05 11:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-03-05 09:16:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-03-05 04:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-03-05 11:03:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 10:35:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 14:12:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-03-05 04:14:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-05 09:16:50.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-03-05 11:03:01.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"81e02028c17304b3e67e4113672a8af631b8323c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-03 07:47:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4369f5ff6025908e3ce8af6fceed02b9a364751f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 10:35:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4c27da09ded4a50a7208d0a54ab48ee : SUCCESS in 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25ae9a64020744ad97e4ed6c381cd10d : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce8150b10e664bff8ce98bbf95579478 : FAILURE in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35c1e75d094e4b7c9fce69ebcddf3d4c : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02842411e8bc4e6ab8fbd516e03ee304 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29a51f0a1f644ebba9b028b893efa67 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/021b893ff07a4ec194825cf79b7e4b6f : SUCCESS in 6m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f3cbd82a33544c1cad660dd01ac7b312 : SUCCESS in 8m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b36c36e418da49888ed32ca02b8903ea : SUCCESS in 9m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6ef5fcacca114bafacf6f059ea12595e : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0011737e32264ce79a174a2b29c065dd : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcbfc22122d54cfa84242bad44404d7a : SUCCESS in 1h 19m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8168ad1b713f46c3876dfba71ba8d3ac : SUCCESS in 2h 21m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/470fb703fb40417ea1943909cb30e2f7 : SUCCESS in 39m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be1f0f67532c4fe680487ee092efb48f : SUCCESS in 59m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0ee4413cde74b9b9572c97811b2a262 : SUCCESS in 45m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29a608f9b40a403680817b8c68514193 : FAILURE in 40m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/213beeb52fe34401aaa5baa8c954d2fe : SUCCESS in 1h 44m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d9523bccf1e48198195ce9a8dac0264 : TIMED_OUT in 2h 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e2528ba08b22496e9b870db4aa55525b : SUCCESS in 1h 53m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a951a95f0cfe4734901106727ce4548c : SUCCESS in 1h 58m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cb9b119589f84150b94fd931d846829f : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/941087b7d01d4aa38a8ff6f400e20224 : SUCCESS in 1h 55m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py","accounts_in_message":[],"_revision_number":1},{"id":"6c78126bb61659f2c9280d4cfec2f3e2a213de59","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-03 14:12:02.000000000","message":"Patch Set 1:\n\nI just added Michael to https://review.opendev.org/c/openstack/octavia/+/776926 since that pylint pin should fix the pep8 issue.","accounts_in_message":[],"_revision_number":1},{"id":"1f2535c370133bb19f6435d3d6bcec805b971ec5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 04:14:28.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce84c22ec352314d0ec4d8f875f259fdc61e5b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 06:23:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72f9eaeef83844cd85baccd63b0650fc : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7f9a6a2dc0341b584ad383ac14c15da : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852d539a48954a27b60ca22b2a9feef7 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7533e3d14dac4e9986f8e3c2bf0268b5 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e6eea25bd25400687059959eb1cf74e : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf3cf2cbe4f4602a755980b3931ec8b : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00d79828163042d68104fd845c8cfaeb : SUCCESS in 5m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/664eaa613d154b2da68f2eb03b74b9af : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0c5626816f2c49f6a714d3c457e188f2 : SUCCESS in 8m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4933f7227b9e494bbdaa4389ec749efc : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fe658d5c8d2405d926aa34e775ae228 : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50ad39cc0faa491cb28d179097148a60 : SUCCESS in 1h 26m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/451ba10cffe6443182ad8a639671236a : SUCCESS in 1h 34m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77128c5bbb1445b1a6052b3c01562712 : SUCCESS in 45m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/86e6cd0393434620844550243e26e347 : SUCCESS in 41m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf52cb629812404593089d4d270f753a : SUCCESS in 43m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74c126bdfe974a139dc9a0c453a1a179 : SUCCESS in 50m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a4e7a029010740998e6c93331da23aed : FAILURE in 1h 43m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbf88057ac1440df83337dfb8b721aa0 : FAILURE in 2h 03m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b1255845c6f54343b8aac509f0472af5 : SUCCESS in 1h 31m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2a3134b6bb064892b34a8e845f06e30c : SUCCESS in 1h 44m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b94b535ab7694c21b20d200e390817ef : SUCCESS in 1h 24m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2ef90138de1549cabedb44cc2e563310 : SUCCESS in 1h 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99720ee3665bd6b9f5962844572ae63d266db3c8","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-03-05 09:16:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a5e2eb88da7d79a25282220d1a43a2fde457459c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-05 11:03:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5b41e1d4036a507d346b360e23bbc4ef7c22bcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:09:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59e4669f42d47da8e18bc41b2ecadd7bf6992a30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:09:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea242032a64c4175903a7e0439bc426f : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2acd0722e18949b396bc0fb6b046297c : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f938188703ca40a49ac1507b291be2ca : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba7f38f7ae5494bb3a1dc96d40004ed : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3237dbe5c844dd5803aafe76b770d9d : SUCCESS in 5m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/15e706cd49f24a418ae6f43799320def : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a752dbed212c4476a063e2ed32cea56d : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28a336df69bb48338284f4b9dd8aaeaf : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e39a23f12054d75b47fb1c1ac76aa46 : SUCCESS in 1h 50m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/936c72f87ad34b789da23c0e735372ee : SUCCESS in 55m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/174b2bf423e04976bc93e117b63d5f96 : SUCCESS in 37m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b1a5e8ed666464abb516423e5682857 : SUCCESS in 46m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69b3497b404f44d48ad72332ed6892ba : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8c3b98eeec023bff6e4e8d944981f502c3b8a9a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:09:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4502414dbd80d445c60a25d02873e22c3a57c4e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:15:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d018d2a495f545d7bb604a8df21eb332 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2173a538a8594c2984ccf67c587589cf : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf7632e65b8458ff5e2e5dbf739299a36eac3d26","revisions":{"bf7632e65b8458ff5e2e5dbf739299a36eac3d26":{"kind":"REWORK","_number":1,"created":"2021-03-03 07:47:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/778377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/778377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/778377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/778377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/778377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/778377/1"}}},"commit":{"parents":[{"commit":"b5c6c884a4a4de48eea7ea08ba865f08fbfcf972","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5c6c884a4a4de48eea7ea08ba865f08fbfcf972"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-26 13:48:40.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-03 07:47:18.000000000","tz":60},"subject":"Fix overriding default listener timeout values in config file","message":"Fix overriding default listener timeout values in config file\n\nThe default value for timeout parameters in the BaseListenerType was\nnot correctly set because the class was defined before reading the\nconfig file.\n\nStory 2008666\nTask 41953\n\nChange-Id: Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88\n(cherry picked from commit b95fbe9ed4d8585e72bfc4f93aefb3d3ca1ab932)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf7632e65b8458ff5e2e5dbf739299a36eac3d26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf7632e65b8458ff5e2e5dbf739299a36eac3d26"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
