)]}'
{"id":"openstack%2Foctavia~778378","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88","subject":"Fix overriding default listener timeout values in config file","status":"MERGED","created":"2021-03-03 07:53:04.000000000","updated":"2021-03-05 13:13:34.000000000","submitted":"2021-03-05 13:06:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778378","meta_rev_id":"08a5baa95833a4db70aa0eb31bee84da55c13e92","_number":778378,"virtual_id_number":778378,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-05 09:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-03 14:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-04 00:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-05 09:25:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 10:33:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 14:12:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-04 00:43:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-05 09:25:58.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"d016387d757e4db5460cf84d52eeae63adc3d80c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-03 07:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70a55b66d1c027f3c25173746a3ee7dcc413734c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 10:33:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d99c90fac10849e3b01fa75e39a84acc : SUCCESS in 11m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a026f5d8f0c940fb9ed0a62e20d4c103 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246d8d59138b4db483bfe3564e10f8f8 : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00e5c173d48b4cf08ba86584f1c56450 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9431f61dd4b84477a6937a783b6658ac : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb74c72f698428bb3f6cf62a764d24f : SUCCESS in 14m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caaf0d2f934b46cf876bc6f64c2a144d : SUCCESS in 6m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ae19a5dbfef346f69509496316168b89 : SUCCESS in 8m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6d535228a12745d5a05985117ac64ddc : SUCCESS in 8m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3992880708b94a5fa1743b86612caa8f : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4191e7e57b5646799dae7ed6dccf8ccc : SUCCESS in 1h 11m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecbd6d02a2214f15b5bee7734ffa5606 : SUCCESS in 2h 11m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e61d7c582fd42bab134df3bcede4993 : SUCCESS in 49m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eff6d666568047289758fcda0cbc78d1 : SUCCESS in 38m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f7f6adcd22d4015884775a43b0379c0 : SUCCESS in 51m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54a77ea6e7864238a1753c568e2238d8 : SUCCESS in 1h 22m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/17b197c135ab43e9b13369ec6a366cf7 : SUCCESS in 1h 49m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc653038dea54c59a96daaf9f85b7305 : FAILURE in 52m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a9497d6d90434454be3b903f93046df9 : TIMED_OUT in 2h 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02be23458a17c6f68e15c6b26eb243a8ffb3ac93","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-03 14:12:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a9e3e2153aa41079bbb259694e588311dbd18bde","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:43:57.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4068f0cd6bb9f658a688e8aeb2ede1e3027bcfb3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:25:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15eb441f4cebaaefe52f514ad79efa3f1db5c777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 09:26:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32deff949b4eafe01e661790c3457b8ff427fbfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:05:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0783437d7f2a4661b3971ad55314f2f9 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ddd42229ef540bcb840b3f3060d80b5 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8d7ec56794749c09597159c4ef2f1b9 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d258bc5f82f462e9ab2722d5a830352 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3188b94e14b54d4983f8366de0e3bca4 : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05a79491923f4517a861e26330a82bd3 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3d40d56dc844440938b31839c71f935 : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f49dd4e2043a4afd89d2f2fad57851ba : SUCCESS in 1h 23m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b72272a482354b109b3457739098acfc : SUCCESS in 2h 01m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9b20053cdb0443ae905ad2ef88d80a16 : SUCCESS in 48m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b3520e7398e145a5b7e9ae56ea86e786 : SUCCESS in 54m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a5800482c454b82a3b02bfa450257f3 : SUCCESS in 45m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3b5d8039bb04fbea1930c8568cb0f26 : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bc88543292490627fc970e47c09ef6529f4ec518","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:06:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"08a5baa95833a4db70aa0eb31bee84da55c13e92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:13:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d630f4078fdb4998a6960619117930b9 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57f24751eb274d118f6d557cbc17cbf7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27230ba862cb0250463a4a2e04e96866681170f0","revisions":{"27230ba862cb0250463a4a2e04e96866681170f0":{"kind":"REWORK","_number":1,"created":"2021-03-03 07:53:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/78/778378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/778378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/778378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/778378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/778378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/778378/1"}}},"commit":{"parents":[{"commit":"4b06bd1c4b947e8b63d1d45a6de20661cec32542","subject":"Merge \"Fix AttributeError on TLS-enabled pool provisioning\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b06bd1c4b947e8b63d1d45a6de20661cec32542"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-26 13:48:40.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-03 07:48:34.000000000","tz":60},"subject":"Fix overriding default listener timeout values in config file","message":"Fix overriding default listener timeout values in config file\n\nThe default value for timeout parameters in the BaseListenerType was\nnot correctly set because the class was defined before reading the\nconfig file.\n\nConflicts:\n\toctavia/db/prepare.py\n\nStory 2008666\nTask 41953\n\nChange-Id: Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88\n(cherry picked from commit b95fbe9ed4d8585e72bfc4f93aefb3d3ca1ab932)\n(cherry picked from commit bf7632e65b8458ff5e2e5dbf739299a36eac3d26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27230ba862cb0250463a4a2e04e96866681170f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27230ba862cb0250463a4a2e04e96866681170f0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
