)]}'
{"id":"openstack%2Foctavia~778379","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88","subject":"Fix overriding default listener timeout values in config file","status":"MERGED","created":"2021-03-03 07:54:31.000000000","updated":"2021-03-04 10:25:09.000000000","submitted":"2021-03-04 10:20:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778379","meta_rev_id":"26e30512d9dd727290bd5490d95ea35d0db6cd53","_number":778379,"virtual_id_number":778379,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 10:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-03-04 08:23:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-03 14:13:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-04 00:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-03-04 08:23:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 10:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 14:13:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-04 00:44:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-04 08:23:34.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"5e5f1bf1b69df9b9bf84351e6540d2d9193b458a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-03 07:54:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9563f93c3e43ef2a9d9cd736adf6bb7c4fdfa83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 10:36:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/499f0e948386480aa21cad4be8b05a2c : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13233ecf9d7f41b38d512354f3b71cca : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1459c9b6704aac856a842256fa59ea : SUCCESS in 9m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6f5f45e8dc4430086f7f1c11430a709 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ec6cdaad2d941759a9d31a20074d05b : SUCCESS in 11m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/420c6b932cd84d4aa9767252809693a3 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c7681c71c94d398a950c711cbab262 : SUCCESS in 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/801f9ce6ba7e418d97ff603585f9087d : SUCCESS in 6m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5421cec74f1b4468bc7428272b7c8af5 : SUCCESS in 8m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4312fe5e2025466a919393eafbfb4e17 : SUCCESS in 11m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6cabb35af5f04cfb93a05abe7a8580c8 : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb3dea2f26974341b8e4dab22d479e41 : SUCCESS in 9m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f248779841524cd8b0d0603ae8aae329 : SUCCESS in 1h 20m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/421b2433e1d743a6a78bba359d2579bf : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/801f40a5369143a89e90bd54a2990c06 : FAILURE in 2h 29m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/71102f49e75a427da8a8973868f1e973 : SUCCESS in 2h 20m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8abc5546807c4ea898f8a7b054b8586a : FAILURE in 10m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1d1624fe39bb47659d5cf72d49f510b8 : FAILURE in 48m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7219a412f33c4868b044fbfda736895f : TIMED_OUT in 2h 32m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f772b65ead642ac9dbc01553be4aaf3 : SUCCESS in 41m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/903a9c75681244c8a0688482692313d0 : SUCCESS in 2h 14m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/565c702be33d43eaa9c9bbf4186ccb2d : SUCCESS in 1h 10m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2a00a93ff9bb490daee174308f1f6118 : SUCCESS in 43m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/deac4d925cc64ceeb758dba636e2492e : FAILURE in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0344e77e06dc86626778b9f6928799128690bd9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-03 10:42:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e6e7723e8030dc09dc8ba2a5878702a288981ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 12:40:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00d7a9256be34557ae22162de40a5818 : SUCCESS in 13m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c86e758ffe174aedb3519be87553080d : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2958ee263de49a6a5a8dd1dd22a4e9d : SUCCESS in 12m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48e290d06c294681a4260fe2ef204852 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51fabe0a5b6843db839ca3a199304c87 : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41295efc73224e61851ad1ea12d71048 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd4496304ba4222a8c26a9ff48c7e04 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb57b955d7b24a4c8227d967e9595ab2 : SUCCESS in 6m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7160312fc1cb4ff086a19353ad439d0d : SUCCESS in 7m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/624612a50b5040c999b76973044c0174 : SUCCESS in 10m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69db353e3e474002a0981eee2c1fbfe3 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a8bd5c326bb4f35a52da0cf77c5a65d : SUCCESS in 13m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/afe1facef41941a1b9245a3a244f69c1 : SUCCESS in 1h 18m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3d04ffa16d184ef5be64fd1c6e32762a : SUCCESS in 1h 39m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e07dbac1b354ce9911652a462d4673a : SUCCESS in 1h 48m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bbcef5b7aa214afe85ac1023239bbe58 : SUCCESS in 1h 54m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/87b4f8eda6134dd7991f00f7a40b4453 : FAILURE in 11m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6897c4cab294192b5bc35b0ffa71da1 : FAILURE in 41m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8f0eceb939e64a358e4beafe3892d76e : FAILURE in 1h 01m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a3e2e36ab3c4bff802223fb7d8907ab : SUCCESS in 40m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/421e639579b2427f844db67ab55fd2aa : SUCCESS in 1h 45m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a5517b45c444a438ea3b54f398b6efd : SUCCESS in 40m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d3b1dc5187c24898b2b23ac7ec431052 : SUCCESS in 42m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ad362765fa342e49fb05ebfc544be79 : FAILURE in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e682a464f6c2f30d4695b8acbcd6adee5bca3d7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-03 14:13:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c3c4b5067fa60604e56b5cdbb6d6a7411babe00e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:44:31.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"d19cfd6d4350c39dae317c24cfaad91f3e9c9f87","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-04 08:23:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3865b3b9a80bd8ce90aa8772b79adeb70ac71a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 08:24:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"596b4cfb18d00e51269800ce637402b783e27ef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:20:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c77f6b703cb54a5b843cb1169dccb92d : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/751954f62cb34a97821b4f1cd22a29aa : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b84f873bb9a42629623bf592e9079c3 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bafd3c179fb49239db715a252ab1dd5 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72e5dcc0ad5d4961bf9e315fd771f086 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc6b1d6084942968e38a80bb1718979 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eb4967a6a0f450c9808ede20f325df2 : SUCCESS in 7m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d2cf1f72b174424b87d393d0cce191c : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee219df85001483184b000b684b29948 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3eacaebcd4b24695956bfa2eb361cf1d : SUCCESS in 1h 15m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f3b3a66508974d9a90b1970932e73d70 : SUCCESS in 1h 46m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61417ed64eaa4d65bab7337f3eeb70af : SUCCESS in 1h 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e639079632a0bee06e35bb53a687bbc208da4759","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:20:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26e30512d9dd727290bd5490d95ea35d0db6cd53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:25:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe1b0ab174449d1a95f56ecb8daea6c : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe09f96d687e4c328480a077a3c4e8fc : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b357237aa9ec7ea7128122d350e92a6eeee1f10","revisions":{"0b357237aa9ec7ea7128122d350e92a6eeee1f10":{"kind":"REWORK","_number":1,"created":"2021-03-03 07:54:31.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/778379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/778379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/778379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/778379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/778379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/778379/1"}}},"commit":{"parents":[{"commit":"adce3dce8a017f30c2214df4802dd9b6bd79cce6","subject":"Merge \"Support deletion of all tags\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adce3dce8a017f30c2214df4802dd9b6bd79cce6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-26 13:48:40.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-03 07:53:47.000000000","tz":60},"subject":"Fix overriding default listener timeout values in config file","message":"Fix overriding default listener timeout values in config file\n\nThe default value for timeout parameters in the BaseListenerType was\nnot correctly set because the class was defined before reading the\nconfig file.\n\nConflicts:\n\toctavia/db/prepare.py\n\nStory 2008666\nTask 41953\n\nChange-Id: Ia4aa2047a79ad6fc3e33c7ebe2da9438914f7a88\n(cherry picked from commit b95fbe9ed4d8585e72bfc4f93aefb3d3ca1ab932)\n(cherry picked from commit bf7632e65b8458ff5e2e5dbf739299a36eac3d26)\n(cherry picked from commit 27230ba862cb0250463a4a2e04e96866681170f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b357237aa9ec7ea7128122d350e92a6eeee1f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b357237aa9ec7ea7128122d350e92a6eeee1f10"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
