)]}'
{"id":"openstack%2Foctavia~778411","triplet_id":"openstack%2Foctavia~master~I82c99f35f957c87ed9f08b7bdf88a5b0e212547f","project":"openstack/octavia","branch":"master","topic":"ovn-devstack","hashtags":[],"change_id":"I82c99f35f957c87ed9f08b7bdf88a5b0e212547f","subject":"DO NOT REVIEW: Test OVN devstack module","status":"ABANDONED","created":"2021-03-03 11:07:28.000000000","updated":"2021-05-20 08:41:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb2cf3fe747cf8a219bdff4c5216ebb0f2a14ec4","_number":778411,"virtual_id_number":778411,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-23 14:48:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 13:53:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 17:28:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-04-22 13:34:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"d92709fc6b2ceda6939ce0f2210a08534c58ba29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-03 11:07:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f37aae03184820069e3a677f239f8536ac3ccc55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:53:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3113b49852d412fa7cb2a5298495d1b : SUCCESS in 12m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b1ba15014564f44b2c59a84ab4276f1 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec154f245fd340d78d13e2f116cc1bc7 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/955021f2981b49b28ab17e61c8bdd97b : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bab9fd0250049358da0edecf63219df : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7070de532ad4850be2c7cc15630ee43 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e422cab2004b148f2445e8020eab4a : SUCCESS in 14m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ed950787b06f4ddd9c7e1082214b8a9d : SUCCESS in 7m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/36988b8a12bc44ab84a56b0fe2ca6098 : SUCCESS in 9m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7131ce7df24e44ceb484874336b27058 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2af170d5f5824e9b83be01bd2cfa9ef0 : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4dce53ad74649c4a241e8f996219b56 : SUCCESS in 1h 25m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3124270c47c040189a52bb277fd40b24 : TIMED_OUT in 2h 33m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9a71eeb98d348a6860dbf6014b3a6fe : SUCCESS in 49m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a38f789054eb49428b09e75116c80ab6 : SUCCESS in 51m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6021095fc4d4b12aa18a7f067d1320b : SUCCESS in 43m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/962f56b36eb84afabb64a31906a748ea : SUCCESS in 1h 05m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34bcbfd86c7947e0a5d4e629d646f25f : TIMED_OUT in 2h 35m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5c32c9eb3f74429f8ac5194f89f945ab : FAILURE in 19m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/341a92bde05b41598b0b786845dce5a5 : TIMED_OUT in 2h 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2c21cafbfec74e74bad7084f6e6bb5fe : TIMED_OUT in 2h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/deb8f4b06785482b907557fd5b1ead47 : TIMED_OUT in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/57baba3da5814fc0a3302c67583498da : FAILURE in 38m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d21c744931049b9346aa98390559eb734a1f45cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-15 17:28:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"326aad15dec11edadf9397d96f3355568426ac52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:56:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e95d03e0e4e48069ae7617a3fe60cf2 : SUCCESS in 12m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb0810f11d52469194790af39ab7f32c : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb6559387714f45b9ddea6db51050a7 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c7ab76589b24f7aa01e01d01066f78e : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf8a91431c9145bb837754462a3fbadc : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b54bd60f8bd4aea94e4cb3728201430 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e71007c02b43789997c2aaa3c11d17 : SUCCESS in 13m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/02bd308849914794bffdfe628491c495 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1739863cc20c46c3bcd35f448ea4fcfd : SUCCESS in 9m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ba466664e3694e91a283c5908a45ff6f : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2235c2dbf7f4a779c8f4b4e01c5180c : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a375e4a6c62b4dc1ab52ad1af1945ad0 : SUCCESS in 1h 25m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26add61ea8184d81b110a21065ad92c7 : FAILURE in 2h 24m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d99c8942e0248af97854e8002569155 : SUCCESS in 54m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e59204e851354c30a22fbb55e313a606 : SUCCESS in 36m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/044f0619d23f4078a0a0b159f10233f8 : SUCCESS in 53m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cf747252efba4bdbb0f383b8339296f7 : SUCCESS in 1h 00m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30113cb8032d4f07bba76bd7ae877639 : SUCCESS in 1h 38m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/124af7155f4e48fd94701907985e1d2a : FAILURE in 16m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e6d659b4576942f28638f0f82dcd368e : SUCCESS in 1h 41m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7fa38f0925b04e099b709b78820aed38 : SUCCESS in 2h 07m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f19db3611d9b4fb7ac423314bba1185c : FAILURE in 2h 12m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/733d1da6ab314e8b8d29b0cf41b9fa5a : FAILURE in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3e40e3cf8514d4c6f3c98e2b69cd31fbf685841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-15 21:39:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd3257bccf29dbc082c7abd7fa2e550780ebe5fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 00:16:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90cf3cda2ffd421abc50fb5dee1c6b56 : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ece5917554d4640801df1782279b7b4 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5094b68f84ab4683a103b71b385588a5 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/982b9230c98045c8a0376f3af5d2202e : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ef7c67c90ea420780b0fd63ee68dacc : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a03d8a98bd854eb083532be155de0dcc : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc52856c495448bb22a84e3dc954333 : SUCCESS in 12m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b672312d0b5142798d554ac55560167f : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0a724dc163ff44b7b2d1e0ca6741b910 : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1dfebf9497b14cd890b602f9ffec8f1c : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a707c052ecee465faa4929ea3ffce9a2 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48dd61247c1f4b759b734fc3020e39c0 : SUCCESS in 1h 32m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b514b1ddeb6b418bbe8ee19312606093 : TIMED_OUT in 2h 33m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7a5d974a7e3483b9982a42669b7cb18 : SUCCESS in 1h 15m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/23f62c69450b4804817e491cc9df45dd : SUCCESS in 44m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24c09ef0ce8f447f88fe09eb385b54b1 : SUCCESS in 1h 03m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1be4b1589e8b454994aa2ddca5418940 : SUCCESS in 1h 01m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5322b32c15014e4f8203335f716192d2 : SUCCESS in 2h 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6f71b63c559846669620b63599697ce5 : FAILURE in 29m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3de72847df5844d18658c3126cea065a : SUCCESS in 1h 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e83d2d888a1a43d387be777f447bd9fa : TIMED_OUT in 2h 33m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0b549490dc614f4ba6f7058f013acb06 : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0f7bc0b0766f4de6b3737176d99ce02d : FAILURE in 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"27b0b4582f213f667a65bc723010e23e387b244e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-16 07:51:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bf80459fe53bbe58e45b097977542fa1a2f29901","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-16 08:28:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a8cd1719967f2425dd5c3a4ad498707e0f00965d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 09:10:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e106056fcb3c44b1a223f2592c9f9b58 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7461f70e80834baf955fbca51146bf21 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42087d6b0c6f47dd88ebcbc5e227fcb3 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef639dcc09254918a6e53d91f3a08070 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a559e1c5254400becbc0250128077f : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/773ac42a287a44c3bb3d9ffe56ea4995 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/945305c5a49a4770bbd2c55d19d4cd7a : SUCCESS in 12m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/00aef923c7e04d5995d360ffbdc2393a : SUCCESS in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/242964f4eb624968b8f11b39137447e2 : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de88094b533a4a6281d2a13dfdd720ff : FAILURE in 30m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c1b586d2f4a48d281c6eb8dc5e6e91e : FAILURE in 39m 00s","accounts_in_message":[],"_revision_number":5},{"id":"9ce1a943f7c526a62fc19e7831270056fe24bb4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-16 09:23:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"87396a95b6fc2c355680b4f07c681004abd81143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 10:06:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a30c1fbdec674a058c8c182018bcf317 : SUCCESS in 13m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a62ca8e00c974a91b8cfe0aeba855053 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6b9d86cab04626a7c9183d7e114d00 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6398d457b73a4dc4b0eab4cd068115c4 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fc9f637bbac41779fd5ada5bdfe3f2a : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe5e6c6a3ab24f96b545c52564c892fa : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf1d438991641b4aab78cefdfaa5e39 : SUCCESS in 13m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c57e56615220491ab8e181c63c4dad12 : SUCCESS in 7m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b64fcbe9537f4b039ff27ab39519c9c0 : SUCCESS in 9m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/032e42cbe2bc49ec8361fe4bff231d65 : FAILURE in 35m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/616df59b7ec54dc8b075de555cc75420 : FAILURE in 31m 24s","accounts_in_message":[],"_revision_number":6},{"id":"7ba97b812c7d989eba558fea245325cda525e586","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-16 11:29:25.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8769239e92ec36f0e69e725b0c87bf15b0e6bb7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 12:38:04.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"140c8cef4d4645e6c9a547cf5338268848321c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-16 12:54:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"46bc2f31630e619e66690332051b18cb14103d78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 13:57:45.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"20654b5d3b6ad810d2832603c8ff838535627304","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-22 13:34:19.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"ab1b739fba7eef1bca6bd12593f32a52724daf3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 15:17:47.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abcd5586f3d043ec92f4ed3cc72e6d1f : FAILURE in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cda05ace389742cd9ec0a2fe49208909 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda0dc097d3842c9bebbdb2d5870e498 : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e80b6a05362e4195a5b052e257005d7d : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/035767c8e97f4d8bb4a368ad9026f683 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdc123563c434bf29bc3e8349ef6f474 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1735e8fcff89489ab15d242e03378500 : SUCCESS in 13m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9a58f06f528f4ce5867d13e6d62f9109 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d553ee85b9a9417bb29d43993d7f95f9 : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0bd9d0f8e4746f8b051dd4e75345e26 : SUCCESS in 1h 34m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ec6579702e541ddb187fbdc71208fa9 : SUCCESS in 1h 40m 05s\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":7},{"id":"9f01c615480e9cea6620b5a41ada6dc5c130d421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-23 13:04:16.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"314111fca466aa26fd3083743a8b4cc488bb8f41","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-23 13:05:10.000000000","message":"Patch Set 8:\n\nJust rebased to make sure it\u0027s good.","accounts_in_message":[],"_revision_number":8},{"id":"851e9e2b8ab5f9d0b2d5741679475678aafe5be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 14:48:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8a989dea1b4bcc84e82f5bebd80b1b : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/081c2a11884d4b068c9418234957a507 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528a4451d41941629f6c0f5dd6a2af8b : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4311fb8894614190b952811821f125f7 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeb183d2c1734d6c857fee96f9cce614 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52f7000a94a34d6aa2b6247213c16d86 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba96aa111eb4300aea076f1510e5f3c : SUCCESS in 14m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b841fe36a2914a7090f7423bb704d496 : SUCCESS in 6m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d7a87be4ea50453685698e79810caf5e : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6fe1bc36b6f45fab5742edcda3507ac : SUCCESS in 1h 38m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7f5ee4ce4dd4a4b95077c5fd74af2c4 : SUCCESS in 1h 24m 25s","accounts_in_message":[],"_revision_number":8},{"id":"fb2cf3fe747cf8a219bdff4c5216ebb0f2a14ec4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-20 08:41:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e0dfc083075555c5314d55ac77420766ea8756be","revisions":{"d3056ae01cd342370fcbc8707ab765a51c3cec95":{"kind":"REWORK","_number":1,"created":"2021-03-03 11:07:28.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/11/778411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/1"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:07:25.000000000","tz":0},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-On: https://review.opendev.org/#/c/735097\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3056ae01cd342370fcbc8707ab765a51c3cec95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3056ae01cd342370fcbc8707ab765a51c3cec95"}]},"branch":"refs/heads/master"},"25e5c1eb76ec3cfa8b1de953dae7a37def38e70e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-15 17:28:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/778411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/2"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-15 17:27:32.000000000","tz":60},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-ON: https://review.opendev.org/c/openstack/devstack/+/780628\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25e5c1eb76ec3cfa8b1de953dae7a37def38e70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25e5c1eb76ec3cfa8b1de953dae7a37def38e70e"}]},"branch":"refs/heads/master"},"51ffe3999fc780bd53679adac43e40285072e544":{"kind":"REWORK","_number":3,"created":"2021-03-15 21:39:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/778411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/3"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-15 21:38:20.000000000","tz":60},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-ON: https://review.opendev.org/c/openstack/devstack/+/780628\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ffe3999fc780bd53679adac43e40285072e544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ffe3999fc780bd53679adac43e40285072e544"}]},"branch":"refs/heads/master"},"27af369e733963090ffc44120d02c3c89eccf5ce":{"kind":"REWORK","_number":4,"created":"2021-03-16 07:51:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/778411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/4"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-16 07:50:44.000000000","tz":60},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-ON: https://review.opendev.org/c/openstack/devstack/+/780628\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27af369e733963090ffc44120d02c3c89eccf5ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27af369e733963090ffc44120d02c3c89eccf5ce"}]},"branch":"refs/heads/master"},"239e20947912fbdbe5dadff015f7a740c5e5f264":{"kind":"REWORK","_number":5,"created":"2021-03-16 08:28:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/778411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/5"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-16 08:28:34.000000000","tz":60},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-ON: https://review.opendev.org/c/openstack/devstack/+/780628\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/239e20947912fbdbe5dadff015f7a740c5e5f264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/239e20947912fbdbe5dadff015f7a740c5e5f264"}]},"branch":"refs/heads/master"},"3f83d94249b0624737f7db3c997fd63bdccb97be":{"kind":"REWORK","_number":6,"created":"2021-03-16 09:23:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/778411/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/6"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-16 09:23:09.000000000","tz":60},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-ON: https://review.opendev.org/c/openstack/devstack/+/780628\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f83d94249b0624737f7db3c997fd63bdccb97be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f83d94249b0624737f7db3c997fd63bdccb97be"}]},"branch":"refs/heads/master"},"9d8fb000d187f674ef869e1065ce51f254d99585":{"kind":"REWORK","_number":7,"created":"2021-03-16 12:54:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/778411/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/7"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-16 12:18:24.000000000","tz":60},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-ON: https://review.opendev.org/c/openstack/devstack/+/780628\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d8fb000d187f674ef869e1065ce51f254d99585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d8fb000d187f674ef869e1065ce51f254d99585"}]},"branch":"refs/heads/master"},"e0dfc083075555c5314d55ac77420766ea8756be":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-04-23 13:04:16.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/11/778411/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/778411/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/778411/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/778411/8"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-03-03 11:06:44.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-04-23 13:04:16.000000000","tz":0},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-ON: https://review.opendev.org/c/openstack/devstack/+/780628\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I82c99f35f957c87ed9f08b7bdf88a5b0e212547f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0dfc083075555c5314d55ac77420766ea8756be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0dfc083075555c5314d55ac77420766ea8756be"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
