)]}'
{"id":"openstack%2Foctavia~778427","triplet_id":"openstack%2Foctavia~stable%2Fstein~I7b8f7c7d653b37395f9a660be67f954a3a6f26d9","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I7b8f7c7d653b37395f9a660be67f954a3a6f26d9","subject":"Fix neutron subnet lookup ignoring endpoint_type","status":"ABANDONED","created":"2021-03-03 16:36:37.000000000","updated":"2021-03-03 20:18:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":738265,"cherry_pick_of_patch_set":1,"meta_rev_id":"a90b53e4bff004e958f4a4c3729c7734cbf57826","_number":778427,"virtual_id_number":778427,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-03 18:36:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 16:36:37.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-03 18:36:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"792626200138bc7b53f6139b7b33c3797fdb65e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2021-03-03 16:36:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"303bd68f6ea37cac4d19c7ccad9b58bd38922b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 18:36:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0f4e205b90c43adbf2793d7af6c5163 : SUCCESS in 13m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d0b9007c187461b976140b6ad2a27b5 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2efc826d6004440dba2ca5b3b229d4c6 : SUCCESS in 12m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c07f0433cb3496c8cb717dd4a9564f2 : SUCCESS in 8m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/38a6db71ff8f457ca4da8446b1740b33 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49c8ebe9440d41c6ac4661ddaef88d25 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74ef0955d7d443cabb79a8c28bc0f26 : SUCCESS in 8m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/be357f876cdc4f1aaa7c7c13d3fa34b0 : SUCCESS in 1h 13m 04s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0e21631df1b34dc0bae99bd962e2ef0d : SUCCESS in 1h 08m 57s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac152270d0e9483d83647fbf49a0506e : FAILURE in 26m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/709e2c632350470eab236a4843abd73e : FAILURE in 27m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3492fed9b2604292ab7ca82d556283d2 : SUCCESS in 10m 28s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/859d9e8a2c8f4408b630301f9193bd39 : SUCCESS in 9m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4adbc63cd6204cf2b8fdf7dd61221755 : SUCCESS in 11m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92baebe0ccc3434fbfb0092ddf4ae8b8 : SUCCESS in 1h 00m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f354f16d295e4b859e0c42b3793cfc10 : SUCCESS in 54m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9f04d034d064014942f2cf4e7376cbc : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b1512877c64b4c4baa1515ce77490709 : SUCCESS in 1h 18m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/10b0a7e0d1b8438fafaf6d21b8cdb917 : FAILURE in 12m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/76bfb08d6450417da42235b827da8b14 : FAILURE in 48m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/eae14b9b3aee48df8919fb1881c76ac6 : FAILURE in 54m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3624a5dd02884131afa6a625409ed4fd : FAILURE in 40m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/868427302e0f4b908112ac700991d044 : SUCCESS in 58m 28s (non-voting)\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30162310bdee4458b60f25d9a35bc4bd : FAILURE in 32m 17s (non-voting)\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e25ad3fbf0df4a458224dea475e9a3c4 : FAILURE in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a90b53e4bff004e958f4a4c3729c7734cbf57826","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-03 20:18:42.000000000","message":"Abandoned\n\nThe Stein stable branch has been marked EOL here: https://review.opendev.org/c/openstack/releases/+/772847 Abandoning this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b606cd6ab3f9e4beb10f3353297ab43f5cc5078","revisions":{"4b606cd6ab3f9e4beb10f3353297ab43f5cc5078":{"kind":"REWORK","_number":1,"created":"2021-03-03 16:36:37.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/27/778427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/778427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/778427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/778427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/778427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/778427/1"}}},"commit":{"parents":[{"commit":"b248d8c12519cdd99e51f241c1b538a7fc242731","subject":"Fix getting user neutron client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b248d8c12519cdd99e51f241c1b538a7fc242731"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-25 17:05:54.000000000","tz":-420},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2021-03-03 16:36:37.000000000","tz":0},"subject":"Fix neutron subnet lookup ignoring endpoint_type","message":"Fix neutron subnet lookup ignoring endpoint_type\n\nThe recent patch[1] that adds a neutron subnet lookup using the user\ncontext is not honoring the interface/endpoint_type or region_name\nsettings for neutron in the octavia configuration file.\nThis is causing problems for deployments that use the \"internal\"\nendpoint for neutron and the current code will always return\nthe \"public\" endpoint.\nThis patch corrects this problem by including those filter\nparameters when the neutron endpoint is looked up in keystone.\n\n[1] https://review.opendev.org/726042\n\nChange-Id: I7b8f7c7d653b37395f9a660be67f954a3a6f26d9\nStory: 2007863\nTask: 40173\n(cherry picked from commit 40731e572f2b503a3ccc603642a3721a5ce1cf7a)\n(cherry picked from commit 2a18b6cc996e4c64d650c516071c57358681af3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b606cd6ab3f9e4beb10f3353297ab43f5cc5078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b606cd6ab3f9e4beb10f3353297ab43f5cc5078"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
