)]}'
{"id":"openstack%2Foctavia~778582","triplet_id":"openstack%2Foctavia~master~I4165631e3f2e050b6ca610407154a3ab8ad48e10","project":"openstack/octavia","branch":"master","topic":"get_listener_realserver_mapping","hashtags":[],"change_id":"I4165631e3f2e050b6ca610407154a3ab8ad48e10","subject":"Change return value of get_listener_realserver_mapping()","status":"ABANDONED","created":"2021-03-03 21:27:42.000000000","updated":"2021-07-28 21:03:22.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"7b3960da96cc7bd282dbb599e187be49da672a1b","_number":778582,"virtual_id_number":778582,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-04 00:17:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":-1,"date":"2021-03-05 09:45:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 00:17:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 09:45:47.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"d875ce14237a57feddec4b58df5441b97ba6cac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-03 21:27:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318048e00b07f064647ca45082fb802e2fff39f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:17:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d138c38ec30940ecb2c4c95831283977 : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58d24362590e4df88033f770981132d5 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87af9d870e8436b821ddbb3ea847387 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd725626a66e43e79cedd0fec51b34cd : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e069d1738ea943ff8762b4e1bd719810 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8683fc1371ff4953ac5358092c882846 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cef2e621cb743529df4771c71a44982 : SUCCESS in 13m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2daba96a7d5a4da89bdd779e8b65dc42 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4817440cd63e4f2d894eb1b75f0e6536 : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ebda8f586c44471e93179c2fb1e69d8a : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70123677b0d740b28c93ed097b7fc588 : SUCCESS in 10m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1ad93f95b2046a9a2bba12964db1cc2 : SUCCESS in 1h 28m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0aa95c0fb55a4c4099d938affce66549 : SUCCESS in 2h 26m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcc1af01933245409159f3bba15d18db : SUCCESS in 48m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e8b6fc7cfa514a1a85e22ad3379631a3 : SUCCESS in 45m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a0091566c134b80add2e2d784a923c5 : SUCCESS in 45m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b9ed7672169148ca96464922fec0e446 : SUCCESS in 1h 04m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4f571b335ed642d48f7c547658d9a094 : SUCCESS in 2h 02m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35283663d5c649ce99ef64fc2d721616 : FAILURE in 2h 31m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5a082c63e1f4a158dbaeef48c55eebb : SUCCESS in 1h 43m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2af86a69262a4d65aa3513d3fdefa92a : SUCCESS in 1h 39m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a56f959a7ce145d68a827dfd38c4831c : SUCCESS in 2h 09m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4d458960699647dfa33bda5d6302a978 : FAILURE in 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7c2a620b6c7e6b22b20edabaca1caadbdc23c1d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:45:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\n@zhaobo If you around, please, take a look to clarify.","accounts_in_message":[],"_revision_number":1},{"id":"3a15b5357c68d04410f52c2b0b8609181c999b9b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-05 15:18:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b3960da96cc7bd282dbb599e187be49da672a1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-28 21:03:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fae3961b2ebe3d01c51f35b6e0257d2d0e823737","revisions":{"fae3961b2ebe3d01c51f35b6e0257d2d0e823737":{"kind":"REWORK","_number":1,"created":"2021-03-03 21:27:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/82/778582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/778582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/778582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/778582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/778582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/778582/1"}}},"commit":{"parents":[{"commit":"c93a76b9f3d22bc131d899554c88ba1caf6dd249","subject":"Add HTTP/2 example to the load balancing cookbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c93a76b9f3d22bc131d899554c88ba1caf6dd249"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-03-03 03:06:30.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-03-03 03:06:30.000000000","tz":-300},"subject":"Change return value of get_listener_realserver_mapping()","message":"Change return value of get_listener_realserver_mapping()\n\nChange from a tuple to a dictionary since the first\nargument was never used by callers.\n\nChange-Id: I4165631e3f2e050b6ca610407154a3ab8ad48e10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fae3961b2ebe3d01c51f35b6e0257d2d0e823737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fae3961b2ebe3d01c51f35b6e0257d2d0e823737"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
