)]}'
{"id":"openstack%2Foctavia~778797","triplet_id":"openstack%2Foctavia~master~I1fe342d2bdf1301dd89ab7dfaa8e6a23e69c252b","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I1fe342d2bdf1301dd89ab7dfaa8e6a23e69c252b","subject":"Fix LB failover when IP addresses exhausted","status":"MERGED","created":"2021-03-04 23:50:33.000000000","updated":"2021-03-05 13:21:57.000000000","submitted":"2021-03-05 13:09:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778797","meta_rev_id":"d3d12f75e018d3c3ab14ab97732818769cea5dc9","_number":778797,"virtual_id_number":778797,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-05 09:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-05 11:01:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-05 11:01:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-03-05 09:14:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-03-04 23:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 02:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 02:11:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-05 09:14:53.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-03-05 09:22:16.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-05 11:01:37.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"0b95188869dc5b7af6485e22204ff88ee5e0b3a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 23:50:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55e9ba6a793d2b0da7345260fe6324f827a3e877","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 23:53:07.000000000","message":"Patch Set 1: Backport-Candidate+2\n\nThis should be backported as far as the new failover flow was backported.","accounts_in_message":[],"_revision_number":1},{"id":"626dd77afcee91b1d8a76f18e4bff670d85b3f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 02:03:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e38ffc84398847908167c6b63bf9a796 : SUCCESS in 11m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/230c8ff5d37b4015aee38a4287b0585b : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3885056942794e7e91037a1406e1c174 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/570838f59d254e1ca5c623f7132b47d0 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1599947749c54868a89ab9dd16fbc916 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb280ad8ae11474ba969f496cf5e25fd : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919b2197b0c949169c15968bc9f2e135 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b668c557ee945958ac778da11f5f5be : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/48606151c61243ee90de482c4711946b : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5b5d55652649448fbb17fcee449552c0 : SUCCESS in 9m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b2333702fa0f4828b8d138b511b53fac : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef48041beca148169bf4fe5bcd541075 : SUCCESS in 9m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f35c1e3ec7c41308728e4f97e180a16 : SUCCESS in 1h 23m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1d047c6f09744c89f5b5a2c3f1c3deb : SUCCESS in 2h 05m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/954287a334894fac8b20e50861342e28 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/354cf762027c478c9e11bf162aa07f76 : SUCCESS in 55m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91dc4c82d3ab4551a3ba22c92d314626 : SUCCESS in 36m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c71c66c291c4053a7d70fd8efc8aaa8 : FAILURE in 42m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ea4eff18b3984e058f0b004b22e02b73 : SUCCESS in 1h 25m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91bde3d368434a94a7b521382973d05c : FAILURE in 2h 04m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e82ee0c657284fd5b10447d1dbde344e : SUCCESS in 1h 44m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/26bfa6386b3f49d593a18c889e06eabf : SUCCESS in 2h 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/94e892679a5a48fbafcec7ee14f61476 : SUCCESS in 1h 27m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/21e96bcbdcb84dc4ac547cfb91451a92 : FAILURE in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3a715908dafb40f2891a7b33002b6cb492b84af","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 02:04:24.000000000","message":"Patch Set 1:\n\nrecheck\nunrelated grenade job timing issue","accounts_in_message":[],"_revision_number":1},{"id":"e7dde0bce7d3251865f655fbae84a9f5bf7a2352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 04:37:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4825a8344be41d2b4afb82e1a76e0b2 : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e002cde9df341e0b943067dfea9b3fa : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3dfe73d55e5473d98c62400f2f01c33 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4bb53e9f29441a8370dd3154102fb5 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c831be851c24ad48939feff4e24d6e9 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/573ab1532b524123b4007b2d2bfe98a1 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fd5b839ac74d4aae38e2d4c6be7aae : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aa6407f93574021a0f9099167ff17e7 : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d73eeb991d7944b9bc77a0336a11be28 : SUCCESS in 6m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/47476474e86140a8ba4066ff0668d386 : SUCCESS in 8m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aa8da27cf2464b5b9b1c6dc166a6ce46 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c61e5c54d43d48b6a4a80ece8a450813 : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b29e737bf734d2fb9c1ce7bc3307cbf : SUCCESS in 1h 09m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42928d81bede404b854a8c8f736858a9 : FAILURE in 2h 10m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3d2156cdfc74f6aad3c19ed4f2a4ff8 : SUCCESS in 44m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/76d02c5bed62458db79aa28379a5425c : SUCCESS in 46m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33abc43ece47429295b75ad53795f6c7 : SUCCESS in 39m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9d1e7a3d7a384ff1a2184e3552c808ba : SUCCESS in 1h 06m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7cb89583ea4e4681a76037716d240738 : SUCCESS in 1h 36m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/41c51086878447019fd9ec14f7ae3807 : TIMED_OUT in 2h 31m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/46ce55a049754931b55c8adf848d687e : SUCCESS in 1h 35m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1c6e55d702e44c6d9fad53f8a1d00266 : SUCCESS in 2h 20m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b29003e216ee43b49971521b216ec7ac : SUCCESS in 1h 39m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/239f2549351647e09e8fe69151fba089 : FAILURE in 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78c082ea55a31afef37e184005a0a3758c1213c1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 04:40:36.000000000","message":"Patch Set 1:\n\nrecheck\nIPv6 issue","accounts_in_message":[],"_revision_number":1},{"id":"9d85f733a3273f9a3ec0b2f401a197003905d50d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 06:33:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c70ab6f2cf34e8a9311bfdec062b202 : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b60b9f11522473cab19bbb4bd3a9d41 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ba035df14c426292e9bab98818d38f : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfc227e420d744e4867bcba197622f70 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46400512b25e4854826a23659f2e5729 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82d2fe1c1bd54f6e83e00ee49d145064 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ed12e4936e4d3cb71470376eaee1bc : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f3c1d985eb4b76a458abecf0cbe8ee : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/17939dffed8e4930bfbd655b76dc6cee : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f7437f81394481ab6365cc367a6600d : SUCCESS in 7m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/19ffed01d1f843d094aed5511fba1a0c : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/270356207ed8438189f35abc252f23ee : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f445d2e64ec44c4b8a9913f8ee81ed2e : SUCCESS in 1h 03m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b30534424f924469b34d25b00f6a6479 : SUCCESS in 1h 27m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8dc7e3e5fe264fbe8ce107a742135cb8 : SUCCESS in 38m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be34f2e3c93745f19726fbfba41a0635 : SUCCESS in 32m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/294586a480e04a7cb037e98e11f7769d : SUCCESS in 40m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/64d70025d05f495d93cc47a0a5965b22 : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8810d144e164ace90480392bbf38207 : FAILURE in 1h 51m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/828189f6d7074d2f8f874901cae3a360 : FAILURE in 1h 40m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/76ecd32b681a4eb594b9cf85bb388d02 : FAILURE in 1h 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f5800902521a465b9f441b4a7a9cbb2c : SUCCESS in 1h 26m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/40cc1cf3e180461b8b8552cdac5228c2 : SUCCESS in 1h 46m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8c1ad16703e54d08826e17b841c5b5a3 : FAILURE in 32m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ae7401f50da6ba9e359a58970c819d7dbbbfce4","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-03-05 09:14:53.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"010c8cce5c1bcca6482aae7effc0ca0f0402e88e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:22:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0abbf1552fd5f3dc990c38f8fe14016c8fa4d87d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-05 11:01:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecac960de0b78562b4df2ab9931c4c0c8155eaef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:09:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3f495e5303eecd473236b4eb2795720c0f00fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:09:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c554b3ee2b7345b2bb08f54d301473ec : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076ec3a7619f4c72b0ccf5f78e08f8ec : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b7f0c64b6e147189c8ce88da5ec8755 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97479fa133de48368fa62f89e1a6d026 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef3fd40727fd4beaa5b88a683257d39a : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23254cc78b764e01b387b10267f9c7b9 : SUCCESS in 5m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe8b003527ae496f9d7f59e32ae3b9d9 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1986ab2fdeaa4829b0ced8295ad286fb : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dbd47afb1301446da37ba1a5170b39b8 : SUCCESS in 1h 19m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9452f11d64c2444dbecc60c21d4cde72 : SUCCESS in 1h 55m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b344d88b207e4f0e80a0d14b4508fb1a : SUCCESS in 47m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/925ba7dc85554d36b1cfd7f244eec7c0 : SUCCESS in 41m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a919667de5a4755b41dd7dedccbfd23 : SUCCESS in 43m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/76e4543da54f49e09f4d6266e3bb1bc1 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9bfab612d6cf780656fd7742fd099d6ed0aff9f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:09:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d3d12f75e018d3c3ab14ab97732818769cea5dc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:21:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492ce34e55c640b591797685273375b6 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fc0dcfd87364d0b8781a67f5333f05c : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3ab49c60aa835cb4034568ea25f06472127388c","revisions":{"e3ab49c60aa835cb4034568ea25f06472127388c":{"kind":"REWORK","_number":1,"created":"2021-03-04 23:50:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/778797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/778797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/778797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/778797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/778797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/778797/1"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-04 23:47:15.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-04 23:50:32.000000000","tz":0},"subject":"Fix LB failover when IP addresses exhausted","message":"Fix LB failover when IP addresses exhausted\n\nWhen a load balancer failover was performed on a load balancer where\nthe VIP address is on a subnet that has no IP addresses available,\nthe VIP address may be deactivated.\nThis patch corrects the failover flow to not deallocate the VIP\naddress on a failover revert flow due to the subnet being out of\nIP addresses.\n\nStory: 2008625\nTask: 41827\nChange-Id: I1fe342d2bdf1301dd89ab7dfaa8e6a23e69c252b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3ab49c60aa835cb4034568ea25f06472127388c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3ab49c60aa835cb4034568ea25f06472127388c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
