)]}'
{"id":"openstack%2Foctavia~778804","triplet_id":"openstack%2Foctavia~master~I6505243ddbf1eab7d110e9bfa03bffda840f07ae","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6505243ddbf1eab7d110e9bfa03bffda840f07ae","subject":"Fix grenade resource waiters","status":"MERGED","created":"2021-03-05 01:40:26.000000000","updated":"2021-03-05 13:13:48.000000000","submitted":"2021-03-05 13:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778804","meta_rev_id":"db21aa8a1c93b2fc50a73adc0b6fec2ea2999f3a","_number":778804,"virtual_id_number":778804,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-05 09:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-05 09:56:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-05 09:56:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-05 09:56:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 02:13:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-05 03:48:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 09:23:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-05 09:56:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"f955bd2ac98524ebca561a544e3de0b8e4dc9d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 01:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b0b8fa87e779ee2545a0f44358fa854ce0529a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 03:48:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/251076f9b3794fa590ec5d2632343bff : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/069fa53fc1d3480085b7dd7c4de3dabc : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8a79c1b2a8446d9d0a946125b97438 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/235b0daaf78f4493be416e2ed3e5821e : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c5a1765b19341958b6e898aa9b56a95 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb5bc5efa8ea4495aa30516dfc867f82 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7ec82ed9fe40ec83d486ccf620b1bb : SUCCESS in 12m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1426852dd1df44fdbc03c2c4eac9c258 : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/734621110bd4411ea87949875e0a0139 : SUCCESS in 8m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d6dce44246ed44d2958df55ce5343a5c : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32bd7ac023aa419bb640378d907a6fbc : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3777ac51f51a4e9bb8776e7460666d27 : SUCCESS in 1h 07m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a8c88ebed1a4b1aaabe74eaf9ef6e8b : SUCCESS in 1h 49m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44309f79d3d946b7a7a4e35a36c9293f : SUCCESS in 40m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fd14b2be64cd4bb3b6aeced9f2cb6215 : SUCCESS in 39m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/094ffed17c85456082b9226d7d681d32 : SUCCESS in 40m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7cf03600491c4dd994535bf67f008df2 : SUCCESS in 52m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7cfac64e037b42939a3fe4acbb4e4ec2 : SUCCESS in 1h 46m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b3d2920348384fe99968a3ff1cafc0da : FAILURE in 1h 52m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3a647f9713dd4e1ca95fb65c06eb8ea8 : SUCCESS in 1h 35m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4c07e4edd06e4140851dc7c8840814df : SUCCESS in 1h 39m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6371fe404fc640f3b2dead4cbb959d97 : SUCCESS in 2h 07m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/38e78b6e8ba64c2587e031b9fdd5ddf6 : FAILURE in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be3c39bd04041bf7b98b8e9fe8ecb09d2c2144fe","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:23:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a740da397c866a197066e2c55ea9c4c744621ff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-05 09:56:27.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"480de9a33aa2cc0cd9b6e59cfc21954eb27c2343","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 09:56:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7007ec1390b388f9629067554d902748ac140cde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:07:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8af64af21d444ccb998e6be740bcc43 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b78be2d5db54f1eb1068a432f9e2b49 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10540febd7cc4d6b80d4cae08231ad1f : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe01db84223946d087628ec7b67cbb0c : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ccdc32c15546fdae708f781ce5c2ac : SUCCESS in 12m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aa681d98cbd646bdbfcc57957c20e371 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7345b4fa3f54e9bbe6bec4f0c83f17e : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2340251fad44744952309844e37c96a : SUCCESS in 1h 19m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9c872503ee542b3b95192cc9572eb7a : SUCCESS in 1h 42m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d57cf18bf9f4206b600fa076c414a3e : SUCCESS in 48m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d9b46e59d354ec0b27d333fe2e06f7f : SUCCESS in 38m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e5098a580cf4e79aa4be3e066bb2332 : SUCCESS in 39m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e164b72860f64b70ab4c3c2725ef24fe : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"95fb3137021ae8f76f9e95a22fe37fad44101ba6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:07:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"db21aa8a1c93b2fc50a73adc0b6fec2ea2999f3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:13:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3f8154cbaa4bcc9e4ca873f4a09cb9 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d17b23282cf7c315251a37f39fa8d3fc35c05e3c","revisions":{"d17b23282cf7c315251a37f39fa8d3fc35c05e3c":{"kind":"REWORK","_number":1,"created":"2021-03-05 01:40:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/04/778804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/778804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/778804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/778804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/778804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/778804/1"}}},"commit":{"parents":[{"commit":"79268cc9e36ffdc4f9d73959166a52a67b9bf00a","subject":"Merge \"Fix overriding default listener timeout values in config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79268cc9e36ffdc4f9d73959166a52a67b9bf00a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-05 01:37:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-05 01:40:25.000000000","tz":0},"subject":"Fix grenade resource waiters","message":"Fix grenade resource waiters\n\nThe grenade job resource.sh script is waiting for the created object\nstatus but not the overall load balancer status to go ACTIVE.\nThis can lead to the script failing with a 409 error.\n\nThis patch adds a check for the load balancer to be unlocked before\nadvancing to the next create call.\n\nChange-Id: I6505243ddbf1eab7d110e9bfa03bffda840f07ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d17b23282cf7c315251a37f39fa8d3fc35c05e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d17b23282cf7c315251a37f39fa8d3fc35c05e3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
