)]}'
{"id":"openstack%2Foctavia~778944","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I17dd7ad2865fdc0b29308e482dd4ca9d385d4d7b","project":"openstack/octavia","branch":"stable/ussuri","topic":"oslo_lc_drop","hashtags":[],"change_id":"I17dd7ad2865fdc0b29308e482dd4ca9d385d4d7b","subject":"Drop lower constraints testing (stable Ussuri)","status":"MERGED","created":"2021-03-05 14:31:21.000000000","updated":"2021-03-06 08:53:57.000000000","submitted":"2021-03-06 08:46:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778944-oslo_lc_drop","meta_rev_id":"b9d22383fdf87744e2aada8090e6c59355a7a24f","_number":778944,"virtual_id_number":778944,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-06 08:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-06 06:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-05 17:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-03-06 06:55:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 16:50:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 17:56:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-06 06:55:21.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"284f2e3066094078bdecee713e7e90730a8bdb24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-05 14:31:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef06211e9759a988361a8b2b26a08f1fb95bdf3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:50:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa2a31c633fc4d2980c8e989044856b6 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/860686c92aad4b5db00197b5b1e5e92c : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a24d5dfffd941a3b50da0544086a1aa : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07089e43e39945ffad0da6f309c3f3d8 : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4695d83329aa44129d731368a7b40222 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad38179f15346f2858b48d2c8891ca9 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa1eb69a72d3499e8a5bbd3157d7837b : SUCCESS in 6m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/54a5fc1240944e3d8121ccf23d941e5d : SUCCESS in 9m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/055fa011b3124057ac1828d9f4382c8e : SUCCESS in 9m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6335c6c7c094341a74294009236cccb : SUCCESS in 8m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25281b4347ec487fbe2cce7d65c797ec : SUCCESS in 1h 26m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3473c60970f4982bddeea9bb467cfba : SUCCESS in 1h 49m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0ab5a273cd840278b790cd0e591818d : SUCCESS in 45m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/06b7439029994c20b820989dfe55b73b : SUCCESS in 45m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8769c0b0ba6d4635ba01b157c9c3e46d : SUCCESS in 41m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/017016a9ddd24dfd9690e7e1c5636e04 : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/95005b160c2b46ffb6d37197735976c1 : SUCCESS in 2h 13m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b7ec157ef1343c0888fbf05f7871070 : FAILURE in 45m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/166204a9bb8a44f2825f2929a57cc7f7 : SUCCESS in 1h 52m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9485ac92ea6930255b9dbd14c7d41eaf887e6a0e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 17:56:32.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"aef5c5c84bd5930d70df8ca3fc0a8751e34804ac","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-06 06:55:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e0c9b406ef878f2b0f9758634cac381d5f59524","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 06:55:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"533de4a514128950002482cc605d571e48f3932b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 08:46:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c11cf7fec3b432b8cc03a8807d45906 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1039f745ba3a4b539aa9cfa48da68fac : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7c1ec34874144bdac3d2b5348ec1a1b : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0315f42208ed402b96ba82878bc16eff : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341f3595e1bb452f9822a0750f07fb5d : SUCCESS in 14m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dee4ee152844db3bfdaf4538c51a3f1 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c351015ee99b4be6af780fbc0643d671 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5dd9a35f01084342965f44c55716ed48 : SUCCESS in 1h 37m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84146f6330994c698add5ea4fdef333c : SUCCESS in 1h 37m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4106fbd406624c068780f3d003d3855c : SUCCESS in 40m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4bd04983715545ba8a262794bee8a710 : SUCCESS in 37m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eeb80ec936d5424fae0dd51158ed3a95 : SUCCESS in 42m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/892f2c4d589546f39f7aa082317fc3cb : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":1},{"id":"916508b37d549f495a9664e5873947ad97142298","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 08:46:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b9d22383fdf87744e2aada8090e6c59355a7a24f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 08:53:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e0b18b6b9f48b0a4304ae35315b481 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d6f78376b584f1a80d1a69dfb68961c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cddbc3c83729ceb34069ed17dd4d553b9f6f0f1b","revisions":{"cddbc3c83729ceb34069ed17dd4d553b9f6f0f1b":{"kind":"REWORK","_number":1,"created":"2021-03-05 14:31:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/44/778944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/778944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/778944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/778944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/778944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/778944/1"}}},"commit":{"parents":[{"commit":"a125aa1439b72d7d7636861f11a9222dd26444f3","subject":"Merge \"Fix pools going into ERROR when updating the pool\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a125aa1439b72d7d7636861f11a9222dd26444f3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-03-05 14:26:14.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-03-05 14:26:14.000000000","tz":-300},"subject":"Drop lower constraints testing (stable Ussuri)","message":"Drop lower constraints testing (stable Ussuri)\n\nThis topic was discussed on the ML and QA team proposed to\nto drop testing lower-constraints [1].\n\nProposing to drop this test because the complexity and\nrecurring pain needed to maintain it now exceeds the\nbenefits provided by it on the stable branches.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: I17dd7ad2865fdc0b29308e482dd4ca9d385d4d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cddbc3c83729ceb34069ed17dd4d553b9f6f0f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cddbc3c83729ceb34069ed17dd4d553b9f6f0f1b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
