)]}'
{"id":"openstack%2Foctavia~780215","triplet_id":"openstack%2Foctavia~master~I34341e1aaad1524e3e0834309c5f6897f176af53","project":"openstack/octavia","branch":"master","topic":"780215-patch4","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-11-19 18:07:44.000000000","reason":"Change was submitted"},"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2025-11-17 19:23:06.000000000","reason":"\u003cGERRIT_ACCOUNT_19298\u003e replied on the change","reason_account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2025-11-19 18:07:44.000000000","reason":"Change was submitted"},"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2025-11-19 18:07:44.000000000","reason":"Change was submitted"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2025-11-19 18:07:44.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-08-21 21:11:46.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-11-19 18:07:44.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-11-19 16:09:37.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-02-06 11:36:56.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I34341e1aaad1524e3e0834309c5f6897f176af53","subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","status":"MERGED","created":"2021-03-12 10:14:21.000000000","updated":"2025-11-19 18:08:45.000000000","submitted":"2025-11-19 18:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"780215-780215-patch4","meta_rev_id":"aadfd5280f52e428fcc63f8a181c9e04d5198374","_number":780215,"virtual_id_number":780215,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-19 18:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2025-11-17 19:23:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-19 16:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-11-14 06:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-19 16:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-11-14 06:41:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 10:57:48.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-12 12:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 13:29:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-22 23:46:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-07-29 22:20:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"},{"updated":"2023-08-04 20:01:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-08-07 12:08:38.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-08-11 19:53:44.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2024-01-16 15:50:01.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2024-02-05 22:00:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-07-03 20:53:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-12-16 21:36:58.000000000","updated_by":{"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},"reviewer":{"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},"state":"REVIEWER"},{"updated":"2025-10-15 17:50:53.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"REVIEWER"}],"messages":[{"id":"a17f3572133423cf29cf5edd66fe26cb63b2e1fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-12 10:14:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf3b2be668f81ac92da216a445f2736cc13345a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-12 10:57:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI\u0027m not sure the API service should silently (without paying attention to the logs) continue without the missing/invalid/faulty provider driver or just exit with error. The provider driver will show up in a provider list GET call which could mislead admins/users to believe it is correctly configured and available.\n\nRelated story: https://storyboard.openstack.org/#!/story/2007412","accounts_in_message":[],"_revision_number":1},{"id":"20bfc63af661afee1889bdb96da85ea720e25041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 12:33:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4da7f7c4f9e947cc81f70ffd86a5c559 : SUCCESS in 13m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da217e58624445ba8bafb0dbc4b9e80a : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf99de8792704892895f39bb19fe4af6 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e09abad7ec4f4350a7d3986565733257 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07a9e9db1a284c46852d187d55f50d11 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4de8d4ed1ca49a483839e62c3c7995f : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0076faac61594c85affe91ed47b3ce93 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be85f430beab4e7f916b4c812ae9c440 : SUCCESS in 5m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d92e0cac329a4502bd9b8df5b076a894 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/878faedd53cf44fa95a515d353462cd9 : SUCCESS in 10m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0e9f619b8a924da09b7f29079efe7efc : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73cd4c98318e4ac6b533dd21ee696e6c : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98bf20aa646d49adb06de2ca500c5b75 : SUCCESS in 1h 20m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7dee9361972c4f7caa5dbf97a74aa0c3 : FAILURE in 2h 01m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de9e17a570aa4a6081fe994ded442fe5 : SUCCESS in 49m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b1f90b2762347f5889690ed2914f880 : SUCCESS in 34m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c7acbdfede14a97bf8dbcd87cac6951 : SUCCESS in 1h 07m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b92b8ada65743a380ee1d741ff7d59c : SUCCESS in 1h 08m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d6b86e4f777048999a3fb61a3fa85d9f : SUCCESS in 1h 38m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b3a5e0ddf7fa4ceb8412242326adc50c : FAILURE in 45m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/61805d9c1814450bb4c33240778489b9 : FAILURE in 2h 14m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ea8dc4254d3d4b8fac8e877430b10f6f : SUCCESS in 1h 37m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f92d1c2e1c844615bfcba0c4b5595562 : SUCCESS in 1h 32m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9289c895e8b8463f8bc8f7bc951c70fc : FAILURE in 39m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13f67dccefbf201b6dfa96274b4ebff845e822b0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-22 23:46:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26245f5658e72fad50963e7292c1a2893939e958","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:20:20.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":1},{"id":"bf4395bf97b5d587101781822cc180c0af348eac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-04 18:32:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e99ad77bbdc1497a26538d38ba4dc1a259444319","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-04 20:01:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8be145ef4d78cb8758f3c720d4e9f634fb8f59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 21:36:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/682f2f2193a14f23b6deaa0fa0e7848d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3f6ed3a764743a2aea497381e9c6763 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a15604165b40d3a6f718296e35a3c0 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b4bcf6e36fb42c9840f0effca0f6a16 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b3b71e346224160a7a9d49d68ff9902 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93ba7612112c44ccaacc8bae576eebb0 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4dad2697a2b426cb4e44d24be3dd760 : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f62bdf8159a45d986e235034fbf5738 : SUCCESS in 5m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4a00ab4b9df64fafa77bd475a3daf3ca : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1547eae2e5d24066b8ca8800d4931f5d : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b89b42c6fd24b528b70f601290adb03 : SUCCESS in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/141ad22c5dd447faacede680421546ea : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c670edcdb96b4034ad4e8338db82c9e6 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e020244e1e3424b8991f20eef7e0b25 : SUCCESS in 6m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/875e483548914986aa897e71c48d2bbe : SUCCESS in 1h 25m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/da39515f4c5b491996db492935c35675 : SUCCESS in 1h 50m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/502d0f7203f041fa86e0b8d0fb56b8d1 : SUCCESS in 1h 10m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/19e711539a254b75bac5dd1f8233870b : SUCCESS in 1h 58m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2e757a678a8a4a2cb29acdd7fb3d8a51 : SUCCESS in 1h 33m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1cc2408b368144ce8865e7555ac6c9a9 : SUCCESS in 38m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e91f2d838194814af033f8261899407 : SUCCESS in 51m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0cac2535153c439dbfcb6e5836d5cc9f : FAILURE in 41m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/713618aa5fa6485fb41389940a045547 : SUCCESS in 36m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4d1d754ac7f94e979760ed1e5828d6df : SUCCESS in 2h 13m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/433743d04ca5445fbff54cf119a27e7f : TIMED_OUT in 3h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4eb5bed2f8a24d36b8e16cc22762d9d0 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/06e7da737bea4af493710498cc6b7743 : FAILURE in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84d6b4e612c30a7aa969f47080fef892f003957d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-07 12:08:38.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3eb020f8ebaff8ff0de0f4a736ab05d72a876f2f","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2023-08-11 19:53:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ca889b3cf0fc81b6f72da7d3ad6a85ea0a8d735","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-25 22:05:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"62d5b8341d1e24927285e9324d8ad012d0172aed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 20:11:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ea22a2a17b9f3fa16cda1dfa05ec483e5049458","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2023-11-03 14:18:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6917c6abf0cf402f68e15e92292055c389fe5b91","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-10 15:17:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1609bd634dfc037ffe1ba1de317e5b86b1d4e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 18:26:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f03c4dce11f44a1fa02677d121183306\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53f1b1a4c2f54f48b3bf40f2038795d6 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96680c285bc74049bc652d372fdbe873 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bd7261148cc4e2284523e755f040f1b : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d198df4090d4699b2835c0e3acae657 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5183681737e405b880fd7b70593f585 : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c96e47bd9f04d0c908467f2df4324ac : SUCCESS in 5m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/080d5dacffc141d19fcb7735404f7839 : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/29624817dcd849f4b2cf5d6c22a7d355 : SUCCESS in 7m 59s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7d22c6b2474f4d21aff98c5e2318a413 : SUCCESS in 13m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c03749869c7642b39b63bf48317bc042 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d9e208d072ed4dbd861a165c39fd4706 : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/241b19ce20104e82a5bd856862370ad3 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bcf48bf4af9b4db08d529a4bbaf82ce5 : SUCCESS in 8m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/afaa49bf587f42fabf4acb70ba845e1e : SUCCESS in 1h 23m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6994b985ffaa473ca39e38f7f9681dc4 : SUCCESS in 2h 18m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8a072bb91563462aaae4cb4ae6e3f86c : SUCCESS in 57m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/88c5c7066fcf4c9b9ae770753353aad4 : SUCCESS in 1h 38m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3e40f82056fc499b9333b76ebb8e2fef : SUCCESS in 1h 35m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0216f9dcc464c39817c09272d723b4e : SUCCESS in 59m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd7c256583cc4806a4bf47da83a1600b : SUCCESS in 40m 05s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1600afff8d6849ebbc05b8366b711dfb : SUCCESS in 52m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2246a126ac0c40fa99fa03a80e9c2692 : SUCCESS in 1h 19m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3028e9534ef4a968369e3e3e4d884c1 : SUCCESS in 34m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b4b7bf134134ad08ea3f57ea94e2753 : SUCCESS in 2h 07m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d20ff1fac4e4b44a3f7174d241f4d5f : TIMED_OUT in 3h 04m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e15fa2ab10e647a59d5be465e004d38c : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf2a7f9d4f07421f93e572525edd1a1a : SUCCESS in 2h 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40026f8874677d65dc693e931ec331dd5c390bcf","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2024-01-16 15:50:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f72dfd2bdb0147f3fc3c3e28edcda1137e7e50f5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-05 22:00:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2977a768fcc2674b8a5df6834802e26ad1f7751b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 08:24:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"247bdd7aa47e154cb96cee9fdaf0245a08c463c2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-06 08:37:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3501691e954d557e77a44f21cda32ad74ae0a25d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-06 08:49:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eaf100b991fe624ccdaffdb76fb401679782aa18","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 11:36:56.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"3c271a0dc2ae04c4bcc0cb540ddf5a0e27de39e2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-07 22:09:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19edd744768f82c17305084b287eb22a36b63c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-03 20:53:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"70b58dddecec88a4fcd3879f1072463d79f5f525","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-03 20:53:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9bb19b39c9f9a0bda70162ee62d91ae0889445d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 00:26:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba812d15cee24495af7b8efc5b366d2d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3af32ed364c949178ef015cc464b5cfd : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb56850e3b8b41798af740776e637984 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4c05b684b354df3a04ab41fcb75087e : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2b4b9435f094c1ba5a3fd8fe4dfe2a1 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e26805d3dc8453699ae228d5f64f1a5 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a73742b02554750ba532ead24cfeb95 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e49e0c4e7575402c82849120d51ed9e7 : SUCCESS in 4m 20s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/60a04c2a2d6a4f198025c9e2bf13bfd4 : SUCCESS in 4m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/07e31b0c40c74dc595925eec05747894 : SUCCESS in 10m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b191b804a6e24fd39c8614aa205e848c : SUCCESS in 4m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/427858d88fb2459189bc69c9fe00addc : SUCCESS in 7m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/196022e143784ebe8aa9dcab4ace6363 : SUCCESS in 8m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/72bd443387ad46f383aadc03f568e6e9 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c1f5f447ee3947e7893b92080108dfc1 : SUCCESS in 2h 03m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ec7b3a414ea44b94836330350a4a2951 : TIMED_OUT in 2h 34m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/07209d743fde4b5cb031dac95cda3e64 : SUCCESS in 1h 20m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5b7c84008d594ff9bd2cbfd08010f6b1 : SUCCESS in 1h 54m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2905ec0e79eb426283ce1014bab9c512 : SUCCESS in 1h 11m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2972de7c69f64e4c94bab157743ca63b : SUCCESS in 34m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f23956cc043d47c2a917acfb8b576fce : SUCCESS in 56m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8ceadeb2cd3e401c8539d5c6b0e7f67d : NODE_FAILURE Node request 300-0024637358 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/467910282da94944b8193abb4feb8640 : SUCCESS in 48m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/edf1de79e1c3479b908e1ef123f8a517 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c10ce10924b43989eee1abb164d4fa1 : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/640c5940ba894d4b8ccbba9f12a301e4 : FAILURE in 26m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/be3c77f3c9544593938b83a9a4ccabd9 : TIMED_OUT in 2h 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc56d75a7489c4c91d5056594ce3adfbcb3470bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-20 22:25:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6dde5651d20900541b79ff45c8cbcbf336365d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 01:30:04.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10eb9675db334cf3a3246ea357e832a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8c8d8059fab48ebb5b747463a5f4ea3 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b3499b505c54976a934073c19c578cb : FAILURE in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17cdf29776154bd4af260de65b3b3780 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/307a69b98bd74439b182662acb8978d2 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/084c7c533b9540d5b3dd8066ec23e770 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde1b328437944ada25735155122e449 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0b64e5430fa4ceb8db118e9d94bfd20 : SUCCESS in 4m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9e83a133f489461aa56c25d36cb6f933 : SUCCESS in 4m 01s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ee35c58a90314b22a2a83b5c36b0390c : FAILURE in 9m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7a17d9a59af4a4795779eace678bc64 : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47639deb0b5e407bbf0b505a51c3477c : SUCCESS in 10m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92e52cfee68e4293b2003e874b0da15f : SUCCESS in 8m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/953553d994f04dd4995de15f110645bf : SUCCESS in 8m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d82c98f080b40039b7055ef4dacc785 : SUCCESS in 1h 14m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ed42c6aa01a94f7d8013385501a65ea5 : SUCCESS in 1h 44m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/89130d5d15494784b05b084e284f9b62 : SUCCESS in 57m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ccfa149a9ffa4669a4d9eb5a28453844 : SUCCESS in 2h 24m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f9862689d53e404fb6b1bddc91534ce8 : SUCCESS in 1h 22m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8f29d1501d34dc58de7c50ee2af72b5 : SUCCESS in 49m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0e1fc653cbf649db99b6d7c779ebfda0 : SUCCESS in 52m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8bd9736db9f548aa95906f29f7e90ced : SUCCESS in 1h 11m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88c9ced918af4d3ba8728f5c92124e63 : SUCCESS in 41m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/75d2a2ea248446be8dbf5a11f31adc7a : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9668639a5c0646c7b71badceeec95f29 : TIMED_OUT in 3h 02m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/22942ea599334e68a3f4628219ee6507 : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/eaa10491606e4d668ea4345a4ef511bd : TIMED_OUT in 2h 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d67f83e8d3c2d1ee4548fe53f1feeb626c54cf34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-21 18:45:37.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e1e179a5302a59fe89da7318088a25463f2bd63a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-21 21:34:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf3fe6df92698681af154666b12d0fb5e796da59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 21:51:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ae4b299771e4968beb16e77615e1a28\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adbf482af1cf46eb8fc7e73de42ad890 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31cda36ade6040749756d328a7cd0c6c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ed1de447014448494d1f2c552012204 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdc606777df548feb8eb2411238d6aee : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3ed0ae1b0414cd1803e0614a3235f0c : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610f901d775b453a9ec902c5fa1b3c31 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094074277fe94ebab37ff914c5043c13 : RETRY_LIMIT in 4m 57s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7e4e1e15ef414f7aa81e16ac3a6b545b : SUCCESS in 4m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/d11663e8cc7a48adab2d56519065e983 : FAILURE in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f0b9a02771b14585a488b7f7456cdbb1 : SUCCESS in 3m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7c4b6510fb441cb9d770239839b3047 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8bd57438a633415cb1c01e1ca428fa41 : SUCCESS in 6m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5165fd322cf9475b977f93608aa50357 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7c87236e27a49ce9a46220b1a30e6d0 : SUCCESS in 1h 45m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c9bc3fa904854b54addb474a3e8cdeeb : SUCCESS in 2h 18m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8060df176b8f471b9c00db7debdc85fe : SUCCESS in 1h 23m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ec1670faa828438da9c32555b7b682e5 : SUCCESS in 2h 26m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c20907b1dd2541bcac5052cf1e8573d5 : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b8e3deddccc4fdea33de9c8aea9074c : SUCCESS in 42m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce128f12bd7843b293da681744125fa4 : SUCCESS in 1h 04m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c297b2ead02c4b66abd84d22e025525f : SUCCESS in 1h 06m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36b74d6a540445ba96188d6a4aeed087 : SUCCESS in 49m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7f9c97b6a24740d2a9f328834db51353 : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd22d32099ae42e0b2853b8c342f9ba6 : TIMED_OUT in 3h 04m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3543a4b28814ebeb6dc71bd4efad63e : TIMED_OUT in 2h 32m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a4c63e35c4f4be899d146f279bc90ec : TIMED_OUT in 2h 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a5b46bd1dd001740a509ecf6c1b85360d31bd027","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-21 23:36:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b776abff3cc45094506b1032efbb3617f707b790","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-22 13:46:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9c27283de93650e7db972dc568e7bd0a3bc8d72b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-08 15:03:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2df195660030a93fca6598bd6a17113cfdc9a020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 18:11:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1fdd99d9e984f41806eccd8f27ed51c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f4e0bdfd40441bb849a2db47f0c8755 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b07bd550d044c50949445c7d241eb9c : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9297f67abc24a8e946acf606b6e56f8 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b0d6683386c4e5f9de5bde51af5da44 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0efd27781ba4faaa75a22079c82685e : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0851a7796b404b17ab61fe40ae7c35fc : SUCCESS in 5m 17s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7f37f3f552e64886a9e693018281eb69 : SUCCESS in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6be487eb6cf74e8c8acac78ad54d54a4 : SUCCESS in 7m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9fd631524cb548999b279b0092f5b832 : SUCCESS in 4m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2c8f3b3a7ec4fae9fddb520ac3f1172 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/859d3db8204c443d9143ffd9704047ac : SUCCESS in 7m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/195f84c8412b453c95b158dee580f0ec : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42cfffdb9bb24a42a3093f5acdbdf60d : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5fa285f924e44366acb8916ff7e7aacf : SUCCESS in 2h 16m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ea07b2ad70534333a3b67854bfa00acc : SUCCESS in 1h 22m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/97ad17568e5b472e8558cd52d1f6a7f7 : SUCCESS in 2h 21m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e3c729d7c82244b29b89c173304e603d : SUCCESS in 1h 04m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6f735af85d846ac9a9e30e28dcece96 : SUCCESS in 51m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db5b52513d4d4bc69d0af25f3ef8f234 : SUCCESS in 48m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5b7441d832a949218419134043bd6c17 : SUCCESS in 42m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ab845112aee44a69f0ab95a53bac841 : SUCCESS in 50m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f054238254db426a89f7e9fa18229ee8 : TIMED_OUT in 2h 35m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d83393c6b2b64548b7be19e9381bea82 : TIMED_OUT in 3h 04m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d4617fc9dcfe46e7865e65a094c48144 : TIMED_OUT in 2h 33m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/36c458cf1de24d1192d91d6bbd9ca63f : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ee1489dce4577dc8f55683189822a121ffc96c29","author":{"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},"date":"2024-12-16 21:36:58.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dcd02570e8e32feb1d573693c0707652e3d825d7","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"},"date":"2024-12-16 21:43:27.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_35648\u003e\n","accounts_in_message":[{"_account_id":35648,"name":"Alan Baghumian","display_name":"Alan Baghumian","email":"alan.baghumian@canonical.com","username":"alanbach"}],"_revision_number":5},{"id":"8e4f2d375f0d04a52bdb83e3441d5e76a0ed4306","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-15 03:58:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"81c1c3b857c02f1ef01411b59562cd60a4987184","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-21 16:35:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce19b8637fc5295a9066bc3a0cb875e9bc2813d0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-05 21:26:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0496e90283bfd3d6b95c342c0eae4e3e100baef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-13 20:27:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c1e473c35a2728a36a766fd83dd75d80c858186d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-13 20:28:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8f4e1adb0baf2f9ef991ad50f930af0086330b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-13 23:08:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd883274a97a45c39900c7002f1c978e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fa225b355494aa49efb7bbb6bd2164b : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8030710103844446b2a8d759fe2a911b : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6212029976824049b37efe8e638fa846 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3776c2b922c642b3b526898a58725833 : SUCCESS in 6m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91f265be7c144fcbaa1355d3eabe1d2b : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cedda0eb25d4435a170a538fe23515c : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9334676b99604d40bf1a45e197ba8386 : SUCCESS in 2m 39s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/31bcfe413f734915b7848779da2708ee : SUCCESS in 7m 38s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/80542e79587f473580d555839700c374 : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8082a68a8f844d28a375ee4a5f7cdbb4 : SUCCESS in 3m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f29553f5ebe34a509723ee8a39f2d617 : SUCCESS in 5m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f528f55ec1464f0ab64521f0f4895dc0 : SUCCESS in 8m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9658421af5f44e42935149bc1659b2bf : SUCCESS in 1h 31m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b0a54f781b7d4bf2bb990c58c40123b5 : SUCCESS in 1h 25m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9235c89c30564589bd27d86c1a639d4e : SUCCESS in 1h 14m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd3c6dbb4dc94b11b797ec88ed106e6a : SUCCESS in 1h 31m 30s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8416499dd944d4498e2a05bb3aa090d : SUCCESS in 1h 07m 02s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/119bf7046fbb417294c2ead10c777d47 : SUCCESS in 1h 54m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/421616221f4c415fb8a07d60b2715474 : SUCCESS in 1h 41m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d887cb8fb53442d08ab3c6078804d46e : SUCCESS in 41m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a5964e5b73a4981bc94a63e8b4f15d7 : SUCCESS in 52m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f70579c6d07f497d82cd9d34b3d3ea01 : SUCCESS in 22m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/496641d537dd4691b335b2fbf09b4835 : SUCCESS in 1h 54m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/efa747d9e3124605a09147b835fe995f : SUCCESS in 2h 01m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4edfa927af7e4e49b55af2f5a7313ce5 : TIMED_OUT in 2h 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"59d7ea429ea27a3dd31155a3b884a3d0568a7446","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-08-07 16:31:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"97f332f96404b2b49aa39de9eb0ef67933af0ccd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-08-19 11:44:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b5974753862b688e1b0f0f587adb143203d85477","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-08-21 21:11:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8bf2af74cc2e04dae1f047a9f79f0520c5a9a072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2025-10-15 17:50:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"eaf21ba40cc2b502ce63bb9a6fe3c3cbb45e7940","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2025-10-15 17:57:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab74e3aa698498bc4a6152bd84219c0d0cdcc733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 20:48:04.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/848ca10b65964c428aacaf9a58db7bff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52b7d640e3d84f758d4ffed1a19972d2 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4049d706a541f9bd2c3ed7930e4b4d : FAILURE in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d534c1b584941ebbd01cc524ae89f1d : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91dac5bd48e34b8fac09aa3e2566e224 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc4a70e05e8464c9cf9008e5c2ad394 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/150a2159c600489d94de228e9802289f : SUCCESS in 6m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a292bfe5420649018f30d4e0f0ce72d8 : SUCCESS in 7m 11s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2d45d97643fb42f3bc23c44dbdc5e08b : SUCCESS in 6m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a058c0a903c42dd8f048a37f85f725c : SUCCESS in 4m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/018c2df64e424ca4b4584869b88f94ef : SUCCESS in 7m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/54633b20d46b4cc8b83b15919b117da2 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49cbc91368344ee6b8783c4dc48f148d : SUCCESS in 56m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4decf7cfb16945c88728f539e0910add : SUCCESS in 1h 52m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/863f22f2a97e47129930ae0f40f0c664 : SUCCESS in 1h 02m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0a5fe1f262f44c86bb89c21046462313 : SUCCESS in 1h 58m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/58486e62c68e4f079db830ecb0268843 : SUCCESS in 1h 14m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39c68d12311541cdb49bacd3eb478620 : SUCCESS in 38m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f023bacd68f347e489d2788deac3a67c : SUCCESS in 45m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fb43f20af9845618ee261781c62a1c9 : SUCCESS in 36m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a4128cc0aaeb47ad8288d78087eae8c7 : SUCCESS in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b589782952394aaaba731566fa2b996d : SUCCESS in 2h 55m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e07fa2e199fe4bbd874a937b5d3aaef2 : TIMED_OUT in 2h 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"58e2953d4f28db4932e9839c97d1ef62d74f871d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2025-10-22 15:18:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"f96676505c29a53ecd9f78e1712a68555d8a24b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 18:23:27.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffe0380bc9894c19bfb3097cd60aab49\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e801c80601874808aca105520ad7f442 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0475475f6a8d4b6ea954fec35d9dd2ae : FAILURE in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37fe386597064699adfe01cc44b05c3e : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b0ef31aa1ab4c51a6577b2fc382e8cf : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435303d70fab4beca8f3f4c91df416c4 : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5746975de7c0423bab107d9151b6b79b : SUCCESS in 4m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3a3253049f204707a906207f99fc608b : SUCCESS in 4m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8455467294d9437d8c0a4d7926af7975 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b38a56159a245d0bca1d075faef2d22 : SUCCESS in 3m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ee75e49657d042b2885765b60971d8b1 : SUCCESS in 8m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/840aa46cac4e49d59be84db773670cde : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/137c1f8f79204b17b76ff163434f4fd3 : SUCCESS in 1h 15m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9ac37bfaa1224941b856e0e921144cba : SUCCESS in 1h 59m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fcedc9c626ae435291f7f60e65084416 : SUCCESS in 1h 07m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/756ac505b8c74994b7dd67703fac7c55 : SUCCESS in 2h 03m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8fa25c5ca7784ef18e1f828f8f6bf8c6 : SUCCESS in 1h 20m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/00c24687b32f47949da7732f0f57516d : SUCCESS in 40m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fff2bdb39be94ecfab3b9a1790339d9b : SUCCESS in 45m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59b7a33b7d4a426fa2155272b9791ef8 : SUCCESS in 38m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9366903d1a144f0dbbaae376da8c9233 : SUCCESS in 2h 31m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca0f8b4630844a57a6c059b3bda3b658 : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/18a6fe2ac6f044fc9ab7984dc992b22c : TIMED_OUT in 2h 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"80aec82f78ae0080f3ccbdd571b759711028dd1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2025-10-22 18:56:29.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"ee40b479fd202168334f89360862b17195a7fda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 22:00:40.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/406da06b1097465599595cd9b6e97659\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/459ce0706af74507b0272e2f7e0ab6d8 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc8dc279e5342209df95203a61284c4 : FAILURE in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/473f58165b054ce9a6e134ed10d384f8 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/794d6c3ce8aa443d8c541d37bab6d8ce : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff13e7c1fbe048aa92278d83cce247e9 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3913c4369bc41198d4b7bc2c0e99629 : SUCCESS in 4m 05s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f86320bc82fa4d14ba13c14d4ed9ff02 : SUCCESS in 3m 06s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/31b518cc853b46f298821d5e65f3e74c : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/254b131501134a868293f2521a6d23d7 : SUCCESS in 3m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/91a64b01b4fe4fcfb62052b9f5b43e23 : SUCCESS in 7m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f29a23dec4db44d98c7ff7510021778e : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4133f09900a4f49a71e76e5fb134d8a : SUCCESS in 1h 13m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3c1f343b3afc4e8a99fc0c65caa49021 : SUCCESS in 1h 52m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4da373c55e644bb4a11e129c1bd6c4f1 : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/96ba34a718f94952b6d36ca05f9e0b26 : SUCCESS in 1h 55m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8d3110c29f2e43bdaa6f25ba0c35c743 : SUCCESS in 1h 07m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/988abd6b49bb4ddbb99e455bd932481b : SUCCESS in 35m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ce0b6015b4f42748f66867e26d6d021 : SUCCESS in 56m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa257953057741bda1fdc8d5d91c31ed : SUCCESS in 48m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4705830ab2bc42c5aa927bd4a882c7d7 : TIMED_OUT in 2h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f401ab5063e4d3699ede2bfaea5dae2 : TIMED_OUT in 3h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/efc3b63a1c8848218dd92ce70547c371 : TIMED_OUT in 2h 33m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/common/sample_data_models.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_server.py\n  Comments left for invalid file octavia/tests/functional/api/drivers/driver_agent/test_driver_agent.py\n  Comments left for invalid file octavia/tests/functional/api/v2/base.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/db/base.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/db/test_repositories.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_amphora_info.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_haproxy_compatibility.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalived.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_loadbalancer.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_plug.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_util.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/test_agent_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/health_daemon/test_health_daemon.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_haproxy_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface_file.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_ip_advertisement.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_keepalivedlvs_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_namespace.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_nftable_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver_1_0.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/health/test_heartbeat_udp.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/jinja/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/test_vrrp_rest_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_get.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_listener.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_updater.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_agent.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_factory.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/auth/test_barbican_acl.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/generator/local_csr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican_legacy.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_castellan_mgr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_noop.py\n  Comments left for invalid file octavia/tests/unit/cmd/test_health_checker.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/logging/test_logging_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/test_user_data_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/sample_configs/sample_configs_combined.py\n  Comments left for invalid file octavia/tests/unit/common/test_base_taskflow.py\n  Comments left for invalid file octavia/tests/unit/common/test_config.py\n  Comments left for invalid file octavia/tests/unit/common/test_constants.py\n  Comments left for invalid file octavia/tests/unit/common/test_data_models.py\n  Comments left for invalid file octavia/tests/unit/common/test_decorators.py\n  Comments left for invalid file octavia/tests/unit/common/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/common/test_keystone.py\n  Comments left for invalid file octavia/tests/unit/common/test_policy.py\n  Comments left for invalid file octavia/tests/unit/common/test_utils.py\n  Comments left for invalid file octavia/tests/unit/common/test_validate.py\n  Comments left for invalid file octavia/tests/unit/common/tls_utils/test_cert_parser.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/test_nova_driver.py\n  Comments left for invalid file octavia/tests/unit/controller/healthmanager/test_health_manager.py\n  Comments left for invalid file octavia/tests/unit/controller/housekeeping/test_house_keeping.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_amphora_rate_limit.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_task_utils.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_amphora_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_health_monitor_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7policy_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7rule_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_listener_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_member_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_pool_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_amphora_driver_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_cert_task.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_compute_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks_quota.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_lifecycle_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_network_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_notification_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_retry_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_shim_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_controller_worker.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/tests/unit/image/drivers/test_glance_driver.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_base.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_utils.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/statistics/drivers/test_update_db.py\n  Comments left for invalid file octavia/tests/unit/test_opts.py\n  Comments left for invalid file octavia/tests/unit/test_version.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/test_cinder_driver.py","accounts_in_message":[],"_revision_number":9},{"id":"887db0524883ba7fa5447670322ed41017cffd24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2025-11-05 08:59:37.000000000","message":"Uploaded patch set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"53627cc5ece2a9f01ff9ff74ea4dfca973b2f823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 12:05:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17789286327a4a38be6e9ab810f479d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f19c70ce4cbf4069ae0ed9b2b61b4337 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8499439708f14519b20bd542e099175f : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13115e5aff964683ba0a11e7332f06ef : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/808827a8960549509fbcd303a8b3a881 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f244612dc52e4f6baccaddbf5a0c3285 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a8c321f89a04b909d361bf00d1b06c3 : SUCCESS in 6m 33s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d3ca634104ca45808317d052b8912618 : SUCCESS in 5m 06s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/927ac99d09804f18bf2adb614faae606 : SUCCESS in 9m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1d0232dbf5a64913a3a122565c05922d : SUCCESS in 5m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/04b12e5780654bb9bf11b900e02ac177 : SUCCESS in 8m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7240a57fb757498aa6557be6ba69333c : SUCCESS in 8m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2acdf90856f94385ad82eb780f2506d9 : SUCCESS in 1h 38m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/234abd13e6914241aa7580c0605e6887 : SUCCESS in 1h 51m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/054143afb1a0470987b08ebdf595c5da : SUCCESS in 1h 09m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5c9e773fa5c5494d98852ecc7d75b07c : SUCCESS in 1h 55m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/de602705804f4415ae1a5ec8c8d351ee : SUCCESS in 1h 17m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58ac03d2bc82454abc6ddf8119920ea4 : SUCCESS in 38m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cf0c309954a446a6b1dc00fb6a743f09 : SUCCESS in 42m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cd129d5e683474396c7d015be47ef1f : SUCCESS in 35m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/629fab8263e3431daa259ec043845c49 : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ae24b5a5dc44fc0b1895018d1583158 : SUCCESS in 3h 03m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1af23ce7cab44a52ab28a2ae7e059016 : SUCCESS in 2h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"31008e4bc9a1b8cf60e2f48129957abd72a920eb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-11-12 21:38:50.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2b49b4dafb2f7f15a6dc525954854182053d0a95","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2025-11-13 22:38:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d3cf689c93617eb92a09501ef44a633494f228f1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-14 06:41:59.000000000","message":"Patch Set 10: Backport-Candidate+1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cd3d161c0226038f6a60da31a9e60ae62c1beff4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2025-11-17 19:23:06.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"fb7f7ac698fe4b8786236cb38121e2b105a7b23c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-11-19 16:09:37.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a3c3ba7e3333d0c3ee05b8b8606650db3694ecc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 16:10:10.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"58afe7c812e6d0fcae455f6708fa86c791b0769f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 18:07:44.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c469d46dd4d40848432e60170409dda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad7b2c087fdd4ba2ae31269184b19e7e : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d0299b827fa4a02ae6a109c31a47801 : SUCCESS in 6m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5016f14421164979b6e9c912e1c52534 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcea47e137ea47eaaa551657826139ad : SUCCESS in 14m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad10a9e7089e49c5a049e5b2e10af9ff : SUCCESS in 5m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0fed0016cdb4476b91cf985ab3c1c8fa : SUCCESS in 4m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/741c252ddcb74812a547d51186fd756c : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/620c3fef591f4a69b0b5e35c0406cfdb : SUCCESS in 1h 52m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/af8e36c78eda41d083e6bebef85cbc12 : SUCCESS in 1h 53m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8267112c01d942ca9e729d0a774258dd : SUCCESS in 1h 01m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4420ecd25f294f29a0c82c3d2f4572cc : SUCCESS in 35m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f59daa3be62f4d0fa884731c5d12b202 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":10},{"id":"fdc036225fdbd78db9679c79c1569ff6e70a40dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 18:07:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"aadfd5280f52e428fcc63f8a181c9e04d5198374","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 18:08:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d85a3e764b6d4e42989205bc999b44c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d26ba10da044c6b5b7e3c02ba77f80 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f51f4628e5d044dda320546c4a97f538 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"6fc1abb780a2fc1413d5b206cec4b033a638c3c4","revisions":{"8cdf12ee05d22aba5a223e794fffbc8cc305c81e":{"kind":"REWORK","_number":1,"created":"2021-03-12 10:14:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/15/780215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/1"}}},"commit":{"parents":[{"commit":"217cdbb719a8d303c486ed6526eedd9ca6fe83eb","subject":"Merge \"Add aarch64/amd64 support to amphora and devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/217cdbb719a8d303c486ed6526eedd9ca6fe83eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-12 10:13:49.000000000","tz":60},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, and the other provider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cdf12ee05d22aba5a223e794fffbc8cc305c81e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cdf12ee05d22aba5a223e794fffbc8cc305c81e"}]},"branch":"refs/heads/master"},"285c0ea74de75429a56dbbbce9c8d00b82a3a95c":{"kind":"REWORK","_number":2,"created":"2023-08-04 18:32:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/15/780215/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/2"}}},"commit":{"parents":[{"commit":"8ccc0d01764c234ce9d3dc0191b5a0952111e2ef","subject":"Don\u0027t update the management port when calculating delta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ccc0d01764c234ce9d3dc0191b5a0952111e2ef"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-04 18:32:22.000000000","tz":-240},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/285c0ea74de75429a56dbbbce9c8d00b82a3a95c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/285c0ea74de75429a56dbbbce9c8d00b82a3a95c"}]},"branch":"refs/heads/master"},"da0bf0d39db584cf95518d8e2ad3b672cf043e7e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-03 20:53:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/15/780215/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/3"}}},"commit":{"parents":[{"commit":"7974e755e18bed376ea5b1bce2a927eeb34494b2","subject":"Merge \"jobboard: Support authentication and SSL for Redis Sentinel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7974e755e18bed376ea5b1bce2a927eeb34494b2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-07-03 20:53:15.000000000","tz":0},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da0bf0d39db584cf95518d8e2ad3b672cf043e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da0bf0d39db584cf95518d8e2ad3b672cf043e7e"}]},"branch":"refs/heads/master","description":"Rebase"},"5f449a5c83d2454a5528e3f02cece393255c3d09":{"kind":"REWORK","_number":4,"created":"2024-08-20 22:25:02.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/15/780215/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/4"}}},"commit":{"parents":[{"commit":"8ccc0d01764c234ce9d3dc0191b5a0952111e2ef","subject":"Don\u0027t update the management port when calculating delta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ccc0d01764c234ce9d3dc0191b5a0952111e2ef"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-08-20 22:24:50.000000000","tz":-240},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f449a5c83d2454a5528e3f02cece393255c3d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f449a5c83d2454a5528e3f02cece393255c3d09"}]},"branch":"refs/heads/master"},"c018a602f5bd7e6879d7c4e9fffbcef4e61bda21":{"kind":"REWORK","_number":5,"created":"2024-08-21 18:45:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/15/780215/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/5"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-08-21 18:45:29.000000000","tz":-240},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c018a602f5bd7e6879d7c4e9fffbcef4e61bda21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c018a602f5bd7e6879d7c4e9fffbcef4e61bda21"}]},"branch":"refs/heads/master"},"c6e067e50cc47429d5a9e70cf2f70317ac8b3bff":{"kind":"REWORK","_number":6,"created":"2025-07-13 20:27:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/15/780215/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/6"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-07-13 20:27:14.000000000","tz":-240},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6e067e50cc47429d5a9e70cf2f70317ac8b3bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6e067e50cc47429d5a9e70cf2f70317ac8b3bff"}]},"branch":"refs/heads/master"},"592564392f9ad97de05cb8ee26538ff1551d11c0":{"kind":"REWORK","_number":7,"created":"2025-10-15 17:50:53.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/15/780215/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/7"}}},"commit":{"parents":[{"commit":"40e13e517b3000a0e4b305fdfebe0c12d339f2bd","subject":"Merge \"Remove unused amphora flow and status constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40e13e517b3000a0e4b305fdfebe0c12d339f2bd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2025-10-15 17:50:37.000000000","tz":-300},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/592564392f9ad97de05cb8ee26538ff1551d11c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/592564392f9ad97de05cb8ee26538ff1551d11c0"}]},"branch":"refs/heads/master"},"143300bfd3aed2c9642abd215676db10cff0ece5":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-22 15:18:43.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/15/780215/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/8"}}},"commit":{"parents":[{"commit":"ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c","subject":"Merge \"Fix devstack cleanup for log offload location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2025-10-22 15:17:43.000000000","tz":-300},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/143300bfd3aed2c9642abd215676db10cff0ece5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/143300bfd3aed2c9642abd215676db10cff0ece5"}]},"branch":"refs/heads/master"},"9e28ea7996bc8c09b675a7d4dad8ae2ba2a42f9b":{"kind":"REWORK","_number":9,"created":"2025-10-22 18:56:29.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/15/780215/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/9"}}},"commit":{"parents":[{"commit":"ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c","subject":"Merge \"Fix devstack cleanup for log offload location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2025-10-22 18:52:18.000000000","tz":-300},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e28ea7996bc8c09b675a7d4dad8ae2ba2a42f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e28ea7996bc8c09b675a7d4dad8ae2ba2a42f9b"}]},"branch":"refs/heads/master"},"6fc1abb780a2fc1413d5b206cec4b033a638c3c4":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2025-11-05 08:59:37.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/15/780215/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/780215/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/780215/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/780215/10"}}},"commit":{"parents":[{"commit":"ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c","subject":"Merge \"Fix devstack cleanup for log offload location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2025-11-05 08:58:54.000000000","tz":60},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/964892\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fc1abb780a2fc1413d5b206cec4b033a638c3c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fc1abb780a2fc1413d5b206cec4b033a638c3c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
