)]}'
{"id":"openstack%2Foctavia~780480","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b","subject":"Fix provider driver quota handling","status":"MERGED","created":"2021-03-15 07:36:58.000000000","updated":"2021-03-18 00:25:53.000000000","submitted":"2021-03-18 00:24:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780480","meta_rev_id":"b1aaf242b80962cba13b0274aeef609f61288d73","_number":780480,"virtual_id_number":780480,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 00:24:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-17 08:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-15 13:36:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-16 14:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-17 08:07:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 07:36:58.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-15 09:55:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 13:36:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-16 14:54:31.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-03-17 08:07:01.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"d9c59cece8cc1f71a8f2ff40dc7605334dedb938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-15 07:36:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6db6f1c9ad58defe4f476924b982d54b59266923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:55:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad283112147f45ccb750f62b263e4fe3 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/687fac631cc84429be21ba4b6da2ac37 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ee44d92819440799ea71bbc5b6e3c7a : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b91db41f55984c2f88e10281da4102cf : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d3e2d85548044b79c871b27de7d18df : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a6706bc657c4f609778583749b2562a : SUCCESS in 4m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8ab1e3a62d674a6e809e2feeddd10d78 : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a1d388db9ec64c58833fee1533e07f5d : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/50ec9d33d4d942108657fb71d7c6f3c1 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/69ccb60d716240acb9fa99cae806c642 : SUCCESS in 9m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01de9fb692454b48ad5281ca420c0358 : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/011bfc8620334350aa9660a3d442310f : SUCCESS in 1h 56m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/812d4b3039db4bd49b6554cbf07abc23 : SUCCESS in 51m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1517bd27e3cf403d8b10590c83350689 : SUCCESS in 45m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d74511250f44c12aac5d570bf37be06 : SUCCESS in 48m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dccb4ab6d82643ceb1aa0c8366b28aa5 : SUCCESS in 1h 03m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/21563ef74ce942dbbbd2714af38467f0 : SUCCESS in 2h 17m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47c9a585e3fe436992099bbb5ced7a60 : FAILURE in 28m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/35ea616ba6bb4430a79926a2bd664172 : SUCCESS in 1h 48m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cb33866ec66e403bb291bee6a8d98eb3 : SUCCESS in 2h 12m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b8a9a16424a94dc6bbacd9c8a8a2fbdd : SUCCESS in 1h 37m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0714251d8abb43919198db87a2aba504 : SUCCESS in 1h 47m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b37ac582bd01e18ff92d9aa2ab0491613a14edff","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-15 13:36:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c9728bbda57c25ffebe104badcb5a6d73849a58","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-16 14:54:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6f928f3e0185328bde4025719cd52c4ce80b3486","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-17 08:07:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8bb6c63c19d4261fdb5f5e1d8eba3b764b07b437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 08:07:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11069db483db2ff9d68e3c3a253f274f4b3ca284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 09:59:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1573b65e4a9459ca0e7453c1ca95414 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca33b7ae2e7547f594e4e7c7b6829839 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bd22e39ce8247318a297fda889488a0 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e05d5cade34ae2b7a95d5ea34ef8eb : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffdb9f9b22ce4ce5a500e726c25c1e47 : SUCCESS in 5m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/621430959d774dc489b1aabe2252ff44 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b63d69c68b54683bfbe3e8b4bc774d7 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f0644a4c96c4b038823a0a94f7d187d : SUCCESS in 1h 20m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a44cd9173b0e43ac81d63f8d1251d9f8 : FAILURE in 1h 50m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a5505c304494d9988ef4ac4170c176f : SUCCESS in 42m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d4928a1a96ec4fc2a73728f0d0b8d301 : SUCCESS in 36m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc8ee92cb9e04ed6bdeb3fc5f2c148ef : SUCCESS in 36m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0076c116d42476ea0057f8a3226fd8b : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"557c3fa32b824235d07e84c24d567a2d28080c61","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-17 14:19:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c118f2d723cf0e9a4b4eeff350982abeb04616b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:31:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b220d6dfab4465294364667fe7941f5 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/337a04c6bed54993822bf32ab069e2da : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/224a279840f74bbe98167d2d95812d04 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b245459f5984e4e83d83136c8be0516 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa416fe8c8e04375b56e2e72ee0bc0d7 : SUCCESS in 14m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8dae65fa3c346af98eb9fc2af2e19aa : SUCCESS in 7m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f908d63b14d6437bb8cfd351522ec620 : SUCCESS in 8m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5c0b7db3d0074957bd0895342445bd07 : SUCCESS in 10m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/05ad1845eaf1420ba9d62805b2dd298c : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ce9d37f9c6d4e4b84d494c5febcb231 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dfcea33358c749da94541679136dd833 : SUCCESS in 1h 25m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad14f4a5d13c49e4944230fee1bc4168 : SUCCESS in 1h 45m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df6bb5562852445ab4e7930d62ba5b9a : SUCCESS in 43m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/16a06e6a0ec646e8a917dc061ea66afd : SUCCESS in 40m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f5fff553cfd42a48bc570c598cd28f6 : SUCCESS in 49m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc5b259d7e144e9a9142009e7fce89a8 : SUCCESS in 1h 05m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77bfc7fab63f447c8dafa418f9f23cee : SUCCESS in 1h 52m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d03510924f674197b30b030c06f3aee7 : FAILURE in 1h 53m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/84ba0e6c56e74bab862c1c15c6ee9ad3 : SUCCESS in 1h 41m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0db4f04bc3d44c918d41dd7cf13b3c73 : SUCCESS in 2h 06m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c81e9023cdc14a7ea902b9fcfa830553 : SUCCESS in 2h 06m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6b5494b336e647b080eb12a0663d88aa : SUCCESS in 1h 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc92e497eda131b6618e9651dee9060c12f1afdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:33:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87d9e2af0d3adceea043ddb26a141a5469c6783c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:26:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8349977160b340ba9fae521007836329 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/397d7645bfa94ec68e62646e76317891 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffec753dd9784014b62f05c4d31b7078 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b400c0a374450f8e7153549699d696 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c26f0b368fc14d4ca9798be7ba7a420c : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f5db4e5c5ba4df79768810dacbff2fc : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04d671ee23074ef9a4cf41847c260988 : SUCCESS in 9m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d51a97e3a6144936a8dead2d39c21de0 : CANCELED in 51m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48482e9cd6154c7e9baffc9cf25803f0 : CANCELED in 48m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/065d412b8ab04a049bd60eaecc73ca80 : POST_FAILURE in 50m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f1ef105648534098a077ed37d085ac62 : CANCELED in 48m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78a8f20f14e649518d7cc48dbc5f47c3 : SUCCESS in 37m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/052964fd7a544f1d95d553baa0e1b9dc : CANCELED in 51m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f90e502b36acca7cf564e2edc2b3f8aeed155cf0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-17 18:09:09.000000000","message":"Patch Set 1:\n\nrecheck\nPOST_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8092c7c6d7951ad2e95ac5480f9c7dad743de5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 20:52:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db50c96620ac48af80338c045ad7b684 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1546c6c4d54c45e29716e427f95c06f7 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58071c6a83a048f1b49b9524877e37ec : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f025a2f8d7a49d99115a44bf422f446 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4596d4ef94744baeb844369bccd53642 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/558e490b5a174c0187b4d80da47085ef : SUCCESS in 6m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c69c0910c26a4c398787923fef8ab798 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9980397308644214bcd6ca440ffaf99a : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7225ab9df9064a29bddbe79c4f640c86 : SUCCESS in 10m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80819ffe079d436f90d55dcc64af36c8 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/75c5e29e88634eb6824e67edc7c90852 : SUCCESS in 1h 20m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/652d2ea8026648af8174b55770f448d2 : SUCCESS in 2h 24m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d037ff6905ac4b1aae5ad66bcbbb48cc : SUCCESS in 37m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5b71666d537944939f8730864388c1d4 : SUCCESS in 44m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/350ba183904a426497a274b39c53dde6 : SUCCESS in 43m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b6a182de4ca4e9f97c8ac7e0a59fc54 : SUCCESS in 57m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7039c6341921437fb4c9d2979d92e18b : SUCCESS in 2h 15m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/323bf952bae74483bdbfcbf0d5c8a9fb : FAILURE in 2h 28m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d9eadaf253684e36b69ebc41457a0942 : FAILURE in 2h 29m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3f8af56aaf914c14b242218689d07c06 : FAILURE in 1h 42m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/089a7548836047a8a0eb2b29511468c5 : SUCCESS in 1h 43m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/45d9f67262714741b71ee02a1dc71a3f : SUCCESS in 1h 54m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"715e7c1b3dcc47d21134b0928e5460a1eda5d607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 20:53:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dca2fad9a0db1f9560cc7da962f6e8b706f8491a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 21:53:11.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c5e6a4dd16fc5c1013e74ee9ff39af0ee337b90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 22:07:29.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc907cf656ea7d6b4af87c7202d6799d7ab52ce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:24:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e39d18427e6b414e85d29e23a61dace6 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e1c361971d4f4cabb92165b22968e6 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a57bde9f1d2493c9b9191482a85cedb : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a7d6df2874f40ed92e45994ae2c366c : SUCCESS in 16m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/069bae4be6b84534a709c3b972111f7b : SUCCESS in 7m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3223b972db5d445ea9c8972dd58407bd : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af59bf7156e2401d9ceca38cfff3de72 : SUCCESS in 11m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e722f6292d94fb3a82c1bc471233750 : SUCCESS in 1h 19m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/135ec2d428174b25aea7d0c8f4b93e2d : SUCCESS in 2h 11m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/732d183f253646c6a0a887b8384eda4e : SUCCESS in 46m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c413ccbe383d46848b3800fa213571c8 : SUCCESS in 37m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ebf89be60a94f66a99d07fcb4df87ea : SUCCESS in 48m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0300b9bfa0e4f978b8d6c26fb8166d7 : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"315afe3593662bb39be933036490afa91bb22386","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:24:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1aaf242b80962cba13b0274aeef609f61288d73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:25:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c957e6be4cf2458cb97ce0fa8b3ca619 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcd5955cdd454e98be232ee1b071ce64 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec","revisions":{"0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec":{"kind":"REWORK","_number":1,"created":"2021-03-15 07:36:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/80/780480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/780480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/780480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/780480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/780480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/780480/1"}}},"commit":{"parents":[{"commit":"857533b59cf1ef2b657ce74bc0de89798356ab93","subject":"Merge \"Add validation for VIP network parameters in amphora driver\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/857533b59cf1ef2b657ce74bc0de89798356ab93"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-22 01:40:46.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-03-15 07:25:07.000000000","tz":60},"subject":"Fix provider driver quota handling","message":"Fix provider driver quota handling\n\nWhen provider drivers registered a load balancer object delete,\nthe driver agent was not decrementing the project quota.\nThis patch corrects that by decrementing the proper quota\nwhen a DELETED status is received from the provider driver.\n\nChange-Id: I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b\nStory: 2008268\nTask: 41133\n(cherry picked from commit fc8ee42dfccfd4b0735185896c30b0899daa4c55)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
