)]}'
{"id":"openstack%2Foctavia~781483","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I2fe94cd4c000f143c59c69e82d03c690acf5e0c3","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I2fe94cd4c000f143c59c69e82d03c690acf5e0c3","subject":"Fix incorrect ERROR status with IPv6 UDP members","status":"MERGED","created":"2021-03-18 13:55:22.000000000","updated":"2021-03-31 18:49:00.000000000","submitted":"2021-03-31 18:31:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781483","meta_rev_id":"daf4eaa08cbc33fa2e980aadc8a7366f29736c71","_number":781483,"virtual_id_number":781483,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 18:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-22 09:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-30 13:17:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-18 14:21:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-30 13:17:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 14:21:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-18 16:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 09:29:09.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-30 13:17:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"3f54f4eb78505e52c7e07ad45bfb685f276bd4ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-18 13:55:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d240c26f2846113602924359b8c7d864ada073c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-18 14:21:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c8fefcafbeeb297c1d1565592ac698075ff6c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:34:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/845e374730924201a718a3031beecf44 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f3d2392d672466fb28d59c7f5daf004 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/957596eb7fcf49738f951fe6db62084d : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca927265ed24578a67559041e94ed9c : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9786e876c1574e659c1317de49dac0dd : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04fb1887da644bd0a6e977bcfafb055c : SUCCESS in 6m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5b7e9c2820ed40ff939b06a671d0a81f : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cf30f95d39e24b9a9547fd9cd3825b1d : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bc20b81b54c74445b55d9ea2f2752d3a : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f46ce94075024e83ac79f354f232bc9c : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/480158edec5a46098d95b038946f6d14 : SUCCESS in 1h 19m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6fcb3380ab4541ad9554228822b23f6d : SUCCESS in 1h 42m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c23905480b894457adab0b225be31e54 : SUCCESS in 36m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9b7ead2ada624f8ca80127e591ed55b9 : SUCCESS in 44m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c70f040e2dc43c5846c174870e3767f : SUCCESS in 45m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f5f2e607af4f4e1ca445a015a4a7288d : SUCCESS in 1h 00m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4e646b4557864cb6a4cef8f111d6dc21 : SUCCESS in 1h 53m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/986f22b16ce24da7bc37a68b3b866118 : FAILURE in 2h 08m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1c2d71dcf46448ee8938b831390f6fc9 : SUCCESS in 1h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/62563e0df3c74e8191dbeaff8367c0a8 : SUCCESS in 2h 03m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f7ab1d878f7e47fcb363c9cb9f445b03 : SUCCESS in 1h 41m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/035c1154bd01470ab6a4a65890b4e635 : SUCCESS in 1h 51m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77f9514014f768cc40c0fc9a6f87011e7c7c908e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-22 09:29:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff55be26d2b6dda48b30329a469063b3bf5f3c27","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-30 13:17:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63f4f70c8fb38717b5f44896bff1ded5f5a05d79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:18:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dba46fdd0e2eb0f12ce8b3332b8a783a7079c0ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:20:44.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cab89d273b13d544ded699207c97d752546a8e05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:49:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73cafa52ebf948f68e513dbbacb129c7 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c40dbae52ca4bf794d4b1ce1db4c0fa : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c4ccb8485f34b9ba6fc8ddfe4cf0b1b : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60bb2c961d14f60a2c0a67d37f06fe9 : SUCCESS in 18m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/283fc79bf74e40448a8ab40d4b55cfba : SUCCESS in 6m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aa77c9413b9c43a9a3fd7482d2fef2de : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7eb3b5440b14409f8fe0494f02849b84 : SUCCESS in 9m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd87599d4e86433fb3333cfd2609b02c : SUCCESS in 1h 48m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94638086b05948728c3e84373a110ec3 : FAILURE in 2h 01m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b86dc62d19740bab1ad4d1c7fa0c765 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f767554660bc4040bbcf30278e311c27 : SUCCESS in 42m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db687747a1814fb18e92466e90c1b67c : SUCCESS in 43m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/006ff29498c0405883e515358c290663 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ac67e04646b0b60a151ded4932fb3fdc63a46d66","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 13:09:28.000000000","message":"Patch Set 1:\n\nrecheck\n\ntimeouts while connecting to the amphora\nCould not connect to instance. Retrying.: requests.exceptions.ReadTimeout: HTTPSConnectionPool(host\u003d\u0027192.168.0.143\u0027, port\u003d9443): Read timed out.","accounts_in_message":[],"_revision_number":1},{"id":"51600b6dc2aef505c7ac1d72f6a540db137a9a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:15:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/463d8880fe04426ea5009177299ee44d : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8336c86ec7be46b79b6efba6a3ae0843 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/278055108716467cb00b78609e5e7af7 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0faf8ec1812540a29c44aaafe56888e2 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc697c106f7407fa920f5ac26cecc4c : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06099ac9613f47ad8f3d14310727bbab : SUCCESS in 5m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/afc769c0511c4177a08cb6c9de2799b5 : SUCCESS in 7m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/47890f109c1349b9ba5231c77e27c68d : SUCCESS in 9m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/532f2959e31846f4b7f303ed3b645da6 : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05fc2cb7d3554ba09ca5a9c6a89deac9 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0522af7eace8471999bba8eb57e375e0 : SUCCESS in 1h 25m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8f8b4ecda434516bbefecab8da734b8 : SUCCESS in 1h 47m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d02a5ea27b442588e4ea5b49c03551a : SUCCESS in 38m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ce4b81bedb65470993f832c0ebb23f6f : SUCCESS in 38m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6aac78de65ca4e719336e94b76b966cf : SUCCESS in 37m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68f565bf540e42f496eabceeeac0f4da : SUCCESS in 1h 28m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e4ee2479f33a43a18fca40542f00340a : SUCCESS in 2h 14m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a7feaf96cff64097a5a170d25754d323 : FAILURE in 2h 21m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3a400d7f9a4f83b41fbc9b0bec874e : SUCCESS in 1h 40m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/30d9a91f19a34bdc84502a0863632ef9 : SUCCESS in 1h 51m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5dc734125e6d42e6a20f0f265fede3f1 : FAILURE in 2h 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bcb23ca551b44ce2a4bd43e1082d4980 : SUCCESS in 2h 09m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b485f0b82f6c7b945059c589e997ef1b31492cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:16:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"792d3fc99f646699f89edc2eea98e41b4e94ddf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:31:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb1a13411a64c618910ba4298c69ba5 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7daca8b756540618cf03a3713201100 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/444d2a880a404ffdb1f31e1b606aa8c1 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149dcde158bc438b8ec7cf996c817bad : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40b9850c21bd4afeb0e0ae4c175fde56 : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f720cc8e3924a3a8d4d6250a8e7cfe3 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3739b68c3b1441ba8ae73704c7d067d3 : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8873c7ec2cbf447380273b6a5929b029 : SUCCESS in 1h 43m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/857b6b177c5744bda0836a8f497ac41d : SUCCESS in 1h 40m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee26316ea48a401a9719d46b42bd65f8 : SUCCESS in 39m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b8e260597d1b4bbd885874ed56133886 : SUCCESS in 35m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0501e49a60c144009d76275367ea0abf : SUCCESS in 50m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05a514fd3c9d47279ddd5900027df7ed : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fa577ef01ebf98a16ea1657a570512d919b56e29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:31:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"daf4eaa08cbc33fa2e980aadc8a7366f29736c71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:49:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96bdcacb685148c2a3403c6440c927d8 : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b518d2eef58f4996a2e052db0521713c : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b210bb52a6c44b3bd69acdf2202d1625e763fb6","revisions":{"8b210bb52a6c44b3bd69acdf2202d1625e763fb6":{"kind":"REWORK","_number":1,"created":"2021-03-18 13:55:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/781483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/781483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/781483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/781483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/781483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/781483/1"}}},"commit":{"parents":[{"commit":"0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec","subject":"Fix provider driver quota handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-05 21:03:21.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:52:12.000000000","tz":60},"subject":"Fix incorrect ERROR status with IPv6 UDP members","message":"Fix incorrect ERROR status with IPv6 UDP members\n\nSome IPv6 UDP members were incorrectly marked in ERROR status because of\na formatting issue between the keepalived configuration file and the\nipvsadm output. Both are used to compute the state of the members and\nwhen a member\u0027s address contained \u0027*:0:*\u0027, parsing was incorrect. Now\nthe health message generation function uses only the compressed IPv6\nnotation instead of mixing notations.\n\nStory: 2008604\nTask: 41783\n\nChange-Id: I2fe94cd4c000f143c59c69e82d03c690acf5e0c3\n(cherry picked from commit e5f9f6708c70313bd7e0c570f3541c0afb7fc09c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b210bb52a6c44b3bd69acdf2202d1625e763fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b210bb52a6c44b3bd69acdf2202d1625e763fb6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
