)]}'
{"id":"openstack%2Foctavia~781485","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I2fe94cd4c000f143c59c69e82d03c690acf5e0c3","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I2fe94cd4c000f143c59c69e82d03c690acf5e0c3","subject":"Fix incorrect ERROR status with IPv6 UDP members","status":"MERGED","created":"2021-03-18 13:57:09.000000000","updated":"2021-03-31 18:49:01.000000000","submitted":"2021-03-31 18:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781485","meta_rev_id":"37aa59d0e3123fc051c74c0a5c737f9d67d1c69e","_number":781485,"virtual_id_number":781485,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 18:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-22 09:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-30 13:17:12.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-18 18:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-30 13:17:12.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 16:25:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 18:24:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-22 09:29:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-30 13:17:12.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"0a14442b32dd35f82a1e994ce90122c26b2e8124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-18 13:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0792b8701cea30e4a6629b8321c618035f221713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:25:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/254265fee05d47ac9bb447b8d8e0985e : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9cde72428b4bd8a7e212fea0d37de8 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84ba5c2716cb48c08b07c6ed800e77d1 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0f43369f73b4e8085b2df43b64e8870 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5eb8894f0d54ed29f44df8bc5154898 : SUCCESS in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8c6c09a8dc24b04aed62f85477a6d4d : SUCCESS in 7m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/be9a626f60b046f6a1cd87ef6e0a2a56 : SUCCESS in 7m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/531fbc5f6b14485b90c1276c2b42cd2b : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13d84d4ca7a443bc89095dbe2776a47d : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8bd38fb4d9140ad9de252ccd28acb17 : SUCCESS in 1h 14m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a420c72dfcd6432aa3af20bdd1a8aa4b : SUCCESS in 1h 36m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb5872e1390944fb9cec4846e007702d : SUCCESS in 46m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/469a0065707f457192edaf4a0ca5aa59 : SUCCESS in 44m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b541dacc25f64aa88f48864271ca5a88 : SUCCESS in 45m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54d85ea070124131bdd55ef4ea446998 : SUCCESS in 1h 04m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2aa634c08e094172a0db371c761359ad : SUCCESS in 1h 52m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3f4ba585edc24581990ad5831aa8a5e1 : FAILURE in 49m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e7ee2bb48dad4d2e88cba281ccbc813e : SUCCESS in 2h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c6cb8cee0fbb5fe8cf6b043a2f80297e4ca90e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-18 18:24:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"000c286caa9f5432269424e8cb87f6ad998d885b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-22 09:29:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b2d2684639da6076ad6a693d94578ee9bb31da1b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-30 13:17:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff2227c09bb61b149a51de5aa0148c44c0c2bba0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:17:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3153fa9f9228df4a3bce81361c4d7726c0b2c3a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:20:41.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a145b6ded89adcd09cede50fadbb32729c673919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 22:43:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551ec1ca8a664786918384db920afca6 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ba52dcaf0a7413a82a08ac39147faf9 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e20927215ca41ea9cc58cdf2dc219fd : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1bbfd9024c4a43b4bef3d6e02fe06c : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aad682c4e4214447ae316d584893091d : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/734b208234c440608ed05f3f3c7d69f2 : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/beb40adaae7c41588e51e778bec26f55 : SUCCESS in 1h 31m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0682f14f44248d68ded3638e9fa66d1 : FAILURE in 2h 07m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ffd94f1f685e41b5a82dd534c60b9333 : SUCCESS in 39m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6a8387d757184eb590b6d0e9388a7c34 : SUCCESS in 39m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f121bd529f28436787bec97665a0eee2 : SUCCESS in 1h 00m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87e52a6ec48d4986b17fa37d66ac9638 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"95fbcf6cad48058f987c421d55f34895e0c124e9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 13:11:19.000000000","message":"Patch Set 1:\n\nrecheck\n\ntimeout while building the amphora compute","accounts_in_message":[],"_revision_number":1},{"id":"aad55d8b63bd3c269eb4a3a74552ddda6d28b3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:16:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/024a6439f2184412a4ed8e66031657fa : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf8e6e1e7fe4ed79f30efd8699376df : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcaa4da3cb714d948ffd9e499febe53c : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a12001e432df4ac984032b80ff9f2897 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb151238f484b089b49e213aa950fb6 : SUCCESS in 15m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bbfed9c075740749c858f9720dddd1c : SUCCESS in 8m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6d8a59545e3e44e8af56d4349885a973 : SUCCESS in 7m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7cc188c4c0a048ccbe916ddf675d74c2 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5a2ffd32a5d4ba9b8ef6ed13b42603e : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/330d51aebd84450594350622d2bd07a3 : SUCCESS in 1h 35m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ec118c76543462c99b1ff8b8ef47117 : SUCCESS in 2h 19m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/668692afa9114793824831411847e7e4 : SUCCESS in 1h 04m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fab2a3942a384635a85005fafe435e98 : SUCCESS in 33m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01e3429d262a4b119c54c822e3cf7757 : SUCCESS in 40m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/366dfd81ba5846fa8183913b5b545a10 : SUCCESS in 1h 23m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1936977c39d448d9ad3bcaf38ae07722 : SUCCESS in 1h 40m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1a6549844534cd58b210aacfb8145dd : FAILURE in 1h 12m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d48ec8e4158b477daa5f7edaff4b73a1 : SUCCESS in 1h 52m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a1c5ce583a6c354905f14391c6d63bb31e7f5ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:17:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ceb4f96bcf70873d2018411fbeff30ed56969e36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:42:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4963716e947465cacfccd80184c47be : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dca52b8905024610adcb554338fbceae : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8427317b41ac4bb6afaf8ca7aca8a4a2 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8ff787be424bf29225a64565480bca : SUCCESS in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a869872467cf4e3b9fd2455234b96fcf : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa7e8736245a4718bb80e6be1dff6de0 : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f76dd36b36b414fb0ca310c7554c9a7 : SUCCESS in 1h 17m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b19611db68334559bd04fe0c2ad78202 : SUCCESS in 1h 53m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a1f692b0b60646c89bfbf70a0b2182ec : SUCCESS in 48m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2e2b943211f143dea11084152972206b : SUCCESS in 45m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56eb60659841443688b5e36a294b9511 : SUCCESS in 43m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aedac39640d14d83857f7fcd6897817b : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7346a0e9b0fc99981d63f363ef4b32a1c41f57da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:42:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"37aa59d0e3123fc051c74c0a5c737f9d67d1c69e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:49:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe70cbd90484585b3db7b34a61085b1 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11c9c31460b74c9aaf471a824606b523 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b8de86cf3b2347e38d3a39435c8b609fb055404","revisions":{"4b8de86cf3b2347e38d3a39435c8b609fb055404":{"kind":"REWORK","_number":1,"created":"2021-03-18 13:57:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/781485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/781485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/781485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/781485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/781485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/781485/1"}}},"commit":{"parents":[{"commit":"dcce548ed6a85b0a17dbcba78a14e0166dd282cd","subject":"Fix provider driver quota handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcce548ed6a85b0a17dbcba78a14e0166dd282cd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-05 21:03:21.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 13:56:46.000000000","tz":60},"subject":"Fix incorrect ERROR status with IPv6 UDP members","message":"Fix incorrect ERROR status with IPv6 UDP members\n\nSome IPv6 UDP members were incorrectly marked in ERROR status because of\na formatting issue between the keepalived configuration file and the\nipvsadm output. Both are used to compute the state of the members and\nwhen a member\u0027s address contained \u0027*:0:*\u0027, parsing was incorrect. Now\nthe health message generation function uses only the compressed IPv6\nnotation instead of mixing notations.\n\nStory: 2008604\nTask: 41783\n\nChange-Id: I2fe94cd4c000f143c59c69e82d03c690acf5e0c3\n(cherry picked from commit e5f9f6708c70313bd7e0c570f3541c0afb7fc09c)\n(cherry picked from commit 8b210bb52a6c44b3bd69acdf2202d1625e763fb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b8de86cf3b2347e38d3a39435c8b609fb055404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b8de86cf3b2347e38d3a39435c8b609fb055404"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
