)]}'
{"id":"openstack%2Foctavia~781486","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I2fe94cd4c000f143c59c69e82d03c690acf5e0c3","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I2fe94cd4c000f143c59c69e82d03c690acf5e0c3","subject":"Fix incorrect ERROR status with IPv6 UDP members","status":"MERGED","created":"2021-03-18 14:00:51.000000000","updated":"2021-03-31 02:46:56.000000000","submitted":"2021-03-31 02:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781486","meta_rev_id":"e069525ef3f9284fe1a0abbab7856503dc8ab331","_number":781486,"virtual_id_number":781486,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 02:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-22 09:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-30 13:17:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-18 18:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-30 13:17:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 16:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 18:25:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-22 09:28:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-30 13:17:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"00a5b45b45806ef7cb27d1593a92a227069a66f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-18 14:00:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24b83eebcf2964deaf18c82a818f6a771b2ad542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ada136b14677485f90933a9bf63bc9d2 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f096580b4e224668b2a7fc2dc7e15b16 : SUCCESS in 8m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d10d93d3e19640e7b5fbe4b70e203758 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20b3aa52e8e443a08f910a18fedb4a99 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52239344a3bd4577a714c6aaec2071ba : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14460b1870f140f98c2efff5f3083994 : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f831ce5824b440b2a58e30ad269597d3 : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5ed629ed139b4ab1bfa9aae89c9908fa : SUCCESS in 9m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a72f1922f8ab485b9c003df7167fcd7a : SUCCESS in 12m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7e71cdc8091d4170a55da91efc03a77f : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90e4a7916af44351ac74fee25cd5c751 : SUCCESS in 10m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93e056c8e4da4def878c211bbd683057 : SUCCESS in 1h 21m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e35cd628c0414e4189d34cfc29a46326 : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b10ceac4f6e0485882c4592e3d59b423 : SUCCESS in 1h 47m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/de7106e7dc3a448a993f93a6b49a0456 : SUCCESS in 2h 04m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/83d8b26e2c7a47158a663026e6ad5b30 : FAILURE in 11m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3602b03f15e24ebeacfdf5977a179343 : FAILURE in 49m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f833feb2a59453f8dc7715bb34e4125 : FAILURE in 43m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c62767dd23b142a1a1ea0185af3462ee : SUCCESS in 48m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/09c0b5d0f19c4594bfe2f9dbc515d107 : SUCCESS in 2h 09m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6192e6c100bb417aabb9ac924d5cf2f7 : FAILURE in 55m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ee00d8cdd92f4c84adcd65ae3b9b51a1 : SUCCESS in 37m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f42d52ac49294fea9cc98c700b544a70 : FAILURE in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd2baa79cc9dcde5f5fa170f3e6572829121f24c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-18 18:25:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b30e231f9f152ba1d6f41488a02222b2f5df032f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-22 09:28:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca402f680dd51e618bead10df1988cd85d31ba2a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-30 13:17:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8eb2d45337d5a69d360a94523560d242800374f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:17:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b20940b75873be1cad7fe81209e71ec28580af4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 16:41:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba4b34f712942c89da6679c34b34404 : SUCCESS in 10m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34fffc00a31d41c0b31605d7b9651c98 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/275ae9ba2c5a47669747ba8a7c79d31f : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4deec1288e704fb48a7676a23de5c5c6 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da16f21cc0a4d61b59982ee3eca3db7 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb9ec07a52364838b9c72318b97fdca8 : SUCCESS in 6m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a3ef86ce28ab4408ac34644608bca39c : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/427d01c1c3ba4d259537bd732237dc22 : SUCCESS in 9m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8e53cb29d9145f9a464d279c39df095 : SUCCESS in 1h 17m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b963edd8dffa426bbd8e370a65e8eae4 : SUCCESS in 1h 17m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/344094723ee74b4f9bb03a7446be487d : RETRY_LIMIT in 1h 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"cd6eb0df93a4167e928da414de3a6775752a8386","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-30 21:42:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c4212b30659a1ef492c37b72979e1e67a05c3c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1930102535c44d4aaf2cafd3a5bf930d : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b0b21c00b54cecbda9169da255ff89 : SUCCESS in 9m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b972f41922fd4bd78503bcfbfc53e816 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b77033a3a52d420dacf6b72f3c33dc6b : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbe767e5cfb546beb8a2031341691241 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1d87e59bd1455ea3716aa55f7251bd : SUCCESS in 15m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fe94326b64d424ab4aec42b419c15b8 : SUCCESS in 7m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c50b34537f3e45738713aafa65c0d648 : SUCCESS in 9m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/13332dae3ab3430fa72271c14166362c : SUCCESS in 9m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/39afc8c20bb24da390e4056190e9a877 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c95a69f18bc648f69b71af069e748395 : SUCCESS in 10m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31d04647af144377978802ffc8f44027 : SUCCESS in 1h 19m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/424b05e9aea1416199e38ae37d7ced56 : SUCCESS in 1h 31m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06c43f7dde4d49d590c60d1a8cb54041 : SUCCESS in 1h 30m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a523b5de34f24dd4a12ba850650edb49 : SUCCESS in 1h 48m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9edbf454baf74e49b8dda4b5552507c6 : FAILURE in 13m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4d1a4c86a944413867ecad448d3159d : FAILURE in 51m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/68dc6016186a4bd3a4b8a9b410a1811a : FAILURE in 58m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29c779a4154544c893843202fd38d231 : SUCCESS in 44m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/facd2bdf8113457d86c3e7f5235e4e0d : SUCCESS in 2h 13m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/450b0063115b4fb19c7501b69a89e936 : SUCCESS in 45m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b558de3f39d04cfabf102e8fafd9df85 : SUCCESS in 47m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/407c590e9f6f40a88352710f6bed5679 : FAILURE in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0587d7753e6da4a9662f23a549a83ad76b184f59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:17:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f582d61df9ac0f08a074a453eb2ebc087ca689b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:45:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d672c6f346ac419e88caa814f2a1e004 : SUCCESS in 9m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/80ed5ef0e5174c0a86f5525d925bb834 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5409bf9c2dd943e292d761241df220f2 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6c911b27c554d2eb2f389720ccb34d8 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54515ffaf744e9c879dfbd58a7d7137 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/892016698be940eaa396a8cc1d1a878b : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8bebf4855c894e9aaff2d75b0aacfac6 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c916931bcc49421f8d6354be540726d2 : SUCCESS in 9m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59b8846c050047c9a811acf3f6089b9f : SUCCESS in 1h 14m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/feeebbb744a048388ab6446147152dcf : SUCCESS in 1h 22m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/460d17c36e0a41e8a5a47a492020f7d6 : SUCCESS in 1h 46m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2aac4caf0487813a1cc14962897d4afb6915ecc6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:45:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e069525ef3f9284fe1a0abbab7856503dc8ab331","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:46:56.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae52ddd24bf742c9a6871340d6aa51a3 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95d0d88e0d2e46599ca3c833d67a1712 : FAILURE in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"411a0b7bef2c5a09c38b949b5884d9b0def53a27","revisions":{"411a0b7bef2c5a09c38b949b5884d9b0def53a27":{"kind":"REWORK","_number":1,"created":"2021-03-18 14:00:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/781486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/781486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/781486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/781486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/781486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/781486/1"}}},"commit":{"parents":[{"commit":"93eb3c8cd92587f633a8d2b7920148ec49d95432","subject":"Fix provider driver quota handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93eb3c8cd92587f633a8d2b7920148ec49d95432"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-05 21:03:21.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-18 14:00:27.000000000","tz":60},"subject":"Fix incorrect ERROR status with IPv6 UDP members","message":"Fix incorrect ERROR status with IPv6 UDP members\n\nSome IPv6 UDP members were incorrectly marked in ERROR status because of\na formatting issue between the keepalived configuration file and the\nipvsadm output. Both are used to compute the state of the members and\nwhen a member\u0027s address contained \u0027*:0:*\u0027, parsing was incorrect. Now\nthe health message generation function uses only the compressed IPv6\nnotation instead of mixing notations.\n\nStory: 2008604\nTask: 41783\n\nConflicts:\n    octavia/amphorae/backends/utils/keepalivedlvs_query.py\n\nChange-Id: I2fe94cd4c000f143c59c69e82d03c690acf5e0c3\n(cherry picked from commit e5f9f6708c70313bd7e0c570f3541c0afb7fc09c)\n(cherry picked from commit 8b210bb52a6c44b3bd69acdf2202d1625e763fb6)\n(cherry picked from commit 4b8de86cf3b2347e38d3a39435c8b609fb055404)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/411a0b7bef2c5a09c38b949b5884d9b0def53a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/411a0b7bef2c5a09c38b949b5884d9b0def53a27"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
