)]}'
{"id":"openstack%2Foctavia~782279","triplet_id":"openstack%2Foctavia~master~I39391070cea170a6039f901093f09fc89ba06123","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I39391070cea170a6039f901093f09fc89ba06123","subject":"Fix using subnets with host_routes in amphorav2 driver","status":"MERGED","created":"2021-03-22 15:32:00.000000000","updated":"2021-03-24 23:58:50.000000000","submitted":"2021-03-24 23:57:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"782279","meta_rev_id":"3f2b2d89620c7eed7a743de4f22823d616c02209","_number":782279,"virtual_id_number":782279,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-24 23:57:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-24 09:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-24 21:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-03-23 16:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-24 21:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 18:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 23:50:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-03-23 16:40:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-24 09:39:10.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-24 21:43:09.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"7a99ecddb1046482370f2cdd1e05cb41421b69ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 15:32:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c21d25028547c04779a3b863877274c6bd2eec2c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-22 15:32:14.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"bca6dbf4cb04e2ff958c7bbd1e44359e56b8c786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 18:23:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16653f30cf2b45bdbc37c2295fc316bd : SUCCESS in 13m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f55abab5f270413596911ad1a18d1457 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a102aed855e84e3eb8e05b68997d6096 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea14ab2321f4a23b0c5af6db46fdee8 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/059f52c4069e4b7bb3b87d825b8260a5 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/031fb80838c44b6195005e876faf9d71 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b46f3017f7546689ff928e8b62a29d3 : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77fee1d830824999afb9fdf0b1181f34 : SUCCESS in 4m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/597d3c62366f4c06bfa633248b17947b : SUCCESS in 9m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5a21306f81c4474789cdcd0c23516aa7 : SUCCESS in 15m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/594d9aa57c5040d4b8275f7ad1dbeecc : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9eeacbe234204204a3405ccb6c9a87e2 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae7410b971db411ab57d5e9715a991a3 : SUCCESS in 1h 41m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89fc610b9a524160a1e7767fd6b74ad6 : SUCCESS in 1h 52m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c53bad180eff4e439100674f55a36fac : SUCCESS in 39m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/61ebe64cf7634066a56767b28e26c2f2 : SUCCESS in 35m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c446afdd02bc45d1b452f54ca381bf95 : SUCCESS in 40m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/427dab6f62fd4939bb614fa19e9519cd : SUCCESS in 1h 03m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73542d32a29549d1b4a66ffb0a6b5cc9 : SUCCESS in 1h 44m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/75290c58d2864db4bb71370f78673b89 : FAILURE in 48m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c559a6ca1dbe41b4a02ce11f56e66cbb : SUCCESS in 1h 31m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0b90ad8eb7f1453c849bc99c09050cdb : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d576318c8f3d45439e2ddd9886c1f6e5 : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b7ecea194d6a4db2bf74e01430ac75ee : FAILURE in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30e887038d181a706adfbeed65af0f268f654a18","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-22 23:50:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb2af16fd45f30cb821b0daaa333027b72d1403a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-23 08:01:07.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd5b89eb849db9658aecb88ec9533913ff6f5880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-23 10:37:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d0ce06b8a67dd536b301b7a57a2a43a63ed47991","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-23 10:38:30.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"ef194f84a036672adb0c8743b46bd9883feef9ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:42:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3ffc3953a964528af98d77ddc437460 : SUCCESS in 12m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a285c3400da8482d81db4a9275ca7a1f : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d1160b81f34a1a9ef68839b30f9c42 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b58dd99172864837a77cff6fe46d0096 : FAILURE in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8cbfac2229c41aa830167306f3ffa3f : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e20658ccc7a47d184b16982b4a3efbb : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd61d68b06314cd08cab7489d0ff72df : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a2c2e07d6f4ffb884cdc98e7069015 : SUCCESS in 7m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f1b2ed32cc3042e69aeaed4e444baa7e : FAILURE in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3d1096875740465dbc6aaf190676ff69 : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6fa93799454344df924663781add7073 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e063f0ac34c4f40a71715feeef0250a : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d37f314f828b43f09085f5f8e2708f5a : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d52b59f1b04402dac2b920ab5259cfd : SUCCESS in 2h 53m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee32463008b4427eb36f1195582e2454 : SUCCESS in 51m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4c5af3a86f834b1387ad2f9c3312c554 : SUCCESS in 45m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45c38552157f46efab3db83ca4a92aea : SUCCESS in 40m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1d026f40d95545eead39a696d654e974 : TIMED_OUT in 3h 03m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/70acc3d4fa464d6c886adbcbbe6f7ff8 : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/98ce1e66d1794089a6d5497dbd8bdd4b : TIMED_OUT in 2h 32m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/09aede5e2b0b4413a8a95e4a3a9947c5 : POST_FAILURE in 2h 22m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a7bcba3eca10450281e08629d778186d : SUCCESS in 1h 52m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3fa0e5a986fc4e0da15e455330acf74a : SUCCESS in 2h 23m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1c796c2d6c724ed0b16a36c218fa2c08 : FAILURE in 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1813c320c2bcebcf227136d2f7b25a202688f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-23 14:10:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e94f75b63b7c38984a8d5719f1f35f1e9499c699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 16:18:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/446153a752f94d9f9816a0656fb77617 : SUCCESS in 12m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/126eae6865714ab4bc82e48cbd8fbc1c : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec715ee4f5f4e94a1e59f770936678a : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b1a0766032a49dca8d4677cbbfcfa35 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/622050554e464bbbb2ae83683c2cf92a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98c56eec0ac64484b84d338eca5b6beb : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694d3b4bf00344818110c6293c940e66 : SUCCESS in 15m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5763518d8cc643b8b7f12dfe045146e1 : SUCCESS in 5m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/379b3d40e57b4d80baebb5dbb3a967b4 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cfa80b672302426f9892cbc872b9691b : SUCCESS in 8m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d01d39b4b0a64506a8a0f97a3b942734 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed31e45cbcbf4e3fbd1dc710b3e65e05 : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bda106509d5c43468bbd1fb93072e04f : SUCCESS in 1h 21m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/122cf89efc08484f9bbe9390738a2a05 : SUCCESS in 1h 41m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af89258286474701b4ad8f07c2466fdd : SUCCESS in 41m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/46c458d152914007ae83ec3cfeb33101 : SUCCESS in 38m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cba87e78d3a4c708db5f19e77a20ef6 : SUCCESS in 42m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6f841c6cf8384c9e9cba02b78d5ddc57 : SUCCESS in 58m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f233f34f259149599b6dfacd02d1b570 : SUCCESS in 2h 04m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65cea9e3de8d49a4ae337a6af7e397b4 : FAILURE in 43m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4c801b2d56c04e249079fd528d1510a7 : SUCCESS in 1h 43m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e5e013ee981241edb24c6c0eafebcdf2 : SUCCESS in 1h 51m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f233b89caedf48bbaaa811e97de48663 : SUCCESS in 1h 41m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bddc2095d3374ee1bfe685ab0dba605c : FAILURE in 33m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"00ba6b510807029e0625c73fa8e2628764b7a694","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-23 16:40:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3b0f169734444e0bd1c3751533314c76d426197e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-24 09:39:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"95e2eff527caf7bf3d2c4fcfab7bbdf52d53735d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-24 09:43:26.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"65333b4135fca04ecc810016ef73c9484fbbe835","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 11:46:12.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/47fa254185294861ad6a82bc4eef04ca : SUCCESS in 1h 50m 44s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/a35f2a07096e4cfa87db3854e8341aa0 : FAILURE in 1h 42m 07s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/7d7733cd19b948419eb3f8fc716988bb : SUCCESS in 42m 54s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/3a389bb173c44fb5a7acd23dbc41a701 : FAILURE in 45m 57s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dfac676b33d24d1d9ef9900996036bdc : FAILURE in 53m 47s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/52276b6908e441c7957db6f1f6e7744a : SUCCESS in 1h 05m 10s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/b006b75f93dc46cb93268a9875447c98 : SUCCESS in 2h 00m 17s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/1af174ef9635468bb5d3b59e12715898 : FAILURE in 55m 28s","accounts_in_message":[],"_revision_number":3},{"id":"3f1ae368bcd2c4148504e0a1d4131acafea345c4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-24 21:43:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"95977d1c1fa131aa45de44f70fd7aa79f1564770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 21:43:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8f4d6a0140eaf0e169b2fa4e980d74c8350c9e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 23:57:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a93150c084f14032831b21d914a17578 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3888d5ecdd5a4a6b866ac7a78b6b7064 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe20905920ea446cb4458c29b71332ca : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3707c9d7364d457abd702b0d460a84db : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73e357652a34ef4b090f86340d1e6f3 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49b472f830794f73b8119dd8fc8f17e4 : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe1ae1d764fd469ba483ea66c149b04c : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bbbcf48f5945447fb89e8d091d271291 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e22cab1a474c4b5c99263dd84e1ee9c2 : SUCCESS in 1h 25m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84fe175f305a4e72bc6bc53dba8b4297 : SUCCESS in 2h 12m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2a882af8e65745538ed8d05f05a924aa : SUCCESS in 1h 17m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f5b462acf78d4ab69ce5d2909095dfb6 : SUCCESS in 38m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a597a3c5e71d4b259a4a70bc625bb6a8 : SUCCESS in 40m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/13f98a1772be4a30a5c10586b3a2a556 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":3},{"id":"705f87b5afbfe746a64723db02723244bb04d652","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 23:57:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3f2b2d89620c7eed7a743de4f22823d616c02209","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 23:58:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fede35fc8db40f9b590f304998f672d : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dc76c5ed901499ba3f98cd71dbb1833 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0ec3aaf23eba0ffc8fe7bc86b51d4857e0cc507","revisions":{"8d8dadd2ada9730d5528a36353e10cfdf0d5fcf3":{"kind":"REWORK","_number":1,"created":"2021-03-22 15:32:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/782279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/782279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/782279/1"}}},"commit":{"parents":[{"commit":"6c54eab5b5d8fd5300cd5d62e23e6cadc1641636","subject":"Make /healthcheck cache results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c54eab5b5d8fd5300cd5d62e23e6cadc1641636"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 08:15:20.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 15:31:46.000000000","tz":60},"subject":"Fix using subnets with host_routes in amphorav2 driver","message":"Fix using subnets with host_routes in amphorav2 driver\n\nWhen using subnets with host_routes in amphorav2, the host_routes\nattribute in the Subnet data structure was not correctly converted to a\nHostRoute data structure. It triggered exceptions and failed to\nprovision the load balancer.\n\nStory 2008738\nTask 42092\n\nChange-Id: I39391070cea170a6039f901093f09fc89ba06123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d8dadd2ada9730d5528a36353e10cfdf0d5fcf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d8dadd2ada9730d5528a36353e10cfdf0d5fcf3"}]},"branch":"refs/heads/master"},"e19bfa18d8beb39917ac49c119fcbdd1e568ee94":{"kind":"REWORK","_number":2,"created":"2021-03-23 10:37:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/782279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/782279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/782279/2"}}},"commit":{"parents":[{"commit":"6c54eab5b5d8fd5300cd5d62e23e6cadc1641636","subject":"Make /healthcheck cache results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c54eab5b5d8fd5300cd5d62e23e6cadc1641636"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 08:15:20.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-23 10:27:23.000000000","tz":60},"subject":"Fix using subnets with host_routes in amphorav2 driver","message":"Fix using subnets with host_routes in amphorav2 driver\n\nWhen using subnets with host_routes in amphorav2, the host_routes\nattribute in the Subnet data structure was not correctly converted to a\nHostRoute data structure. It triggered exceptions and failed to\nprovision the load balancer.\n\nStory 2008738\nTask 42092\n\nChange-Id: I39391070cea170a6039f901093f09fc89ba06123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19bfa18d8beb39917ac49c119fcbdd1e568ee94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19bfa18d8beb39917ac49c119fcbdd1e568ee94"}]},"branch":"refs/heads/master"},"d0ec3aaf23eba0ffc8fe7bc86b51d4857e0cc507":{"kind":"REWORK","_number":3,"created":"2021-03-23 14:10:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/782279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/782279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/782279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/782279/3"}}},"commit":{"parents":[{"commit":"6c54eab5b5d8fd5300cd5d62e23e6cadc1641636","subject":"Make /healthcheck cache results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c54eab5b5d8fd5300cd5d62e23e6cadc1641636"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 08:15:20.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-23 14:10:45.000000000","tz":60},"subject":"Fix using subnets with host_routes in amphorav2 driver","message":"Fix using subnets with host_routes in amphorav2 driver\n\nWhen using subnets with host_routes in amphorav2, the host_routes\nattribute in the Subnet data structure was not correctly converted to a\nHostRoute data structure. It triggered exceptions and failed to\nprovision the load balancer.\n\nStory 2008738\nTask 42092\n\nChange-Id: I39391070cea170a6039f901093f09fc89ba06123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0ec3aaf23eba0ffc8fe7bc86b51d4857e0cc507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0ec3aaf23eba0ffc8fe7bc86b51d4857e0cc507"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
