)]}'
{"id":"openstack%2Foctavia~783465","triplet_id":"openstack%2Foctavia~master~I9e37db9f45fe1b8880cc1034aee32a551fa339c7","project":"openstack/octavia","branch":"master","topic":"add-xena-python-jobtemplates","hashtags":[],"change_id":"I9e37db9f45fe1b8880cc1034aee32a551fa339c7","subject":"Add Python3 xena unit tests","status":"MERGED","created":"2021-03-26 19:14:22.000000000","updated":"2021-09-24 13:45:24.000000000","submitted":"2021-09-24 13:39:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783465-add-xena-python-jobtemplates","meta_rev_id":"5280be2acf07eaccfc325fc971aa7d7b786a9776","_number":783465,"virtual_id_number":783465,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-24 13:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-03-29 21:02:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-03-26 19:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-03-29 21:02:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 19:16:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-26 21:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 21:02:31.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-09-24 09:19:01.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"6fe13ef52d2baf90c458941d5b9ea21155632d8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-26 19:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74f9f94c50fbf233e1b5f373dc740ce3ece6967f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-26 19:16:40.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"24e419009e5ffae754df4efa4ca5b5819c910025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 21:26:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d877aa1c453f48719022a44e184563f9 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a169e5dde89e4a65be330672bb24a1e8 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfe49d2f3f443ef9d97b6466dacdd10 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/896111dd592e44d1b03e04062a043e2e : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b430ff1268f48d7a33536dd4bcaff95 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f90c31da970545a3af3fecfd75b731ed : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b730a949467a4630b85ebdaf30f4ac05 : SUCCESS in 11m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/064e35f0abec4628a1de5d2182193d6c : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/145ea7434f984ea19f1f1ab1a8341159 : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/121a58a97b014ce4a021182fb10a4438 : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8c70d40633a45c0abca43c2ce84496f : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fffbb3849afe4ff8bda9db96b281ab6b : SUCCESS in 1h 02m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c2b98c34ee24182a7607ff6abbe3482 : FAILURE in 25m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/080a44ba6deb45deb53d04e9beae89dd : SUCCESS in 40m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3c0fd57879004b1eb4da886dc12b3515 : SUCCESS in 44m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af94925fc03943fbbc9b13c42daaf2da : SUCCESS in 45m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6476f390f5b7479db105cc242182d4e1 : SUCCESS in 1h 07m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/93194104aebc4372832c32285269e1a5 : SUCCESS in 1h 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2799ecc934824870b587e8b504fe0e63 : FAILURE in 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/edd0eaed505745ffb960fff27d7bc9b8 : SUCCESS in 1h 59m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce6024a9db444784872aefa8298e474f : SUCCESS in 2h 00m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2917d74d92d548be94be9a99cd3954e6 : SUCCESS in 2h 05m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c618f1bdace94ef4bb52ee2ce2b66aba : FAILURE in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b58e1bfd13736539281b2e05338c7f58f4f1290f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-29 21:02:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abbdc355be9014b0403ccde2bdc010632e9ec673","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-29 21:05:23.000000000","message":"Patch Set 1:\n\nrecheck\nopendev.org net connect issues","accounts_in_message":[],"_revision_number":1},{"id":"58171991fce07a460fbc46c860f0afde8cfd1cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 23:17:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cffb7095a7741e2820e4b694e17430b : SUCCESS in 12m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bfd748ab5ca468b920a9cea0692aaa9 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd7167a69e0b465f946df4b7e306e31e : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0827ded7f59c49608654509a76783f5b : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feec960f7c8d4da68f512c2f9a58e144 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9715a00788d64ebab884e153d2e67de8 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a31d46a8784cb8946ad3e97f7bf909 : SUCCESS in 17m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1aca5cfc71a84794a573d6e2d0863ee4 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a8750324d853406e8b3bef899dfbad78 : FAILURE in 10m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a0bc1ac236b74adebcf2345be1fddfa2 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/afa36d1005444e529a24967876f35dfe : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f42c206cd9843b6bbf56961f218d654 : SUCCESS in 1h 21m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdd99b7616254d6c95794059060a5d08 : SUCCESS in 2h 00m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/124194e381874d2883ea6011357779b1 : SUCCESS in 43m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/18d8928960b94bd69df5109288fd53b9 : SUCCESS in 1h 03m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6e44764e3804b229ba5098407242b6c : SUCCESS in 42m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b37b7251d9264e9e900e92e03bf16637 : SUCCESS in 1h 27m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5b7f9158959945ac9e54eeae1d2f754e : SUCCESS in 1h 47m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b4354a546084be1a396adfda33458fc : FAILURE in 48m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ce591335958744688f3f04835ac39768 : SUCCESS in 2h 03m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/483e2d0995bb40c39fe92566cb96a726 : SUCCESS in 1h 50m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/12a0260f59924c24a3936979eeba2b50 : SUCCESS in 1h 49m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/681ce67a36f14e4b9eb1957dd1713ab9 : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6475f8bf7f15a8f2dd244241b5ca98d91f9f51dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-07 19:43:11.000000000","message":"Patch Set 1:\n\nrecheck\nsocket timeout","accounts_in_message":[],"_revision_number":1},{"id":"a59389de33ca31b9fc67d54386d73fa521b72c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 21:46:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c01ab5c097440bdaa62a033af14e7df : SUCCESS in 11m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32e7f7f8e0d54d69845787ca18f32a5f : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b944f13147004b919d7d13335930edc0 : SUCCESS in 16m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e430c6e361d48a494aad5a49a06f207 : SUCCESS in 17m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cbf8fe876234092bb0e6e3729c9683d : SUCCESS in 13m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6471483bf90b4b78b07541a097242df6 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf39ba998074bf59caf1403bd69fc03 : SUCCESS in 14m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ca5ac565a9ca4783b9802b076765297c : SUCCESS in 6m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2673cf369e8041fcb43c9299fb8ba5c0 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ddd55e7c08d8431f978fd511afcc5ed2 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb6874ea5fec415fbecd22cb1d053783 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e09885d55fb48a794feab165bba3fc7 : SUCCESS in 1h 19m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d0f89777aab4b189d37f87fe5fea60b : SUCCESS in 1h 41m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90a402f6bd6042bc84b3f2795e5547b6 : SUCCESS in 45m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b5508093314a42da9dcd86cb82a9dd95 : SUCCESS in 38m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba7911cd9c4f48e2b395e14bb8e688b4 : SUCCESS in 46m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7749632c2f2c426a8ab396b53a75ae12 : SUCCESS in 1h 01m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/068fd5b6b0904f8aa88af14a5a41db63 : SUCCESS in 1h 39m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/037eb4d0f5024cc4a400dd315fd42c1a : FAILURE in 43m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6d9a3837f0424c47a97e8136d8f6304a : SUCCESS in 2h 00m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1ac80cdd34645eebc86eda51f42dcfe : SUCCESS in 1h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/07bec51b24964d2388ed1518e7c84570 : SUCCESS in 1h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9e3ac487bb814d428251118dbc026ea3 : FAILURE in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f36000f3e02b24dc7bd6c6ae99d97d0ed341325a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 21:47:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33c6ee9aa7e6bd70cf065ab551b7fab82fcff114","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:38:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c1a789c3f5942eca51948016a16544c : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de0a6da0dfa499f811724b543e2561f : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee943640b17b497e98c39ef7e2f43ef0 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0103333c533d4770801a1f5f61c7ff29 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0ef4b27a2c487395fa63952aa649ba : SUCCESS in 13m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/536019eae7b94d9d9f5b9dbfa939a04e : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4883fefa8cef46e1ac3bc36245231c4b : SUCCESS in 9m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/386ef41ba1e7428c85eaa09178e03076 : SUCCESS in 1h 36m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6200f911fce14d9bacf9d09ca7c237c4 : FAILURE in 1h 45m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d61f330e62664e4f80b0c85176366ea6 : SUCCESS in 43m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/47e5fc0d25ea4988b2508f6a8800e109 : SUCCESS in 42m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/338401ec957b4870adac4614760247a2 : SUCCESS in 44m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3fb81dcc8564d15a5e03863112aa7de : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c02675551c55280c8ceccea4f8523c38f9b83163","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:28:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8fa11228bb4b28a096ad50a5624a9db916eeebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 23:09:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/040b4a8192f542858fc9095dbd0051bd : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7df29ad0f3f04e079ab8fce703097ca1 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07bbcc97d2394eda8a22f932189451d0 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c0a18fe91824d5e814460c550894ace : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95bded9919bd4c68a14fe7a3f182095e : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec76fe6a9c844cd6a455e9ab1432e416 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/534d0be85f58469c827f301d20f96e54 : SUCCESS in 13m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f815da471ae04bce967ca37948c8cff0 : SUCCESS in 6m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/078429f29e11474281b1d35c42490861 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/afc4b42e04144a9f9aceedaebaebf46d : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c7e18ab5bab416daedf6ca83a211d85 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc00fab7868a42ba9f55b2f7fbc44ec5 : SUCCESS in 1h 21m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ee750a0de434ef3860a6e77a2b265e6 : SUCCESS in 1h 44m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dfa3638589e1416c8066b21f5a0967eb : SUCCESS in 43m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/77331301701242aea546ede0b5ba378a : SUCCESS in 39m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68d99cab01e14b7693381a97e258c813 : SUCCESS in 39m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f3d6e9466b184ade84caf66b66521820 : SUCCESS in 56m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/275028b47d82443599a21905a77b5cf1 : SUCCESS in 1h 35m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ac409c14b974a12bb4757203a7a6f4a : FAILURE in 59m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5eb32d390f4d4e39a728a5439f148463 : SUCCESS in 1h 48m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/46ab0f6ef31d43fc85a254c71d97883a : SUCCESS in 1h 42m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f14a803316fd4f729a050cdcccb5ad3f : SUCCESS in 2h 09m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3ae37799e4af4679a3a2d677250623b4 : FAILURE in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ecd1d4341c885d633f64f54fc53a9be5fdddb6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 23:09:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"200d18a4f90a0a3b8d46f11cfbf93bd3a4860fdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-24 00:48:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfc0402fe8254437b248fc888740eb63 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d542c61eaf43d0829b086d3a386843 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42a6d68ed8ed4f0f9370adb5bb90b946 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97b49bb64e5340a48901d1bd2c2cf9b4 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be60e42a45842bb8ec1794ae0e5e42d : SUCCESS in 13m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e74a86b9c52a4fd3ba4226939ce10252 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ba59dba44984fbb8de1e26b1f39204f : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e5ae3b8e7b64a48b24828345bd2be72 : SUCCESS in 1h 15m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8506375bd7b54f45b574837368326c70 : FAILURE in 1h 37m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6d73750583cc4e6ca9420a62f2a23535 : SUCCESS in 34m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9a85f58e93ca4b629bdd9e6aeba73055 : SUCCESS in 43m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7496ec260204be1b842e49c4acf224c : SUCCESS in 36m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a836fc897e2a42d9b2d75927ee904653 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f58fee106858e33555225debd7dcf56b969b9a7f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-24 09:19:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb4c2294f5613f0babf9d4aa706ceef40b5d7286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:31:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b187c318544348b08e3db03d6228deea : SUCCESS in 11m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0880023f4bf842a59d8b354f2c7b2abe : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c53345d1d944a2b8862f396a32f475 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1980f095f0084cff86a706a88d3fc720 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a0295f3fa5e4bf7b2f2d3de453d0046 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6be51781458b49319cdb78485b7033bf : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1979f60cca8475b90298ba9eb5f87ad : SUCCESS in 13m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/116557b163a5468a862b32f192024edf : SUCCESS in 7m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/145c0e21475d4d12864afb5b8d9b3d71 : SUCCESS in 10m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e171396954d84c579426f781adbcece5 : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ce1b9403d5f49c794a2875c11e8f8ac : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4265f0e9cec4e5f9117b8e88f706978 : SUCCESS in 2h 09m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b533b1d1534e4f3c88c0723749151150 : SUCCESS in 1h 24m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9664a03904224e49b621018bed0d7b79 : SUCCESS in 37m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/73f90b27842e4757bf6de3d2eb644838 : SUCCESS in 49m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a7212942b004a43b1c3aa432dfc4b6a : SUCCESS in 32m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d7fa0f3e27ed4c6d96a7fbff3ec11642 : POST_FAILURE in 1h 27m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a5f831e85b5443d8d38014e71a7c789 : SUCCESS in 2h 08m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1fe01bb6c71748eda22bf6ba548b3074 : SUCCESS in 1h 28m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bb27de4986b04fab86a6e57c6cb62b39 : SUCCESS in 1h 36m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b5d96a67279d4ed5b6abad9bb4f10e76 : FAILURE in 1h 26m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2a841d4765b04684b8aa2127bb60abc7 : FAILURE in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cb2430679aefd06847507d8841d69bf694cd2c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:32:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12bf109d32a4f1733deaf548222f0ed372bcb113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 13:39:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d39ba0a38e8f417fa9171adca5c50a3f : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eddd484eba6844a4ab7286684b2f7efa : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5dbba876984bc787bc8ca62b3f9391 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8738b07fab0b4e1993d29b125ed438cf : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1bb0426d11448f49f2804f5e257badf : SUCCESS in 13m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b7263e7fc8ed462895197c9a38a44bb3 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/981ed415d397433691ddd539224f5606 : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/979ddb7f1db44f5580c2d290ba145e01 : SUCCESS in 2h 05m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69f7b53549bc4148ba9cb7bf029efe10 : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6f20109bd80420e91cfe5803183df09 : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1be9f07a8f184dfcb4016835f7871524 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7d26f2c3835a059c5e0926eb1fb175bfd5999981","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 13:39:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5280be2acf07eaccfc325fc971aa7d7b786a9776","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 13:45:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b892c0aca4f4d6f9db1ee88de4044f4 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98efe97648330e0a19e7d5519b989093430623e0","revisions":{"98efe97648330e0a19e7d5519b989093430623e0":{"kind":"REWORK","_number":1,"created":"2021-03-26 19:14:22.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/65/783465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/783465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/783465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/783465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/783465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/783465/1"}}},"commit":{"parents":[{"commit":"380f5c4cceee5706d500140f35d54b09e78b95d2","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/380f5c4cceee5706d500140f35d54b09e78b95d2"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 19:14:21.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 19:14:21.000000000","tz":0},"subject":"Add Python3 xena unit tests","message":"Add Python3 xena unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for xena.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I9e37db9f45fe1b8880cc1034aee32a551fa339c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98efe97648330e0a19e7d5519b989093430623e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98efe97648330e0a19e7d5519b989093430623e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
