)]}'
{"id":"openstack%2Foctavia~783576","triplet_id":"openstack%2Foctavia~master~Iac949431f0901aaa0a60e5c39c2aab44cb9c1970","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Iac949431f0901aaa0a60e5c39c2aab44cb9c1970","subject":"Fix pool ALPN compatibility with older amphora","status":"MERGED","created":"2021-03-28 16:57:37.000000000","updated":"2021-03-31 02:47:04.000000000","submitted":"2021-03-31 02:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"783576","meta_rev_id":"56a8df4610f3f1d4e25d9feb9b1062c3fc1a02f4","_number":783576,"virtual_id_number":783576,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 02:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-30 00:31:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-30 16:49:25.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-30 17:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-30 17:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-30 17:45:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-30 16:49:25.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-31 02:47:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-28 17:30:53.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-03-28 19:46:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 13:43:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-03-29 17:11:57.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-30 00:23:21.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-03-30 17:12:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"68c8281bdd096bc958a23e3951fc211b1191d024","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-28 16:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f6364e60aeadff98b1d47d893960853fc868994","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-28 17:00:21.000000000","message":"Patch Set 1: Workflow-1\n\nChecking if my local pep8/pycodestyle version issue is upstream too.\nStill need to finish test updates.","accounts_in_message":[],"_revision_number":1},{"id":"1edb2e89db68c274d022ec32f4750dbd218a8ada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-28 17:27:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9660d9769305d2e00c74b8eee77247e599d67932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 19:46:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fad69a95b8d94d43932f135587c1b884 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/942adcb696dc40828b99b39d7a76ed99 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025d63777a4343c788e95df8c8cefbdd : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54bf130b15aa47508d18e7631cbbf17d : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e20c419902304bf89b71fef56bc3ac0b : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a06858f1903f4cd5be992ae7a91412b0 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52b369c228a444eaf831eac0f3b1f7a : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e454673f4b946548edc98bf0e6af5a1 : SUCCESS in 4m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e683a8e257bd403b9e1409cc6c43543c : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/89c397ba149a42e0b76666cd7afefd2d : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3522e69d26444889b915bc17daf5fb31 : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81b2d0ff7dc54a5789ca241e36d3a3c3 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6fda6eaad5454da392494d0d219a28c5 : SUCCESS in 1h 23m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13d7509feb8944398c3f9f44f5dc10b8 : SUCCESS in 1h 50m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0345fc8e49654caa8167d054719fa57b : SUCCESS in 39m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be92204c20c64a34a549dce5bc822c05 : SUCCESS in 38m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e20167da8eaa432196d9898432083d5a : SUCCESS in 35m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/661a7956c85e4d1ca5b732dcfb35be6b : SUCCESS in 47m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4d7f0e28953848fc87f34e7c8f4591fb : SUCCESS in 1h 43m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e03ef7a0e75d46a8980ec5eccdf1c0b9 : FAILURE in 41m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6928f57955e049a1951eb354d8cb0132 : SUCCESS in 1h 30m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b05d24ac961f473a97e0e8b40be6e6a4 : SUCCESS in 2h 13m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/88de104184194820a22acdd552161520 : SUCCESS in 1h 56m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/77d2665ca4d14db6906d6bf6ca811db1 : FAILURE in 24m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f9ac1eb870900075eea49b67d3a535a6c5d3a40","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 08:18:28.000000000","message":"Patch Set 2:\n\n(2 comments)\n\n2 questions ^","accounts_in_message":[],"_revision_number":2},{"id":"7743a9ae733f86010d8110df42fc7f5bef787eea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-29 13:43:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bfe8067b43f8f9040d36aa9bb469bda1b0afe65","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-29 15:13:12.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"68612ab45aadd938ca2cae1ba2ab82b622afa11a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-29 15:29:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"267d75a950d103f5c5adaeda4f876c9ea86dc6e1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 15:30:36.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\nThanks for the updated release note","accounts_in_message":[],"_revision_number":3},{"id":"c5f79ab516a8d9887cd096acb7e71b01f697aecd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-29 17:11:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47211a6216d2308dc06c5717090a84355c2f4a0b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-29 17:20:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a08088efa6a639f31790a1821bfba0cde1d6c961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 18:06:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b96a4b96ef74ca18f99e938491a2785 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c90c553ebcf442b89bf5a6143fd79337 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ce595aca4442758d89a17ab835478c : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba3422bb241e4f3c9a874abbb85f2801 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/064f915bc1b64ff99b03a587eb9a3207 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51dfd3371a97496e88c6f61975074081 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88b0f4a7fbe4b36ba7c9d6899be6bc9 : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11edba13e76f42d5a636e99d87d4bb6b : SUCCESS in 6m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/33b169cd4d7047078b727af03214c116 : SUCCESS in 8m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d83a9da6468d4ea894ab982b2ca66bc6 : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/961bf1cfa9164e199a0e2309df73b18d : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f9aaf1e33df431e80ee063fa9eceed0 : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6c3842f01a54ab9acb582d2e217a79f : SUCCESS in 1h 04m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0d8513b0ec240b9a3e2869517635f61 : SUCCESS in 2h 08m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/88a7d4100dc64934b2ec818dfd460420 : SUCCESS in 40m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/957f846b225d4cc18f8dc0daa48d1234 : SUCCESS in 37m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d36995b9cdaa4ca0a7a3df0da6cc6dcb : SUCCESS in 1h 15m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb836d1b463f4ec89d24297e95e644b7 : SUCCESS in 56m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10a4f14c0e324848b7cd39ba3609e670 : SUCCESS in 1h 34m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0177c05c49c54cdb9168f010d334e725 : FAILURE in 57m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7198e81559774aea928a1cc9b43b83d7 : SUCCESS in 1h 40m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a48e52f43d4c425ebdfa907fd1edcbdb : SUCCESS in 1h 41m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fecbdb56fefe44deaa8a44cc4022d073 : SUCCESS in 2h 00m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cc0b62b1517c4e8aa728a79fbfdeb19c : FAILURE in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"32e3a14c1e29511da0708997108ddfb901663c6f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-29 18:42:43.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI think Carlos raises a good question.","accounts_in_message":[],"_revision_number":3},{"id":"5bc95bf7a365a500174179aa0482df07294c5d70","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 19:01:17.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16c6f3532b8b864654295694484d4bc271ee268f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-30 00:14:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ac14017c54f7b5df6765cea6da837d65371a9c8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-30 00:23:21.000000000","message":"Patch Set 3: Code-Review+2\n\nI agree with Carlos in principle, but in practice I think this patch is the way to go.\nWhile the purist in me would prefer to never silently ignore a config option (I believe I\u0027ve gone on hour long mouth-frothing tirades about why I hate rsyslog and it\u0027s essentially because of this, so feeling a little hypocritical right now) I think that *breaking* loadbalancers is much worse, especially if this will just *happen* to all LBs by default, as Michael indicated. Also, assuming there\u0027s no newer image available (which is possible), this would break equally cryptically, requiring an operator (the user is helpless in this case as well) to track down why suddenly an entire fleet of loadbalancers just went to ERROR status and couldn\u0027t failover.\n\nI am in favor of merging this change as-is.","accounts_in_message":[],"_revision_number":3},{"id":"e88560c70ee0896a35640d4fab18cee6d46bd3cb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-30 00:28:23.000000000","message":"Patch Set 3:\n\nAt least with this implementation users will report \"ALPN h2 doesn\u0027t work\" instead of the random failures they will get with the current implementation \"Disabling a member caused my load balancer to fail and go offline\".","accounts_in_message":[],"_revision_number":3},{"id":"b5cedaae4728da78952885fa1c1506a95a8960e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-30 00:30:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1cc6e4d8730fac19af162a448c19957ccafb9a86","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-30 00:31:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ab6b02a5754855cd0fe66c69906d8195d671e642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 02:44:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1a66d995eed44a5b3eb30e972f17b2c : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06eba702e41d450190561ad7f0986e8b : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b380e9f7d99c4e81a218498a858bea03 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/080a406da57446fa9f20e87bc5239ca1 : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d95e786a50d4acb9cf6dff4b5515fde : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aad95b85783e47e39b927097761ad41a : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b065839cfbed4e2c8cfe1349d7306106 : SUCCESS in 15m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc8fa4a478a449ffa0072c9c89b93b2e : SUCCESS in 6m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/66ac5669ce1243bc9c17d79955d31076 : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c9cbc21ea884176b1ad3b1e9dbf6396 : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3872ec399984fbb904b1a97833d8fae : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dd50181e8954b48886a60719217fb66 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6605d2af54a461dbeace24d1996215d : SUCCESS in 1h 16m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcf5c72d0bcc4389a82ca45d558749bf : SUCCESS in 1h 46m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8395e39ba7644195b6015b492c539034 : SUCCESS in 52m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ebad5c55d8a843e9abd8067e07ea057a : SUCCESS in 36m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84a4239cf9ab4200abd4472034adecd0 : SUCCESS in 43m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5672f9ec2d54014bc4c1b38a2503f1e : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2bd6de4ea4b3492ab81fd66f1297aef9 : SUCCESS in 1h 55m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ea163e7406447179522f05e9924e034 : FAILURE in 48m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4d2c791701294a8ba635ed7a83690ed6 : SUCCESS in 2h 11m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f0a6b4108ab49ad84c19ef3d3a472df : FAILURE in 1h 39m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4f53f41e4c884f50a6cfe62d1c89719b : FAILURE in 1h 56m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5fa18fcb13394a46a0a11bf39a07db66 : FAILURE in 34m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b0b9945fb109675b06a1e96bace244661842456d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-30 13:42:09.000000000","message":"Patch Set 4: Code-Review-1\n\nThanks all for the comments.\n\nI ran the following manual upgrade test:\n\n1. Deployed stable/ussuri with a Bionic amphora image, created a listener, TLS-enabled pool and a member. Confirmed all operating statuses were ONLINE and provisioning statuses ACTIVE. No ALPN settings applied, only available from Wallaby.\n2. Upgraded Octavia to master. Left amphora image untouched.\n3. Confirmed alpn_protocols is unset\n    $ openstack loadbalancer pool show pool1 -f value -c alpn_protocols                                                                                                                                          \n    None\n4. Created second member in existing load balancer.\n5. Confirmed no errors were raised in worker, health manager and amphora. Also confirmed no ALPN configuration was set in haproxy.cfg.\n6. Deleted second member\n7. Confirmed same as in point 5.\n\nI believe this addresses the expressed concerns that existing load balancers (running old HAProxy versions) would be impacted upon adding/deleting members to existing pools.\n\nStill, there is one more item to consider: what happens when an additional TLS-enabled pool and member are created on the same old load balancer? Here\u0027s my test and observations:\n\n1. Created TLS-enabled pool, without passing in any ALPN protocols\n2. Confirmed alpn_protocols is set to the default value (octavia.conf)\n    $ openstack loadbalancer pool show pool2 -f value -c alpn_protocols\n    [\u0027h2\u0027, \u0027http/1.1\u0027, \u0027http/1.0\u0027]\n3. Confirmed no errors were raised in worker, health manager and amphora.\n4. Created member. The HAProxy backend server was set with ALPN properties and failed with same error message as attached to the story.\n5. New member provisioning_status is in ERROR. All other load balancer objects are ONLINE/ACTIVE. HAProxy reverted to a good haproxy.cfg (i.e. new member excluded).\n\nI would think the question now is: what should we do in this case, if any at all? Load balancer owners cannot create members on pools created *after* the upgrade. An update amphora image would solve this problem with either a manual failover or making an additional create/update/delete operation to the load balancer or its children which would trigger a failover via the health manager. *BUT* if the cloud operator does not upload an updated amphora image / stick with the old one, load balancers will go on an endless failover loop.\n\nSo, I would like to ask for feedback about this edge scenario: octavia upgraded except that amphora image was not which I\u0027d think is unlikely to happen and strongly recommended to update (the upgrade guide is clear about this step --https://docs.openstack.org/octavia/latest/admin/guides/upgrade.html): should we a )move forward with this patch potentially not honoring and hiding the desired ALPN configuration or b) assume cloud operators upload a new amphora image at upgrade phase?\n\nMarking CR-1 to signal community feedback\u0027s been solicited.","accounts_in_message":[],"_revision_number":4},{"id":"3c297e0914451c32f5f794161c0f04570450ac3f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-30 13:51:43.000000000","message":"Patch Set 4:\n\nIn my previous comment, where it reads:\n\"Load balancer owners cannot create members on pools created *after* the upgrade.\"\n\nIt should be:\n\"Load balancer owners cannot create members on *NEW* pools created *after* the upgrade.\"","accounts_in_message":[],"_revision_number":4},{"id":"3fde9672cbf743023e155adad05ae7191813ccc3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-30 14:29:15.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e In my previous comment, where it reads:\n\u003e \"Load balancer owners cannot create members on pools created *after* the upgrade.\"\n\u003e \n\u003e It should be:\n\u003e \"Load balancer owners cannot create members on *NEW* pools created *after* the upgrade.\"\n\nWell, I think it is they can\u0027t do anything with the load balancer after creating a new pool because any update will push the broken config.","accounts_in_message":[],"_revision_number":4},{"id":"d65cd94389bf609e2d70cd9bc8639dac2b314dff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-30 16:49:25.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2\n\nMichael and I further discussed this offline, including alternative solution paths. I am convinced now this patch provides the best balanced solution. I will let Greg, who also reviewed this patch earlier, have another look and approve it.\nMichael, thank you for reporting the bug, proposing a fix and taking the time to chat a bit more.","accounts_in_message":[],"_revision_number":4},{"id":"ccc315b2348866d6e45dbd98e4ac4c259da9392b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-30 17:12:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"57a749b1bb3a4d368fa0517991b600646a363ab2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-30 17:45:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a569b7d7d96507e6452046c672eae15e4f98666c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 17:46:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"020a15384b1487a111e3e15bd10f7db4c15d8e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:21:26.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"091984c72082c8305d3991d7fe480d51f3a5c9c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:44:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcf30ec3540847569dd610a5a9d5f78b : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e05435e891904cffa78f979c0180f6fb : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8525044344849cfbcf265ed656d9c90 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a25afb8f2ea48a480474eac7cc57479 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f98724b667749e2b9acb15e6d087091 : SUCCESS in 13m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c9c350c4d274ee893e5810fa8d30bf7 : SUCCESS in 7m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc6c3c5ad0a9489d837e503129c39507 : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50733394ecd84c859aea0261b6a414ba : SUCCESS in 10m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/735bbe63cc3b4f22b231bd450bfa6361 : SUCCESS in 1h 25m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/855e0bfa9db24578bf427abdb0e0b5c5 : SUCCESS in 1h 51m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63c7b351f58746cc94e44e473ea13e5b : SUCCESS in 49m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d1fb08523fee41f498825ae66a4f5772 : SUCCESS in 1h 09m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21507096bee84960a4c303ee06368f96 : SUCCESS in 59m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/743a7e0216f14e11aaa5d810957fa668 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":4},{"id":"d411f6b3919e3a788d1eaefa2d6f81a6f8b722e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:44:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"952c6b54be8ac8341b55d14fe73f5dfbfa810377","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:47:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff38877f55e46c8b286fe76227c28b9 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b40fbfcf859d4ef78250aefa413b9dc7 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a8e3f4021e658134732d3e4819289df1f794a6cd","revisions":{"f4bcdd48b3b402b8f3c5f16cac422a2505c3bc66":{"kind":"REWORK","_number":1,"created":"2021-03-28 16:57:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/783576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/783576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/783576/1"}}},"commit":{"parents":[{"commit":"3faf24d8f269849b86f1e9d863f81e72cbd53965","subject":"Merge \"Validate user access to vip_subnet_id when creating a LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faf24d8f269849b86f1e9d863f81e72cbd53965"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-28 16:54:17.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-28 16:57:35.000000000","tz":0},"subject":"Fix pool ALPN compatibility with older amphora","message":"Fix pool ALPN compatibility with older amphora\n\nThe ALPN for pools patch introduced a bug that causes amphroa to\nfail if the HAProxy version running inside the amphroa does not\nsupport ALPN on backend members (bionic iamges for example).\nThis patch adds compatibility support for those older HAProxy versions\nby removing the ALPN configuration settings if the HAProxy version\nis too old to support the functionality.\n\nStory: 2008780\nTask: 42171\nChange-Id: Iac949431f0901aaa0a60e5c39c2aab44cb9c1970\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4bcdd48b3b402b8f3c5f16cac422a2505c3bc66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4bcdd48b3b402b8f3c5f16cac422a2505c3bc66"}]},"branch":"refs/heads/master"},"70ea8b2dc3236327393b8dcfa8f90d265642eb84":{"kind":"REWORK","_number":2,"created":"2021-03-28 17:27:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/783576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/783576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/783576/2"}}},"commit":{"parents":[{"commit":"3faf24d8f269849b86f1e9d863f81e72cbd53965","subject":"Merge \"Validate user access to vip_subnet_id when creating a LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faf24d8f269849b86f1e9d863f81e72cbd53965"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-28 16:54:17.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-28 17:26:56.000000000","tz":0},"subject":"Fix pool ALPN compatibility with older amphora","message":"Fix pool ALPN compatibility with older amphora\n\nThe ALPN for pools patch introduced a bug that causes amphroa to\nfail if the HAProxy version running inside the amphroa does not\nsupport ALPN on backend members (bionic iamges for example).\nThis patch adds compatibility support for those older HAProxy versions\nby removing the ALPN configuration settings if the HAProxy version\nis too old to support the functionality.\n\nStory: 2008780\nTask: 42171\nChange-Id: Iac949431f0901aaa0a60e5c39c2aab44cb9c1970\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70ea8b2dc3236327393b8dcfa8f90d265642eb84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70ea8b2dc3236327393b8dcfa8f90d265642eb84"}]},"branch":"refs/heads/master"},"038e83d562f79765ce99ad59b0152674860484e4":{"kind":"REWORK","_number":3,"created":"2021-03-29 15:29:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/783576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/783576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/783576/3"}}},"commit":{"parents":[{"commit":"3faf24d8f269849b86f1e9d863f81e72cbd53965","subject":"Merge \"Validate user access to vip_subnet_id when creating a LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faf24d8f269849b86f1e9d863f81e72cbd53965"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-28 16:54:17.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-29 15:28:54.000000000","tz":0},"subject":"Fix pool ALPN compatibility with older amphora","message":"Fix pool ALPN compatibility with older amphora\n\nThe ALPN for pools patch introduced a bug that causes amphora to\nfail if the HAProxy version running inside the amphroa does not\nsupport ALPN on backend members (bionic iamges for example).\nThis patch adds compatibility support for those older HAProxy versions\nby removing the ALPN configuration settings if the HAProxy version\nis too old to support the functionality.\n\nStory: 2008780\nTask: 42171\nChange-Id: Iac949431f0901aaa0a60e5c39c2aab44cb9c1970\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/038e83d562f79765ce99ad59b0152674860484e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/038e83d562f79765ce99ad59b0152674860484e4"}]},"branch":"refs/heads/master"},"a8e3f4021e658134732d3e4819289df1f794a6cd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-30 00:30:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/76/783576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/783576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/783576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/783576/4"}}},"commit":{"parents":[{"commit":"3faf24d8f269849b86f1e9d863f81e72cbd53965","subject":"Merge \"Validate user access to vip_subnet_id when creating a LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faf24d8f269849b86f1e9d863f81e72cbd53965"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-28 16:54:17.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-30 00:30:25.000000000","tz":0},"subject":"Fix pool ALPN compatibility with older amphora","message":"Fix pool ALPN compatibility with older amphora\n\nThe ALPN for pools patch introduced a bug that causes amphora to\nfail if the HAProxy version running inside the amphroa does not\nsupport ALPN on backend members (bionic images for example).\nThis patch adds compatibility support for those older HAProxy versions\nby removing the ALPN configuration settings if the HAProxy version\nis too old to support the functionality.\n\nStory: 2008780\nTask: 42171\nChange-Id: Iac949431f0901aaa0a60e5c39c2aab44cb9c1970\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8e3f4021e658134732d3e4819289df1f794a6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8e3f4021e658134732d3e4819289df1f794a6cd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
