)]}'
{"id":"openstack%2Foctavia~783608","triplet_id":"openstack%2Foctavia~master~I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7","subject":"Fix rsyslog configuration when disabling logs","status":"MERGED","created":"2021-03-29 07:26:13.000000000","updated":"2021-04-23 22:49:56.000000000","submitted":"2021-04-23 22:48:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783608","meta_rev_id":"6299278e3be4c62cada6e1b0c9e07e4c35f68466","_number":783608,"virtual_id_number":783608,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-23 22:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-04-08 09:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-31 19:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-04-23 21:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-04-23 21:03:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-04-23 21:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 09:42:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 13:42:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-08 09:22:47.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-04-23 21:03:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"fd798ceedb132adee4916917ae8d360b622a3cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 07:26:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e76fa651b88c260d19b98bcafb58126c5ed57298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 09:42:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5188e16b0a447f18b072b69834fb89e : SUCCESS in 12m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e4155229d53458eae78eeecbd1212c4 : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac146d30b654e27af6906a489139e10 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42f1a523a9c849da89f56e317357150e : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f3b2b60b83744c8a5e92a68e062019a : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c30cde9670f4cd78e0084478fa1d1b0 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa2c48d8a5447faa455e70b23ff69f5 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb5c00594da24a899eb7f318eb387617 : SUCCESS in 5m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f55e3b2554a04ad9898ba16af145befa : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/40af93fae02c4dbf89e6786d055c8541 : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b393f584e6ad4a4caeb1e4f91a1970ef : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0751c5fd56094f50879b1aefee9f6089 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfef3bffa6ab4f3781235980374b9647 : SUCCESS in 1h 18m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8607fff7f9743c18683f17de3496afd : SUCCESS in 1h 56m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f0837dbcb214457866aade7b93fbd23 : SUCCESS in 42m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3c6a63b6e46b41cbba092ce261b9595b : SUCCESS in 36m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13cbf4b639ea43e4835ec1fd2b31bf50 : SUCCESS in 37m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3e6f389f6df4b308a1fffd9c673f251 : SUCCESS in 54m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/676fa924633840838c464895dc357e36 : SUCCESS in 1h 34m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05813291efb3406a9f275c1d8d1e3bf8 : FAILURE in 44m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2add460b51e34253af3a2b9cadc9bf91 : SUCCESS in 2h 11m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0940f9434b7d4d309d32ad571bdfa2f1 : SUCCESS in 2h 04m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/665a6785ff4043eeb75aa79892480299 : FAILURE in 1h 54m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2978133478844c40b6478c4002792c32 : FAILURE in 39m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d1355e578e8c8075fc87fe881df5dc3ace2c951","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 09:55:56.000000000","message":"Patch Set 1: Workflow-1\n\nunit test seems to fail randomly","accounts_in_message":[],"_revision_number":1},{"id":"24a7a142168a3ae031c718207ec607fd8efb2b84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 12:22:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6283e187f0f3e042804884718d3bb73aaa2909a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 15:15:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cfa6222142e4140a8120ab383ca4d1a : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52e1f144c39e43cbb15cebb4c58ee76a : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c02fa36ca2140c9ba1a65aab21ddb8d : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eedf3b2017bc498c83797034250c3d6c : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d269006091c54821ba4a792324eb9b39 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45bf14f479294d8ea346388ea4479203 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd4f5cd8db54775b3c47015028f9cce : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9a86eaf0399479cb91a7b0f7cc715d2 : SUCCESS in 4m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/10df9726018b424d8819f06a2072ee05 : SUCCESS in 10m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49edaf9aa4f0486d9e8cf433116cced4 : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/991d80767254427893cd0498c7dd4c09 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca64b65a807b4504aa918f395775a095 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9672366d2fe841aa8cb1753c4440432a : SUCCESS in 1h 18m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a754e9b4468c4daaa0727f7a52f5112f : SUCCESS in 1h 55m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7bd1c2760f44474b8928f3ab1b153f3 : SUCCESS in 1h 00m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/614e6904342d410cbbf6a24ef7e91974 : SUCCESS in 1h 13m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05afb4dd746441c191f4057a75ea241c : FAILURE in 37m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95b6b4a9a76e4e50866e906101fbc8ee : SUCCESS in 1h 17m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/36f00153ec3747ec96af1c5cbb801493 : SUCCESS in 2h 00m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0072cf8cafb843ea8f4b6a6d0b1173c9 : FAILURE in 1h 29m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be06b600d26b4ee0ba4e7ec010dda45b : TIMED_OUT in 2h 34m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/360cd91850a148a2be820eb3b6a6e6bd : SUCCESS in 2h 06m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/420107b1473f40c29c1876c90a2ecc67 : SUCCESS in 2h 21m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cd7f8501e988436aaae657c70d7438ec : FAILURE in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc74afb71d17512cad183d47e52ace5084328e88","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 15:53:30.000000000","message":"Patch Set 2:\n\nrecheck\n\nweird error in DIB","accounts_in_message":[],"_revision_number":2},{"id":"d930ac44c9476088020e95a0b2b6d0618893d89b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:28:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e651e8784154ec6b526fb6a37a36f5c : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6796f3e272142bf89fc95a4f837f505 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6921c35aa424458da3883d292efa7ce8 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82c912c7d1084d55ad0754b997084312 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf46d8c90cce40c2b9e2e8ffe4da1d17 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be11be20e6f94c79ba0acdf471bdc573 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0abed4accb034a3dba34781ad03d69f1 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d62602531fe424c83ff8596650ac2c9 : SUCCESS in 5m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/93b2a27549504fdebfaddfb4c0a87b2d : SUCCESS in 6m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b6c1b99033644b3db58586fd77ce6a44 : SUCCESS in 9m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fa39c292e6714d20b80366e8e20cb967 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08d175f950404c7a801b94f49826f3cb : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19e4cf8a50174fd294d77f353cb51930 : SUCCESS in 1h 29m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97101cde7df44b48893a714f5aa34c0f : SUCCESS in 2h 06m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71a77ffb2bcd4c5cbbaa2a4f6fede152 : SUCCESS in 49m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/072b8c182ce54d30a6adfb10b1571ded : SUCCESS in 39m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01a59a16eba34da8a4fe315f76225e22 : SUCCESS in 38m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10d2a399dd464891bf8d2682b2ba9f8f : SUCCESS in 57m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f80a4bef20844df486926c91fe7c963c : SUCCESS in 1h 45m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b4bd54518f745058ef75a3437a83f64 : FAILURE in 46m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3a1dc0fe64434ae497fe7f718e4e3109 : SUCCESS in 1h 35m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/11c3f7ce621e434eaf32dcbe0c44602f : SUCCESS in 1h 41m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ecd98ee8ea464fe3a984acd883117ed0 : SUCCESS in 1h 43m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f47b6d9d5de449e59421eee2d43790c1 : FAILURE in 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79b9d9a074c3b762613af0f3e01fef4143a2400f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-31 19:00:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1ba8f0a3100ab708bdc5554f7fa50a8028e29abd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-08 09:22:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6336397dcd0906d105d5a4c695f78fa7eb0f3a5f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 21:03:47.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"c0e3849289edce4dbbae734cbd26be1749290a03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 21:04:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6751c31bc6173943704eab179480d5dbe5014ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 22:48:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28be0ca34f054f1698345d4a4d8da3e6 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb55aed1c49647a8997a5ed83ae53c0d : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecb3af18725a407098197c4462c36c87 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/710940bd485648809893293095809740 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba741ea7df5b48cea4311d87b7197300 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a6b9a1608eb4c88a29f15732807835d : SUCCESS in 8m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/277d4e5f09164bc6844076875e82f73b : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8560f512c9b040b9a11016258ac2e2c8 : SUCCESS in 9m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7fed7daecc04d1bb45c35d6c58fb5c6 : SUCCESS in 1h 25m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8f756fdf6f54d79992530ae959c3d8c : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da6c2c1cb0a34ca3847f7e35e297d3d1 : SUCCESS in 41m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/89df25a0e0d343e1867136ce86f7e0fe : SUCCESS in 35m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e043428d59254cb5a6abb8b93c0c9e40 : SUCCESS in 31m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/76f90c7886404b5cbd823abf7120da95 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":2},{"id":"2e20a9d5726daf5e85c4cb8807bed4b89775ceca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 22:48:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6299278e3be4c62cada6e1b0c9e07e4c35f68466","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 22:49:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579a57b0b92740da8c8a207bb6e70f7d : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10cf71c1e13a4af5a07efda898645517 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e26664de59eb872a85520c24b5701dae31cd75e9","revisions":{"66a2a16f3ded5e568352176d681970adefe1d43f":{"kind":"REWORK","_number":1,"created":"2021-03-29 07:26:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/08/783608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/783608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/783608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/783608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/783608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/783608/1"}}},"commit":{"parents":[{"commit":"3faf24d8f269849b86f1e9d863f81e72cbd53965","subject":"Merge \"Validate user access to vip_subnet_id when creating a LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faf24d8f269849b86f1e9d863f81e72cbd53965"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-27 09:14:43.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-29 07:22:56.000000000","tz":120},"subject":"Fix rsyslog configuration when disabling logs","message":"Fix rsyslog configuration when disabling logs\n\nFix an invalid rsyslog configuration when disabling both log offloading\nand local log storage.\n\nStory 2008782\nTask 42173\n\nChange-Id: I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66a2a16f3ded5e568352176d681970adefe1d43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66a2a16f3ded5e568352176d681970adefe1d43f"}]},"branch":"refs/heads/master"},"e26664de59eb872a85520c24b5701dae31cd75e9":{"kind":"REWORK","_number":2,"created":"2021-03-29 12:22:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/08/783608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/783608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/783608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/783608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/783608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/783608/2"}}},"commit":{"parents":[{"commit":"3faf24d8f269849b86f1e9d863f81e72cbd53965","subject":"Merge \"Validate user access to vip_subnet_id when creating a LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faf24d8f269849b86f1e9d863f81e72cbd53965"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-27 09:14:43.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-29 12:22:27.000000000","tz":120},"subject":"Fix rsyslog configuration when disabling logs","message":"Fix rsyslog configuration when disabling logs\n\nFix an invalid rsyslog configuration when disabling both log offloading\nand local log storage.\n\nStory 2008782\nTask 42173\n\nChange-Id: I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e26664de59eb872a85520c24b5701dae31cd75e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e26664de59eb872a85520c24b5701dae31cd75e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
