)]}'
{"id":"openstack%2Foctavia~783609","triplet_id":"openstack%2Foctavia~master~Iabc2f613d1afbfac414d9e98cbd147467bceda73","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Iabc2f613d1afbfac414d9e98cbd147467bceda73","subject":"Fix default amphora output filename in diskimage-create","status":"ABANDONED","created":"2021-03-29 07:32:58.000000000","updated":"2021-11-17 20:14:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fcb5b0b19a48e3dbfbb71159394336d478079042","_number":783609,"virtual_id_number":783609,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-29 09:29:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 09:29:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 13:44:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-29 22:18:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"e5cc2f4a3943962a665da0f8d561a15f17447b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 07:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5fae67ed7a67a887171d8da69081245e819d18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 09:29:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3391e125a5ed41568aada68255f0cda9 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b76892c17928433e80ca677ce19d7064 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0eea25a5b348d59c0e2577d52f5ae7 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ecaed0f61148a18965b489a84619c9 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3211be91c4b84162ab2bb3778a44b86c : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3b70d34f26046df9fb9a66db662d8b8 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e0d55b87c24750b5f4113d57fabfa7 : SUCCESS in 13m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e3ed1d5f483f41049528eca76d56ec9e : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c9e7987057b40c6b02f7dd7b7ce98fe : SUCCESS in 11m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/76709b7d40e14add90d03224249625b4 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8766eeb80d28492ca1dda65514de1c7f : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/852f4f5ed09545e3ac136e401ad10b95 : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cca771a1e2b246119cfa23cc81f9de58 : SUCCESS in 1h 24m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bc2f0f3d13e4575b0cb97e74f34ec33 : SUCCESS in 35m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b3e0ddd56da3483c91e7177efee2d889 : SUCCESS in 38m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab1d2936a3534307b94175c860fb32f4 : FAILURE in 53m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/02594fce0691423e8495d91c0e6f21e0 : SUCCESS in 48m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f0b6c7e342af4523a30ba55319b0e298 : FAILURE in 1h 54m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca410dfc05f1455ba1bc2755caf50419 : FAILURE in 54m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c25aaa555bfd4e778ed788783f14c8ad : SUCCESS in 1h 39m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/67bbd96a3a4b4c6fb197531e9e4dfb7d : FAILURE in 1h 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fb44322dde424795a5799d0c6cb60e5b : FAILURE in 1h 44m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/422dcd9dbe1c4ee48e01b66254ca7cad : FAILURE in 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2af97efaa2bd213ff9562f910c1b537f3bf41d67","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-29 19:02:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7cb73a8e37234caab3180adf798562264f14385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 21:18:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6123b31a849d47a49c714afc00bcba89 : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6abf0f08c0c74a69ab8da399f8e35153 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f596116b5f94c85a436e2367cdeb4e7 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d20edea9c2254db49b52d73517e4dde7 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8baff25307fb41c2bc6c192d760abe48 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af911dc7075149e8aeda08b34e517d5f : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7877986b506f4dd780f89d2735867e07 : SUCCESS in 12m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b3f5549ab6ec41d9aa1294bfa1696046 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1e1de3531d364cecae91cea5a9611300 : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2d97e405ee884caea122987564f73a03 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5fe79755ddb4062b48a218e52e62396 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18d9dce76161430eba9e98c47c218606 : SUCCESS in 1h 25m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e51743a4e3a74fe69fb859037738bd94 : FAILURE in 2h 04m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6f9020e44e04f1984e7caae2c60e37d : SUCCESS in 43m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/71a028cf088b41a69a04bf470a18b94a : SUCCESS in 40m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d315e75dafa4dc98cf524357518f843 : SUCCESS in 44m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/baf9c5c71e304327a07e08be70df2f1c : SUCCESS in 49m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cd791913f6124796bc6dbbe00e0d2614 : SUCCESS in 1h 44m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5dc3be6bfe14d1190d04c1103172f11 : FAILURE in 43m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1653722917ae4764a5b5aa55e09554bc : SUCCESS in 2h 13m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/834cf75baf77474092cea6b6e2bd099a : SUCCESS in 1h 52m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fbebcd6d923844709a7fc71026ee55ba : SUCCESS in 1h 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/327c347708864be38c01f80dfb8f0bf2 : FAILURE in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d12a24fbe0cf295207cf0c0a87d1b3189e61427","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-14 18:08:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1ac604a95355639d190edacb517f82d13755ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 20:20:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79fd8ecf923e4bd5a54846a87471d808 : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48e88cd9f8124c728b5c364643b9bc50 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e392c965258e4e438172957cb7959984 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acfb0b20505349f7a5439eae80014dd5 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbbcafa38df74908a1fcb7b7930cc1f7 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c05bf7b34504a99a6c7ac9f7254a2b4 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/183485837c61406cb4f2524f8e55ab54 : SUCCESS in 12m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fc1375970d16479ab83cab8d5e76ecfd : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c33cb87a5efd435e8a9edf8b3acc62d2 : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6bb93e036908470684395fc8b2c948b0 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e52877dfd6c04b4e9ca3e4254fcb2b0d : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c4a2bdf4da24fffbfc9846a855cb990 : SUCCESS in 1h 12m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce1aa077651e41a099e36dde65003b82 : SUCCESS in 1h 53m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/242d8f2dbf4e4b1bafbbc8cfa9158c0f : FAILURE in 36m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ca1aea397fc544048dce121422398286 : FAILURE in 31m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1db38b90d7e44ef19e41a719140dff3c : SUCCESS in 43m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08b46b8d95024e22975ecdad0900223b : SUCCESS in 1h 05m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8ef0561a29f4fc29e65cba59549d9fa : SUCCESS in 1h 46m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ae90c9245bf44d52bcfe8514e8f11440 : FAILURE in 52m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f2b84b9268db40ae9bbdb8d9e98a7df4 : SUCCESS in 1h 32m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e4d7488f8d6046a2a218ec0bf73c8644 : FAILURE in 35m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c5bc37c12e514572aa06532f9cc045af : FAILURE in 1h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e498bf6beae6494bb00649ac7b055495 : FAILURE in 30m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77d48218cf28237b5ec7f55575ae0b228baff8f6","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:18:44.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":1},{"id":"fcb5b0b19a48e3dbfbb71159394336d478079042","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-17 20:14:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1564f93d248a9ba0716aeeac768abaf0303ef908","revisions":{"1564f93d248a9ba0716aeeac768abaf0303ef908":{"kind":"REWORK","_number":1,"created":"2021-03-29 07:32:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/09/783609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/783609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/783609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/783609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/783609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/783609/1"}}},"commit":{"parents":[{"commit":"3faf24d8f269849b86f1e9d863f81e72cbd53965","subject":"Merge \"Validate user access to vip_subnet_id when creating a LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3faf24d8f269849b86f1e9d863f81e72cbd53965"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-29 07:10:27.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-29 07:29:13.000000000","tz":120},"subject":"Fix default amphora output filename in diskimage-create","message":"Fix default amphora output filename in diskimage-create\n\nThe default amphora image output filename (AMP_OUTPUTFILENAME) doesn\u0027t\ninclude the extension, as DIB may add it automatically.\nBut later in the diskimage-create.sh script, we run \u0027stat\u0027 on\nAMP_OUTPUTFILENAME, and it can only work with the real filename.\n\nChange-Id: Iabc2f613d1afbfac414d9e98cbd147467bceda73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1564f93d248a9ba0716aeeac768abaf0303ef908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1564f93d248a9ba0716aeeac768abaf0303ef908"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
