)]}'
{"id":"openstack%2Foctavia~783680","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ia380b712d9d40d702b99bb3b04e1f45dcb60b96c","project":"openstack/octavia","branch":"stable/wallaby","hashtags":[],"change_id":"Ia380b712d9d40d702b99bb3b04e1f45dcb60b96c","subject":"Make focal the Ubuntu default for diskimage-create","status":"MERGED","created":"2021-03-29 16:17:40.000000000","updated":"2021-03-30 00:19:57.000000000","submitted":"2021-03-30 00:18:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783680","cherry_pick_of_change":762925,"cherry_pick_of_patch_set":2,"meta_rev_id":"32f21aa1cd6dfb1f7bb6902cd0639b9586b443a6","_number":783680,"virtual_id_number":783680,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-30 00:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-30 00:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-29 21:06:25.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-03-30 00:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 16:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 21:06:25.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-30 00:00:27.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"96cdcde813b7b114d9e7d949abed8e283546db03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-29 16:17:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"87b6a9b7797600ce14aed05b1f3859683d512321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 16:49:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff79ed36f1534d189035077ae384a2ee : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26d521a188024aacb417ad8384710169 : FAILURE in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de2ff6dd5174d37bea93c3fdc4962ff : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7965979640f3468a89f8e4a55a991f26 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18dfc32567fd431682017c845121a208 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11615fd6ff904d609300ceededb4e7b3 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83cf45ff4c354d4093a3e850777fae07 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdcd13a28f2046b3ad81e8c7fc6034b3 : SUCCESS in 5m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cdf4594d34914de284a16d1101a164fa : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4765449c7a734ec5ac18472cbc8715b6 : SUCCESS in 13m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/422e67afff574e12b56d7306a0d5a4d8 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a48e5cd29474ef0bc78886ec7f12a35 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":1},{"id":"48971cb377562ff2916f3e16c6ef90222c75de15","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-29 19:06:30.000000000","message":"Patch Set 1:\n\nrecheck\nsocket error","accounts_in_message":[],"_revision_number":1},{"id":"153f730d002ac4c36bda2461f5c46a39c7e8eaf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:19:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ba688859e374f7c8b19fc3709fdba1e : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45efe4c2f32245799c37c1b767b43d03 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2e69ff66b24343abc13556c2953ddf : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82bc644528ad46f0a59c75c87461a6b7 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc01e1b883b48c88dfe8077c0416d70 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/418f750b505a44dd8c29c1234feb64e5 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2420d26b6645129a61326b5b4a6532 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91b7735a711948a6af013899cbce797e : SUCCESS in 11m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/97a9ecb040864c3085835addd91996ad : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7875c02642c1431a88d412dc082ce4c8 : SUCCESS in 12m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ba4acd2e13c3470fa6bebd3b18806232 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df0be4d03e064512a24b7086e219da36 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"691afd46392f794a14b37c2b3fff070f1f0d95f2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-29 21:06:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44765bc6983b080e37e637c33d8458a52fd05cd2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-29 21:07:25.000000000","message":"Patch Set 1:\n\nWe should make a Wallaby RC2 release to include this one before GA.","accounts_in_message":[],"_revision_number":1},{"id":"3da61ada9d510c172131aa82b48f8db9be9637ca","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-30 00:00:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"817c0bd8d54cffa662581e93a73bd2859915340f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 00:00:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9efec9b159ad833b30643702fc16fb812e069cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 00:18:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408894a3cc574a1d8a485ef4d1dd29d2 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94b50a25c9e4eae9810025e0ba97c78 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e09a2d521fd40118583752e6494aeee : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90d1062c778f4894ab58e6e7c729bf5b : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4189c9b3a7cd42b49b969fee2b4192b6 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84f886ccbb324046b84419ee737cbcec : SUCCESS in 4m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/331005002e834951be989c89fee29147 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/57323b7247154d7ea4fd6525d7c485f5 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9c3f3e80c4ec88373cfa5d3457a64dcff1ee1a0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 00:18:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"32f21aa1cd6dfb1f7bb6902cd0639b9586b443a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 00:19:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c3a69db418434ea14ce7a9d0118e55 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c9c49b4ce8840248e7500d3da08dd53 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40f70d6dc516298f14c99b0c7188c2195caa0273","revisions":{"40f70d6dc516298f14c99b0c7188c2195caa0273":{"kind":"REWORK","_number":1,"created":"2021-03-29 16:17:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/783680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/783680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/783680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/783680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/783680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/783680/1"}}},"commit":{"parents":[{"commit":"1b7c421f35359ee939d318cd337e348e61a1e8f4","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b7c421f35359ee939d318cd337e348e61a1e8f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-16 21:35:41.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-29 16:17:40.000000000","tz":0},"subject":"Make focal the Ubuntu default for diskimage-create","message":"Make focal the Ubuntu default for diskimage-create\n\nThe current PTI version for Ubuntu is focal. This patch updates\nthe diskimage-create.sh default to focal from bionic.\n\nChange-Id: Ia380b712d9d40d702b99bb3b04e1f45dcb60b96c\n(cherry picked from commit 40d011ab7f1f7141041440b760a8e68671979016)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40f70d6dc516298f14c99b0c7188c2195caa0273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40f70d6dc516298f14c99b0c7188c2195caa0273"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
