)]}'
{"id":"openstack%2Foctavia~784004","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I7b3693fe3a0b8c5e64936284c0e2f03b12cd0c48","project":"openstack/octavia","branch":"stable/victoria","topic":"docs-centos","hashtags":[],"change_id":"I7b3693fe3a0b8c5e64936284c0e2f03b12cd0c48","subject":"Add docs for centos8 installation","status":"ABANDONED","created":"2021-03-31 06:19:59.000000000","updated":"2021-03-31 07:54:16.000000000","total_comment_count":5,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"e958b68574db1e4a2b395ad9aa29782117d6ceff","_number":784004,"virtual_id_number":784004,"owner":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-31 06:53:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2021-03-31 07:24:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 06:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 07:24:43.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"118cc3ea0ffdf3ada06231df8254aecb26a1bd4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"date":"2021-03-31 06:19:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d08d71d11c075226ce693715bb3c9baf97676f5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 06:53:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb7eb475473425b9d5c84cb370f8098 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62467dc8c0341e18261f53f333e85d1 : FAILURE in 9m 23s\n\nWarning:\n  Comments left for invalid file doc/source/install/install-ubuntu.rst","accounts_in_message":[],"_revision_number":1},{"id":"268f3fd664088ecacaf0a21d1a2b732626f39f30","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 07:24:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nThanks for your contribution to the Octavia doc.\n\nYou sent this patch on the stable/victoria branch, but you should propose it on the master branch first, then you would back port it to the stable branches. Can you cherry-pick your commit on master, then propose a new review?","accounts_in_message":[],"_revision_number":1},{"id":"0ccf9ecdf330c38cc9edb887db162d7f6c919327","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-31 07:47:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e958b68574db1e4a2b395ad9aa29782117d6ceff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"date":"2021-03-31 07:54:16.000000000","message":"Abandoned\n\nReplaced by a new change towards \u0027master\u0027 branch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34dcf441e09af50322e48db2777e6fb98e6ce3c7","revisions":{"34dcf441e09af50322e48db2777e6fb98e6ce3c7":{"kind":"REWORK","_number":1,"created":"2021-03-31 06:19:59.000000000","uploader":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"ref":"refs/changes/04/784004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/784004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/784004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/784004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/784004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/784004/1"}}},"commit":{"parents":[{"commit":"0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec","subject":"Fix provider driver quota handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a8254e04e6c7536564e0497ef6ebe7ce2b3f1ec"}]}],"author":{"name":"Piotr Mossakowski","email":"moss2k03@yahoo.com","date":"2021-03-31 06:19:26.000000000","tz":120},"committer":{"name":"Piotr Mossakowski","email":"moss2k03@yahoo.com","date":"2021-03-31 06:19:26.000000000","tz":120},"subject":"Add docs for centos8 installation","message":"Add docs for centos8 installation\n\nAdd procedure assuming openvswitch mechanism driver\nAdd missing [health_manager] heartbeat_key in octavia.conf\n\nSigned-off-by: Piotr Mossakowski \u003cmoss2k03@yahoo.com\u003e\nChange-Id: I7b3693fe3a0b8c5e64936284c0e2f03b12cd0c48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34dcf441e09af50322e48db2777e6fb98e6ce3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34dcf441e09af50322e48db2777e6fb98e6ce3c7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
