)]}'
{"id":"openstack%2Foctavia~784022","triplet_id":"openstack%2Foctavia~master~Ib5f4aaf873c39220a81a875770b7a2bbd49c8ab1","project":"openstack/octavia","branch":"master","topic":"docs-centos","attention_set":{"33296":{"account":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"last_update":"2021-11-19 14:46:18.000000000","reason":"Gregory Thiemonge replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-01 14:39:00.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib5f4aaf873c39220a81a875770b7a2bbd49c8ab1","subject":"Add docs for centos8 installation","status":"NEW","created":"2021-03-31 08:01:41.000000000","updated":"2023-03-01 14:39:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"6eff2faa75cc155462ba55f73769c7d5c737f9e2","_number":784022,"virtual_id_number":784022,"owner":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-26 09:04:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2021-11-19 14:46:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-03-01 14:39:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 08:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 15:11:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-29 22:19:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"},{"updated":"2021-10-26 08:21:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2021-10-26 20:41:12.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-03-01 14:39:00.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"2fc426b12dc8ec924d302a4062c4b2fb9572514c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"date":"2021-03-31 08:01:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"829a1775ca6bc0f5766a38e9bc5d5e73491a0630","author":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"date":"2021-03-31 08:13:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56182bf364f080ece6aa34bca6adeec08a84e615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 08:17:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2b84ea5a34439195eaadca936d4879 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40f1554835a43698f9053bd91d4e58a : FAILURE in 8m 14s","accounts_in_message":[],"_revision_number":1},{"id":"492a1734005c4fa4789ebf942a9fea13dc26945c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:19:59.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":1},{"id":"663ee161583c6f3d8f3e02f824859a86076a1efd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 08:21:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e8eef88bb2956efd978a7ed5a6b53f5505b1003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 08:39:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b2bfd9c55e4219a5a5d91f6c9d305b : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88fb4705f03a4c9886173eb6e668285f : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"668432fa57102f10c0dad49e994f954fc2c5693c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 08:40:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffc96ecc1df423af58065abdb64c74d7dfb63582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 09:04:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1778149593b4474c9ba74e98cfee6572 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d65a731c380439f9fc697659e2e1758 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":2},{"id":"3e71349d2645eb8d8b28f5ddf7482cef228c5a4a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 20:41:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1409a91286245ad5806ddbaea4726dba7baf185a","author":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"date":"2021-11-05 11:14:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1108b6ac4025ccdeb356f6c474d6750b63538410","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-19 14:46:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6eff2faa75cc155462ba55f73769c7d5c737f9e2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-01 14:39:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c95c56185075a28c5f64d905f9fe618189f1683b","revisions":{"2f9225b135577b94e403a7f5bbcf7d8581ca44de":{"kind":"REWORK","_number":1,"created":"2021-03-31 08:01:41.000000000","uploader":{"_account_id":33296,"name":"moss2k16","email":"moss2k03@yahoo.com","username":"moss2k03"},"ref":"refs/changes/22/784022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/784022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/784022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/784022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/784022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/784022/1"}}},"commit":{"parents":[{"commit":"f769c2b42daec25e680308c225d5ede3632d6f71","subject":"Merge \"Fix pool ALPN compatibility with older amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f769c2b42daec25e680308c225d5ede3632d6f71"}]}],"author":{"name":"Piotr Mossakowski","email":"moss2k03@yahoo.com","date":"2021-03-31 08:00:59.000000000","tz":120},"committer":{"name":"Piotr Mossakowski","email":"moss2k03@yahoo.com","date":"2021-03-31 08:00:59.000000000","tz":120},"subject":"Add docs for centos8 installation","message":"Add docs for centos8 installation\n\nAdd procedure assuming openvswitch mechanism driver\nAdd missing [health_manager] heartbeat_key in octavia.conf\n\nSigned-off-by: Piotr Mossakowski \u003cmoss2k03@yahoo.com\u003e\nChange-Id: Ib5f4aaf873c39220a81a875770b7a2bbd49c8ab1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f9225b135577b94e403a7f5bbcf7d8581ca44de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f9225b135577b94e403a7f5bbcf7d8581ca44de"}]},"branch":"refs/heads/master"},"c95c56185075a28c5f64d905f9fe618189f1683b":{"kind":"REWORK","_number":2,"created":"2021-10-26 08:40:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/784022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/784022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/784022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/784022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/784022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/784022/2"}}},"commit":{"parents":[{"commit":"f769c2b42daec25e680308c225d5ede3632d6f71","subject":"Merge \"Fix pool ALPN compatibility with older amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f769c2b42daec25e680308c225d5ede3632d6f71"}]}],"author":{"name":"Piotr Mossakowski","email":"moss2k03@yahoo.com","date":"2021-03-31 08:00:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-26 08:40:21.000000000","tz":120},"subject":"Add docs for centos8 installation","message":"Add docs for centos8 installation\n\nAdd procedure assuming openvswitch mechanism driver\nAdd missing [health_manager] heartbeat_key in octavia.conf\n\nSigned-off-by: Piotr Mossakowski \u003cmoss2k03@yahoo.com\u003e\nChange-Id: Ib5f4aaf873c39220a81a875770b7a2bbd49c8ab1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c95c56185075a28c5f64d905f9fe618189f1683b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c95c56185075a28c5f64d905f9fe618189f1683b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
