)]}'
{"id":"openstack%2Foctavia~784382","triplet_id":"openstack%2Foctavia~master~I99ce408ec886820c056b69696b26be9521740f1c","project":"openstack/octavia","branch":"master","topic":"ca_cert-comments","hashtags":[],"change_id":"I99ce408ec886820c056b69696b26be9521740f1c","subject":"Fix comment for the ca_certificates_file opt","status":"MERGED","created":"2021-04-01 12:12:15.000000000","updated":"2021-05-05 12:41:58.000000000","submitted":"2021-05-05 12:40:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"784382-ca_cert-comments","meta_rev_id":"920f82a6535acead995c769d2f6eb2ce582fe44a","_number":784382,"virtual_id_number":784382,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 12:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-05-05 10:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-04-08 15:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-23 21:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-04-10 20:45:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-05-05 10:55:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 14:35:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 18:08:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-01 23:10:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-05 10:55:41.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"68fdd29a8f2e4da02899e361d3337f4f30e153fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-01 12:12:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"285aa55b6f4780785bd7581ee585f4a6549f65aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 14:35:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b873e470b9d4ed087a6670e232c8778 : SUCCESS in 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c96f12f04de4b95a786b110d87de600 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8958bebbbc8942109f591df9978cc96e : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c173f2ead15e47b0bf5bad57d1e0fc8e : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af2129a6443c4247aa6281cdb07a87e3 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d2459d143e14907a16067ae184b752c : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a18d98370494189803ade8c66d0d885 : SUCCESS in 12m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fedd6203235d42ba81874eee75bc2cea : SUCCESS in 6m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a1fadd46f02a460eb1863702659d48ed : SUCCESS in 8m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9aa8672a80634bb18edf7214344d0ca7 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bdab602a261a434d9aa27e6a7ba0d2a7 : SUCCESS in 13m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1cde502160347faa1cf7a133f3cc9ce : SUCCESS in 1h 13m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b9cd549c6f3481c9cc966a48813fb33 : SUCCESS in 2h 02m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c545bdf8e20d4f68a16bba4e85fd1bec : SUCCESS in 40m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aee57cf5f31e46f28b177629366ed4c0 : SUCCESS in 36m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/404d5fed86f348d490ab7f315442feb9 : SUCCESS in 36m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3170205feab8478ab1216769a5a33874 : SUCCESS in 1h 14m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9dd3223674394aea84932aa67acfb6a8 : SUCCESS in 2h 02m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6bea169ff9e24a83831fe9c2b6d275fe : FAILURE in 52m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fabd0a2ac614daa91437863794a2f68 : FAILURE in 1h 56m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d407e936b7214e46baffe57a267032ba : SUCCESS in 1h 52m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/845ee2e1a6024db7ac433e733d49039c : SUCCESS in 1h 28m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7a11f85555fb484c9c0fb66b1fdf6bfd : FAILURE in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7a01b85bd2e498c525c88a97f15bfdab43bcaae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-01 18:08:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c81a71dae9ad5fbec121d259e7fce9ab2bcde01a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-01 23:10:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI think we need to acknowledge that barbican is not the only supported option and to update the example etc/octavia.conf comment as well.","accounts_in_message":[],"_revision_number":1},{"id":"d486dcbd8dabb6c244664255260e64e68dedcf3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-02 16:40:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b04ddb71b5f7a1e3df6f356c9f0f1ee0891cea82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 19:01:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46867a98256242bba8e0c75ae7e43b28 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95044d82c3fe45308e1850fa7422f728 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049864e92cbb4561871e6cdb14ef6381 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfa7172528b8428489685e3441791cd1 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a85bba99aa48889476167744c1c2fd : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c154992cfafe474088e1a490765f36a1 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14846117a964c709c304b57cb21c7e3 : SUCCESS in 11m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/310f450c171045f1b055cd37919484ee : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/580d4105ca7d4eefafe0c9575b428621 : SUCCESS in 7m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c9899453294e49f590b3bb10285a3d9e : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06ee84dbce02412aa05ce2501f88ed3b : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20df074483644eea9e42cf7077e7ea27 : SUCCESS in 1h 16m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aab8a19ccacf4891a7a6078452a47c76 : SUCCESS in 1h 38m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b528b620dd914db18c258e3d3315bac0 : FAILURE in 51m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8cbe6e62e66848e9b9ebd8bbd3991725 : SUCCESS in 39m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebe84cc16a534b1a897a47fc997bf158 : SUCCESS in 32m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65dcb3ffce024cc6bbe08b45c8230f80 : SUCCESS in 1h 15m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2c9ca2949b6b443a98e7d195b3231c73 : SUCCESS in 2h 17m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29c01c0b3aea46e196a91d3a977ad186 : FAILURE in 1h 03m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c37adff51cc048b09618f0e5dc3e710e : SUCCESS in 1h 57m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b223afd2adac4087a4744acef1e311b9 : SUCCESS in 1h 45m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/91283a2cb4bf4d318c70dc5fd1d697bc : SUCCESS in 1h 41m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/60b7ee5131b742fd893eb38673d86a77 : FAILURE in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3f411838f3134fda3ef6c5da09b3dbf5a97c45d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-02 21:21:23.000000000","message":"Patch Set 2:\n\nrecheck\nnodepool issue","accounts_in_message":[],"_revision_number":2},{"id":"6b653336461c4737097da5adeab5cfacd8285212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 23:02:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d79c66d61395400b85c251174384de30 : SUCCESS in 10m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba436df848de4fe3a03899cba9a07c00 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11e97b43e7b4ffaadc4755e5ac2eeb0 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8913f8f00024d68aca3de3c82e54ff8 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77cf363beb7f4c27b495db3ed68cba92 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82e4a17f54614aecba091a0c3b2768c0 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee5d82eaf67948929b9010286ff7d871 : SUCCESS in 14m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/133d2e0eb3ac42a6a36ccd0eb452e711 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/08d1eacd35d542d8b127b41ea84a7667 : FAILURE in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23aead0565b945a78ba507fbd8daacf2 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f819dc5793b4420a83ddf6cc9effbfcf : SUCCESS in 6m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f44a02153aac47e78ff8026cf87dc661 : SUCCESS in 1h 04m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d21a1cce8ed4b57bb73db87f0bf5bb4 : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10c75e3dd36f475490c657fb71839aa6 : SUCCESS in 44m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a7fb9a5d78ef4cc3a8b4907a7110fc65 : SUCCESS in 35m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7f1af95b10948c1901fdf7c100cccca : SUCCESS in 43m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60422cf6e73949feb01864c1cd9e5614 : SUCCESS in 53m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c183e7f1daaf48aaa8fe62c0806591bf : SUCCESS in 1h 34m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2e7616663edf48ed97a771abb65c6920 : FAILURE in 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2d482d35e81a4be3aca996874dbbfb79 : SUCCESS in 1h 38m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f0fc69f3ad074cc7bbb4cbc4c9e8e556 : SUCCESS in 1h 28m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/afb46ed8a0844679ae86e68607b072ca : SUCCESS in 1h 23m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/40443c03162e4628b75f8e80eafed3ef : FAILURE in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58b7e65b2a16e4f8cdfd49a98b09a4f9197400ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-02 23:09:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"32e81f1b3c13efdfc3c178e2cd1a90b4e2e7c01f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 01:26:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55c56c15ef244130b0da2f2ccb19c306 : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dafaf52bf2ba4be89a4b951c4ab1fb29 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f6217ba35c47c4947ac29466b761a9 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee5f2b41a5964e649ae2339b387927ce : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c24d208723c146ae9178ca441122f7bf : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb5058090af480fb4e7cd6238afb17c : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2292052d87448ba91158d0bb99b228 : SUCCESS in 11m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d77a2e995f494c24acf3772222e85710 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dd00d3fd2faa4cca97b2b28820585a02 : SUCCESS in 7m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/003d0d372e0a411da138404edb0d7064 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6fbdd59bbbf4a6fb859a092a68344a4 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f00208344df44205a4cc74ddc765ef47 : SUCCESS in 1h 17m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df951be5777d4bc7bb9bbf25fa1e1311 : SUCCESS in 1h 56m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/322f1a58a30c4ef5a5e399a8d5b6b87d : SUCCESS in 43m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8afd9ea332bb4c179d2dbdce2038a9e5 : SUCCESS in 1h 00m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7aa9568f02e449e9b37de1b01064aa84 : SUCCESS in 1h 04m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/216e14b086b44267b9f95b3208386219 : SUCCESS in 1h 14m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3a5490bd34d48e6b3a2cefd93215a9e : SUCCESS in 2h 11m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa0f3e88adce46df86bfb7478941bc65 : FAILURE in 54m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/19b927ec9e9d43d492cdb688f11577f9 : SUCCESS in 2h 02m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f2712744bed44ce78e9df476545fe542 : SUCCESS in 1h 43m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2ff3c6fa24254df691461e68ff23c0e8 : SUCCESS in 1h 54m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/41568747b9194425b26c00bb1cc85998 : FAILURE in 31m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a61bb709336e2c5f935b9c4c7b7590101b9d4b6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-05 19:39:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1aaf7a17d91c1ac921d23412120618378263042d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-06 16:15:57.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d192d1948ffb69a95cbf62449ace6eec62844f75","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-08 15:27:16.000000000","message":"Patch Set 2: Code-Review+1\n\nI\u0027ll open a story about the discrepancies between the config and sample files, including updating text like this to be more generic.","accounts_in_message":[],"_revision_number":2},{"id":"3eb67331fb0c72f61cfe92302e94649609969e43","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-10 20:45:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c4d6d4335b426dd4339021cb188f6a6d166d62b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 21:02:28.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"aadb08afa01ec4da2dfafa36948967454424f8bf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-05 10:55:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e26d29e5137eefa7c29a54e39dfc95b5cea719fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 10:56:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e22947d4f111d87800648f0e2c7619c779e5aceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:40:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d904ad7ecfd04d6794df8c35f02f4db3 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7240f1d84fe24791a6312c2908d6a2ce : SUCCESS in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49c04202e36546bba338bfb2427f047b : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed372165fdab4e85a30fbc39612cfec1 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd565ac2e354445ca8541c3ac8a5ea4d : SUCCESS in 12m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef9b59e6a08d488286b092e4368344f9 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/211d84e4b50f4db7a6f5c4cb9b9e073d : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5dc36b3c2c9d42ec83b9158cb8eb9cd2 : SUCCESS in 1h 31m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee0c3c1ca5b64c91b4f51a975c4a3fda : SUCCESS in 1h 42m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1f72dddd9d54f3295eba9c086cb536f : SUCCESS in 36m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26e9fd52ff1f4c13ae0f6410e339c047 : SUCCESS in 36m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6653fcaf2264c38bd3e862bc2e14f4a : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c482e738f7d0ac46e20920af794f6d2f97e9a168","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:40:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"920f82a6535acead995c769d2f6eb2ce582fe44a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:41:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7256d10eef493b84903712f6329a54 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee0da827b163c15796bf2f19b0785bc0259d5128","revisions":{"8bef15b520235d839d22346a09a1a562049a7448":{"kind":"REWORK","_number":1,"created":"2021-04-01 12:12:15.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/82/784382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/784382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/784382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/784382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/784382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/784382/1"}}},"commit":{"parents":[{"commit":"fa08122b185e14ce28a8455a106bdb675ad8972a","subject":"Merge \"Fix devstack cleanup when using amphorav2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa08122b185e14ce28a8455a106bdb675ad8972a"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 12:11:07.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 12:11:07.000000000","tz":120},"subject":"Fix comment for the ca_certificates_file opt","message":"Fix comment for the ca_certificates_file opt\n\nThe directive [certificates]/ca_certificates_file currently has a\nconfusing comment. This tries to fix it and make it more easy for\nOctavia operators to configure the directive.\n\nChange-Id: I99ce408ec886820c056b69696b26be9521740f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8bef15b520235d839d22346a09a1a562049a7448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8bef15b520235d839d22346a09a1a562049a7448"}]},"branch":"refs/heads/master"},"ee0da827b163c15796bf2f19b0785bc0259d5128":{"kind":"REWORK","_number":2,"created":"2021-04-02 16:40:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/82/784382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/784382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/784382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/784382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/784382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/784382/2"}}},"commit":{"parents":[{"commit":"fa08122b185e14ce28a8455a106bdb675ad8972a","subject":"Merge \"Fix devstack cleanup when using amphorav2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa08122b185e14ce28a8455a106bdb675ad8972a"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 12:11:07.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-02 16:40:08.000000000","tz":0},"subject":"Fix comment for the ca_certificates_file opt","message":"Fix comment for the ca_certificates_file opt\n\nThe directive [certificates]/ca_certificates_file currently has a\nconfusing comment. This tries to fix it and make it more easy for\nOctavia operators to configure the directive.\n\nChange-Id: I99ce408ec886820c056b69696b26be9521740f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee0da827b163c15796bf2f19b0785bc0259d5128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee0da827b163c15796bf2f19b0785bc0259d5128"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
