)]}'
{"id":"openstack%2Foctavia~784514","triplet_id":"openstack%2Foctavia~master~I38a1dd7c4f10717a90d83a568cddf48d0dec2972","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I38a1dd7c4f10717a90d83a568cddf48d0dec2972","subject":"DNM: Testing the agent","status":"ABANDONED","created":"2021-04-01 23:01:04.000000000","updated":"2021-04-02 16:10:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92b2f41a482d4c5b61ff6452a2ead0e309416c37","_number":784514,"virtual_id_number":784514,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-02 01:15:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2021-04-01 23:01:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-04-01 23:01:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 01:15:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c7b88e143c24a93c04557b241410c296a713977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-01 23:01:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"697fecc0c9448675b2891b3ee6a441ac057b52ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-01 23:01:35.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1bf0cd21102f62443c107fb29dc147f0c901df34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 01:15:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/447e8484b43749f0940ea378baad3a64 : SUCCESS in 11m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/279a33b70c23494496bc8582c503e484 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77616c4effa4893a8384fcc2f2edc91 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51cb3b1da93147738967933407ef6302 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a8244ad526a46ee81ba4e4f7525d054 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9fb56d209374be9832f3c56b9f837fb : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a585574be0b460facdac820568fd57d : SUCCESS in 11m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8362fcb3dc594bbbad7f3d9997f5bd20 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4049c3c1f5ff40ef99e60a337318cbd5 : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d0f8252ff7834fa9a57040d198e95b68 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fb51efca4ce4a8ca5bd3faaa2ad29fa : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a76f0c211dc74cf38edf3e502227ab1c : SUCCESS in 1h 14m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fb2427dff124cf1848a43d5adb97858 : SUCCESS in 1h 28m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64c7bbf24bb6445aa2f54fb190290ce9 : SUCCESS in 53m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fb4a3d0470664767baa4327302e155b8 : SUCCESS in 32m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce580f1127b64368afe788c0486d18fb : SUCCESS in 37m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ee91a0e88b34a30bfd6bba4fba93a77 : SUCCESS in 55m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2f14befed81044759d582254057a6e0e : SUCCESS in 1h 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d02169064ef44b5801c7a472391d60f : FAILURE in 41m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f1792696a38245d2a17644bdf38c66e3 : SUCCESS in 2h 09m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e964d90cef974453b547d7f146af813c : SUCCESS in 1h 44m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5ae47d4af50a4e71bbf06e68aa3ce4ed : SUCCESS in 1h 21m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/11edfd1da56d451bb2755c4009d1c06d : FAILURE in 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92b2f41a482d4c5b61ff6452a2ead0e309416c37","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-02 16:10:39.000000000","message":"Abandoned\n\nDone testing","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff362d1acf0438a8924d9062c4155bf471352f89","revisions":{"ff362d1acf0438a8924d9062c4155bf471352f89":{"kind":"REWORK","_number":1,"created":"2021-04-01 23:01:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/14/784514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/784514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/784514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/784514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/784514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/784514/1"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-01 23:00:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-01 23:01:03.000000000","tz":0},"subject":"DNM: Testing the agent","message":"DNM: Testing the agent\n\nChange-Id: I38a1dd7c4f10717a90d83a568cddf48d0dec2972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff362d1acf0438a8924d9062c4155bf471352f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff362d1acf0438a8924d9062c4155bf471352f89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
