)]}'
{"id":"openstack%2Foctavia~784628","triplet_id":"openstack%2Foctavia~master~I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7","project":"openstack/octavia","branch":"master","topic":"784628-patch12","attention_set":{"33130":{"account":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"last_update":"2022-02-01 10:46:56.000000000","reason":"A robot voted negatively on a label"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2021-12-22 11:53:08.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-01-28 20:10:49.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7","subject":"Add simple notifications for load balancers","status":"NEW","created":"2021-04-03 16:45:22.000000000","updated":"2023-03-01 13:26:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":18,"unresolved_comment_count":11,"has_review_started":true,"meta_rev_id":"0848282680aff3d2ec0aa8797f4cab0dc1c4161e","_number":784628,"virtual_id_number":784628,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-31 10:28:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":-2,"date":"2023-03-01 13:26:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-03 18:38:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-19 22:20:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-01-27 17:12:49.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-01-28 18:33:38.000000000","updated_by":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"reviewer":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"state":"CC"},{"updated":"2022-01-31 14:10:13.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"CC"},{"updated":"2023-03-01 13:26:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"141dca8509dbd6ac13d8509dc10fd6fbe3deef97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-04-03 16:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b436e1c589cd896625ad404e6eb575458a2896c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-03 18:38:07.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b436ac76723d475683995f271ac3e1c6 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bdb1c4c561f40469bd578455896f943 : FAILURE in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7e5462a65d6451fa8c435e9bd3241f7 : FAILURE in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b4982ae90e540c7909faee2e7ad7576 : FAILURE in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/218509bba6a041709bad59bfede3ae22 : FAILURE in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/339cb1cc33c64cd98ca1423aa1b338b6 : FAILURE in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be47d3dce1c44a94ada72acf88154fc6 : FAILURE in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353a09eb8c4b4113bd6e54443498c0ff : FAILURE in 5m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/234c2cba594547c3972f4d4a0e916f5b : FAILURE in 5m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2a25364def164d5b8c36e60951217ec6 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e6f194d7c2e40f39a4740c1a797ce17 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e4793b3fccb487dac8c3127257f97e5 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef4f9b5edacd482294521c976cbe01d8 : SUCCESS in 1h 12m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8f937fb9d794f6c92ad8c3ef7481535 : SUCCESS in 1h 51m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b96516b99724358b8c9125604ffa8ac : SUCCESS in 33m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8277e533a883479aa4aec501658e8c92 : SUCCESS in 36m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94cb9972d5a744d8b3bb1f6956c92aa0 : SUCCESS in 32m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/47e760f5122b425db6a7c7acec0611ea : SUCCESS in 52m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ef09b1997de4b91a2745c005a9ce6cb : SUCCESS in 1h 30m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/001edc73cbd84886975fc73028849a5e : FAILURE in 40m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1850c454b4ea40738bef48fdb69f9f98 : SUCCESS in 1h 27m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9517f47e2ff8460daf6ef9fc8abab8ad : SUCCESS in 1h 25m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/63ba160bfa5a4b16931e3af2d8749fb6 : SUCCESS in 1h 20m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5d628d1b6a014bde99bb6a152d45cb76 : FAILURE in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c0d4117d19af321296df9b49218b81a10cb9966","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-21 10:57:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c808c78c77c27af1ab6ebb57a28121b06003f7e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 13:19:50.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ea0e5c0236814b1c9a9075d1bd5902b3 : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0be121bc3da42b5979d51cc4fa41467 : FAILURE in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa567c2a46664d6b9c523519569c38f1 : FAILURE in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f38e61fe52f64afb919b3a7c8c372da5 : FAILURE in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c22d8a08ccf4cf881686ed7912675c5 : FAILURE in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90f1e86d351c4bbc87f4f5c94263e70d : FAILURE in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a10f1ac9cd34232a255a90dbf7888ec : FAILURE in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e118cb8a6d0143babc826e144ef66c07 : FAILURE in 6m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/85a9fe0d65e14245b5ad6479183f5da6 : FAILURE in 7m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ea6e3d1fd3364b63892169c4f25f57ba : SUCCESS in 11m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e79a582e761426d98f553b12bb1b94d : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c997dadd28c34bbc8a155881e742ab83 : SUCCESS in 10m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b662d995cd0443febd261a2eca429837 : SUCCESS in 2h 18m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31139c421efd47fc98d8f2c23ea32214 : SUCCESS in 2h 12m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a4e53c4aebf4cc0b670bd3fc7b397f1 : SUCCESS in 45m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ac426c3e8978471ca7867d8ed8299e1a : SUCCESS in 47m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/beb1fc8955ca4b158761aae0d138533e : SUCCESS in 34m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/edba9c5f2f6f4793a08ef15a45cc6077 : FAILURE in 13m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd271fdfe36c4d0b8dd9d5d6a0e7ee80 : SUCCESS in 2h 01m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cf3379e7072b40d8826e45f085e11288 : FAILURE in 1h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/28b50f1af0514c80bbe27703e5050e07 : SUCCESS in 2h 15m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5cbd63f62723437488500c919dee5778 : FAILURE in 1h 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/393443b77a384868afdae61f98c6667e : SUCCESS in 1h 52m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"412a8daa462dbfeb2f7198e1142dc1258a3d0c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-22 09:08:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"630d62f3f9dc9e5dbf5cf3f94ae0f1999797937e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 11:53:08.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e88eb8941b4e48e1b86cd48be586235a : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b621fbc6a4f4cfb8992817b9c24f958 : FAILURE in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c9c3dcd2c4e41179e8ed40dc308af0c : FAILURE in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fca88e81ca74a549bc718118eaacad9 : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f96a1ba7bc8248e8a03ca0f3573f802b : FAILURE in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ca9dcad67d84b51ad7c06d5a2a21c3c : FAILURE in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01881cf0a60d4c88914dcbac6a58e5d5 : FAILURE in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52313cd1fe9b4620ae73bc953c8fc964 : FAILURE in 5m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3272a4cd5a2045ebb58b9b5ec935e766 : FAILURE in 6m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6e02b6d9712c413ebcdd7ecace71fbff : SUCCESS in 9m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2858031cdfa54232bdc39c155298d4eb : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/96170c62d3634a51995e0771f366910a : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/209df6ece93f48759ee5e8fcac6d72ca : SUCCESS in 2h 38m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d4b1d5948fa4ba3b58ca4b5c5718355 : FAILURE in 2h 20m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5efa0c5ae62c4dceb3d3c14adf7b4983 : SUCCESS in 33m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/caef0ad3444d4c26912533d95f128e88 : SUCCESS in 59m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdfcaa38ed424238aa2cb3ce02b51899 : SUCCESS in 33m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0abb76abea7747079d7cf5fa7c751e0c : SUCCESS in 1h 51m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62caffb257da48559c95937fc177dd38 : SUCCESS in 2h 37m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/56305bc6e12343bcb3cd16b0aea5b445 : SUCCESS in 2h 10m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/be36c937ad9944a09144e7fab88a5384 : SUCCESS in 2h 30m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/714e3cec6e5a42ad9c56bcabbc15bde3 : FAILURE in 1h 57m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/b631a33e80944d9ba8e0cbed0c2ce227 : FAILURE in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7b71ca4aec865e1bdb2a3929976d1cb17deb21c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-22 12:17:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b5a7b0097eac8cbdceadc5a64262b38c9ed3410b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 14:53:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e9e41f66237c4503b60c17b9c7f35027 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/062c89331f6244c4821ea859f5dc15f9 : FAILURE in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94863fdc3c504d0fa076e5cecc89302a : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5283a2943954b49b0029cf1e97b702e : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b53e6a53ff094b66856dbf51088797b6 : FAILURE in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76976177b5d54a2caa2fae76e338073b : FAILURE in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70f407a279a74377b4c04de8be71a079 : FAILURE in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850adfafdc794306ab29e3932e55381c : FAILURE in 6m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/14b5e3d7bd1648ad8e44a7f8ac996df4 : FAILURE in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/baf4fccb8a7f4d27a1c8e2e5bdf0ca11 : SUCCESS in 11m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a806e8c145594f5598ee9a0ab3719a62 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b0051e2c68e4c65a7de00451bf84edc : SUCCESS in 10m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1913e044c3dd4a64b008f6a50267146d : SUCCESS in 2h 16m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd5a55be694b40c482ccdbf2f7e1a74e : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b05fbbc7e61a4425a9fbd557a393892c : SUCCESS in 34m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e167defb2d1b4e539c51eeec3d033a36 : SUCCESS in 46m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdebfe5c65044059b89da4545a470750 : SUCCESS in 31m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a8af41b754c74d9b9cdece30302ef686 : SUCCESS in 1h 37m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e7a356bde99e4b09810042ca21f3b49a : SUCCESS in 2h 02m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ceda2bfff6ea41a8a7aac42b846c8fdc : SUCCESS in 1h 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c2b063353dba4f07a9a5144b69efc0b9 : SUCCESS in 1h 22m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a011593f37fc4d2cb13194d2c6232bbc : FAILURE in 1h 23m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/55cc49e0aa1a4714a1644eec35ea8705 : SUCCESS in 1h 52m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"70896ddf34b85610e45de7b65ab7098be43312e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-22 15:44:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ecb742b7f6174bf1d1d14b41ca8fe44d19447c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:17:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1c039838eeae4691af432af0e81136dc : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dcb4f03f1424b609d81936964e056f3 : FAILURE in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23236b48114845438fc09c44c68cac03 : FAILURE in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a404d3a9e7a44bd5a6a0fab88b15cd99 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e49b2c476f04401a574334e05efc6b7 : FAILURE in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db75edbf5e7c4a9e92bd2a783c85a583 : FAILURE in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09d88a576777460fb6d177df0b8762dc : FAILURE in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e8668b7a6948f3a83a60f8a86530d7 : SUCCESS in 12m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a21dc1617b664e96ab0e2717eba286c9 : FAILURE in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f9d1cfaa39fe462683eca5a07d67e5dd : SUCCESS in 10m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8dfc7f2334a43a18ef01bd0f8e97ecd : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/661e45d94e3945c0aa99837ab0d6890b : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4958178603844f429fdb952aa6942d03 : SUCCESS in 2h 14m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edccacd7bbad4c99a5c89adb7dc6052c : SUCCESS in 1h 25m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/325aa908253d457c87769ecebb9f5c1b : SUCCESS in 35m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9c7a8d1368649cebb526e114079d398 : SUCCESS in 45m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/156bd8d79e3944a794def677b4759d76 : SUCCESS in 30m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eecb112c7f8942d3ae95e68836159460 : SUCCESS in 1h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/02ea7c8fe4484d908bfdc7a37cab892e : SUCCESS in 1h 47m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7108a9392da4481687e8119be81561d7 : SUCCESS in 1h 27m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/13f8b5a427ff460498aca433e894621e : SUCCESS in 1h 40m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/902bc91877a54d60a831c4156b2c902c : SUCCESS in 1h 18m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e1001c6027e426b9335b9723299c804 : SUCCESS in 2h 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ffd8f05d6b720cd0e858e05df8dcea50709242ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-23 01:44:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ca95e6d50c17dfbb9c287d62762c5e537e0dc497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 04:09:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1b47be87d8a643709f0d806c3f82f997 : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc6ac2feaf304852a1d02a375a6965dd : FAILURE in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef71e63d588b4626ac3e41200fe1d40c : FAILURE in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a719eb23667421aa7362df283a73c15 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b59141a7654955b8ea54e024cd0b33 : FAILURE in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6bbd4fbbf0b439885491bd6bdf21e46 : FAILURE in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66073ae537f143bab7bba7ceba8c5687 : FAILURE in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc978807fec44093b396264c57a7a394 : SUCCESS in 11m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/924f274629c244118a011557c7f0bc9b : FAILURE in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6b774485434e49508728198f7ddf59f9 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e7bd8fcd6a4046fc96239956e2762fb8 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c24817bc2a6e4baf84b4ea163a041f91 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/797408dcb3cd429e99f7c2a8ff12973a : SUCCESS in 2h 06m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc42e81bea504456a07246d1be62e9f4 : FAILURE in 1h 54m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc0c712efeef4196a198c6288cd4f961 : SUCCESS in 35m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b2c11b2c41c49d08aa4bd12767882c7 : SUCCESS in 1h 15m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ea42b6689694ecabccf57fcdf0ab414 : SUCCESS in 36m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/38dbd4df85c64ba8aaf2809c57266a50 : SUCCESS in 1h 56m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/02d08381c5e44c8387691f6cbdc6a4f3 : SUCCESS in 1h 52m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7e1d850ba1f54534843d86fdd6f327fa : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a0aec193df7b4872be2bb63e443c4d0c : SUCCESS in 1h 51m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1ed5baaaeb1149e0980d502e087fc25e : SUCCESS in 1h 28m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e0b25ab978f4d2c94732963aa8288a9 : SUCCESS in 2h 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6361c28892e588c2165d58f51cc1ad0981570a50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-23 08:27:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"88eec7071055c8877d078def5d8a92492a83b36c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-23 08:38:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9038c3e08847603025947b4e04fa1bb5016e0ab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-23 10:25:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"02da445c873f63d94f02d9f6490384fcfe04084f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 13:04:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c40807e992b04e37bfab9bc62430e33c : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78c8e83f5cad4122b82d29022bb1efce : FAILURE in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9baaa0d9e5ce4636b766b69c9ff3d8e6 : FAILURE in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc9d7c9c3c274d49b05c814f2d793bd0 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9dabda693e84af483c47940ded35874 : FAILURE in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfc45a565b854c90885c868eac0c75d3 : FAILURE in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dfafb363ad1434984474631fa64805f : FAILURE in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46440cd2c72e4f4aabe5fb0c59976e0e : SUCCESS in 12m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d65e707aae2745f3b4b0ee23dce26d6c : FAILURE in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fa1fb17a3b654119acd8d4c6fe83126e : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c4dfe7a8c9fb425284f83e7965cbc68c : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cb1d81a354d407d98de8ee42809b665 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d38fd7b80a164a59bd80f9bb03fcbfef : SUCCESS in 2h 02m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd3a1f5332d644a1b188abf478942e09 : SUCCESS in 1h 17m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6a6adce260b44ae8b323d8b87c379e0 : SUCCESS in 35m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c63609f18be94f6faefa4a7770d3ad7e : SUCCESS in 1h 10m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b680c7986ff4f6d8736f740c35ee039 : SUCCESS in 38m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a8db479fc24b4828ae0ed553eea2010e : SUCCESS in 2h 25m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00d68b61e71e4a54a1c0831ee6e2c258 : SUCCESS in 2h 17m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/25f70d0617014450b6db3648eef5d172 : SUCCESS in 1h 47m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a34c7de6c8534f50bd661ae54e825d58 : TIMED_OUT in 2h 34m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/020feb4a6c664007b0341715193b1db4 : SUCCESS in 1h 40m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc55a89c1782422a98ca976174ab4514 : SUCCESS in 2h 09m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"54006364e529859d0fca7ba9c74eb77b7258025c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-28 08:13:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ae9bdbd974f853639918baa7402c7ae15bf94300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 10:35:04.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d790c63eefc54bd6bac710e60f5e9370 : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c7d04eeff184221a5e83e01523d2b55 : FAILURE in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a762f16c2e6a4dfbad06252737df48ec : FAILURE in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ced68ce58240c19845130b9b070045 : FAILURE in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b7a558a850456ea75e1e7bc871f3d8 : FAILURE in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65568a5b500c474dbad2a557ff42dd45 : FAILURE in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c6fe38d60c54721b4938d671a4ee5d9 : FAILURE in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a4d6a3a5b94fb48d16795b22c8d53c : SUCCESS in 11m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1373299acfa04440a9a984151fee0ce9 : FAILURE in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b72c353e8faa4f609a4798d66923b74a : SUCCESS in 10m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6b2833460afb47f5a482c6a36f82f242 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a67b1b8ae1bc4226999f351a77ecac19 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97d2a4b160154ccd989fa5a521f64fd8 : SUCCESS in 2h 15m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/491a07d618134a1eb214e28af6cfe271 : FAILURE in 57m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42aea55170d14947a075ed238641769d : FAILURE in 34m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8acdc8653e324c2c982f905947ef62e0 : FAILURE in 59m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50cd094edb3b4eff918ab4ae3aa74def : FAILURE in 33m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/26d0da8d24e644e8b8b9109e4061611c : FAILURE in 1h 28m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7163d11a9fc940fc940872d14b6b46fb : FAILURE in 1h 35m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4ef68c450f5743c0bf06787c7f32f689 : FAILURE in 1h 18m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/038d1055d6f54d269f2557a28e9e9cd7 : FAILURE in 1h 15m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e7bee1f4d29a4bc5b899a6922f1e3255 : FAILURE in 1h 04m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/749889b10aaa4c6f83db013f1809e91c : FAILURE in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8386f18c0cdcc9c46efcfaa9d4481138d462f723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-28 12:35:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bd8d7426c271de27d258cd46171c3e921242b7a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 14:50:44.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/62de3df9d60c46309a72c2af8513d5cd : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04879631a50841709c9f857f5d725513 : FAILURE in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e17e3b888a1248b1b2212e48a7c54d88 : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70897a25f7fc4abdb943c3795327944e : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99e5f3fa56f845e1a3d3d28dfe3c9c4f : FAILURE in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77aa6851cf774db5a544ac1898104451 : FAILURE in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af042ccfd3bb4b60ba0f860804540c32 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6361834f9614c139e8e375a21178a60 : SUCCESS in 11m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/77ee9104310b4d8d8e8e2d927dda5d3a : FAILURE in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8f793c036514d07ad20374c4fafa8d4 : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae9a21e810f74c1b80480f94ca48a011 : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00c57ac7fb8e4699b22d5b40214a0158 : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7646e4ad1bf741fca456b2fa5c01489c : SUCCESS in 2h 09m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81280991189b4817a90f909b6b5411df : FAILURE in 1h 14m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/451f123e92a747c0acd1f88d7d725ddf : FAILURE in 46m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa226675cd354782afa2a88953627d29 : FAILURE in 1h 10m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdc3a4b8ff824ce2864dc615d3546ff7 : FAILURE in 44m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af8130288a914d759c314f1eb2811b36 : FAILURE in 1h 09m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c20350858fcc4ba582f838987bb93afa : FAILURE in 1h 41m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71993161d07740b6a60b529820a90027 : FAILURE in 1h 21m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9f244a3530342bab9e271151376f641 : FAILURE in 59m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b45ab677ee66480f86d7357edf8da1c9 : FAILURE in 1h 24m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6a2ed483b4a4fdebf09822744110a70 : FAILURE in 1h 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1ad3f2580c47265d58ff360dbc0562e219c60b8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-12-28 16:03:45.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5cf37ae52517820db333dcfa291a7e9ab76d5ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 18:09:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3ef2e0cf5b664812b7c599660ceb91e4 : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7672bcf86f294d41b0e4efc298e5c18e : FAILURE in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53e67ab0bf5c4373b11f725831d6dea3 : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e9a7df50435481093bcb2e1e2bec677 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ca80fdbb404c668f3399706153aa92 : FAILURE in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d89103f5c1864f13a9bda21d76ee2265 : FAILURE in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/053b947f7b5a40e68a5c3dc72bfc38b8 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58cfe3d154fe4a318b3902fc93b6bc39 : SUCCESS in 9m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/51c80207125a496a997657198bae3cec : FAILURE in 4m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4be18b90535249c28d230f7f62e90688 : SUCCESS in 8m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5ec5e2d6e7ab4c19a00e527b598ccc79 : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2f85f9c52f24252b344ceb989c8984a : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7bd693b465f2464890ebde253c074c92 : SUCCESS in 2h 03m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0dbbf6ffbe0463dbf1222c6004b48e9 : FAILURE in 1h 04m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/89bd55f658e4442bb332c2315ec95cf6 : FAILURE in 36m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2bd4c1f903294d4faf28fac4f26e90d9 : FAILURE in 45m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18ffb99ec50d49e98a42beae464e12f8 : FAILURE in 34m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5b821212074144babca2f9472186401d : FAILURE in 1h 05m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ef1203201cbb49cc8d0bada91965486d : FAILURE in 1h 23m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d74b7b517b5843be977bc63454ef7510 : FAILURE in 1h 22m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b56fb4801ae040398281ddacc01f58d5 : FAILURE in 57m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6e445d0adb7b432386091af68106459f : FAILURE in 1h 07m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6be31a7d011f4d2dac8f1cd296294f42 : FAILURE in 1h 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"cd63db276a1edec5b644c62c5c3a29be13332957","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-19 22:20:16.000000000","message":"Patch Set 12: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":12},{"id":"16f33ef8043f2a5baad85301d36a7bc5f55d0a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2022-01-27 08:52:06.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"adbdef7358c46fc55e9344c517ecffe32478886a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:08:08.000000000","message":"Patch Set 13: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a4230e312fd4c09b83f7432f0cb4d19 : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87f5a5d0254746429e4e137b80fd5462 : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc1d04b52024ef093c4593a6a4d1b2f : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c59c6851d214071b94a18236cc2d8f6 : FAILURE in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71a2e902e43d450588e78e2ce5b5bfb1 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c40c4f6adf274942914b4957db07cdbd : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fee9f55bf994cdebeb11b1e9372962f : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c27202702d4477b843c09944912d7b : SUCCESS in 11m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8459ef9df9294c3b80513454207991eb : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f54c98f8e9044c7391866e605aaac3b8 : SUCCESS in 8m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b0f22fe2bace4183941c0e303c32287d : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2d6d5ee37c6420bb3a587dbc85dccf2 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67a7c6146e6b4d9cadc45c91a0bd0533 : SUCCESS in 2h 06m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1351366b1d584f009de0e661de833023 : SUCCESS in 2h 04m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd8782b4d9da4c419b18b1c0b3e1837c : SUCCESS in 36m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1b3adc088dd46e08257f3df84734bf7 : SUCCESS in 46m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59abbeb39ac74eb0b9a66fe1d31ad0da : SUCCESS in 31m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c517cea398b24b66a4fbdb30f367233f : SUCCESS in 1h 28m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/68ca4dbdaab04a958005e65716e5679d : SUCCESS in 2h 01m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c7dc2b2fd11489c8c9af2d3d8a2a7d0 : SUCCESS in 1h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/62174f0d35d6455b99922166450a5c18 : SUCCESS in 1h 48m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a820533ce6864ceca7da517321784e82 : SUCCESS in 1h 19m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fec7a57f98f42f6a74bd10f2cc79eda : SUCCESS in 2h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"827af5f04753085ae7e806b152ba8508ff1e7b51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2022-01-27 12:13:29.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9534949790520ed86466303f4ca88799edab3d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 14:22:38.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e35438f60704cd490963ae6d27754d6 : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbadce9b15444adaaffd8fde38a2d613 : SUCCESS in 11m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d8449c2bb164d3fa20c0ebccd9b19b2 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7784084005545cb9a368f76734e00ef : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d56c47e4f534b02811a9e334dc1149f : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e28bbe28972429d8eab937695b92e0e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50ead2c48ad14b88b5ff348f0b236daf : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f49815f99c4666802cf867f43812fe : SUCCESS in 11m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef09c0d44313431d9290002222af2820 : FAILURE in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d95c4f2e4a634a16a9b8f0719af11e82 : SUCCESS in 8m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b7c7ce37ea14ca08eba9ab86bcf8f39 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a8ead45439e463a80e556a2f0ae6fd6 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b69084ec8d2423dacd4faa5240eb52c : SUCCESS in 2h 06m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82c489260d3c4caba91416beef093bcd : SUCCESS in 1h 30m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3528fddffe5646a7bb4b8ee03688abc3 : SUCCESS in 36m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/037e0b29daf449f1abb2f33d9915beef : SUCCESS in 53m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/047ee1aaf05b4240b61a961949480d6c : SUCCESS in 39m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4db8411893d643bfabdeeed7c4405060 : SUCCESS in 1h 46m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72e92ff31ebb4755a1b251c7f11f440e : SUCCESS in 1h 58m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb66c2189d174cee9b6239c6c0f47f1b : SUCCESS in 2h 01m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/886c018296a24bde9ed377b144d64648 : SUCCESS in 1h 41m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6409e15304194ea2bf34290e9c0cd3c7 : SUCCESS in 1h 47m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e9abac2a008469d93b11f7c5c3f486f : SUCCESS in 1h 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0d61d5a8acc0118e5efe6fee8048581b4247079c","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2022-01-28 18:33:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2ae9ad86ba3f7ae947340982758ba933ea2d028a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-28 20:10:49.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b97c19e6780b39c45241b4a6d9ae279c3130da30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 21:10:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/07db923091ea4038b8a8b51c426dcec8 : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20166d19e001413d943e0b2c9575b268 : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb23f610e328403ebaa4a0abd6c4f8c2 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e55dcb0202e4fdf87533fae55fa6807 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e02af9f7ad4a50a9eb3fd884befa0c : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26d3ee8ea387401f826682fbe65e5fa4 : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16f441bfa0924bd2bdace291ac9a10e8 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24d612d34fe467c825e479563230522 : SUCCESS in 12m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8288febddcc240499124837eecb21165 : SUCCESS in 8m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/383066c5e3784494a6b65b131d0dbf57 : SUCCESS in 12m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f5d06456d9f1470c9a965f60a39bc7c3 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fff7e62c52e3435f924fc87d4672b871 : SUCCESS in 11m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6aa52c439d4476abd51cd4f03d3864e : SUCCESS in 2h 05m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b7e1162835f469f95fa41771ea9a247 : SUCCESS in 1h 21m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f0927e26daa411f8b1e308acab9c965 : SUCCESS in 35m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d30a17ecbda406da0e448260aad887a : SUCCESS in 46m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13a2ce9676774f5cad7e06b7835d4178 : SUCCESS in 33m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0016f732156f46c6969cbdb958e33498 : SUCCESS in 1h 31m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a9aeb8ac1914d6bba08d79ffb75fe31 : SUCCESS in 1h 59m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c9dcbecd2a8c443c974e0c486df6082e : FAILURE in 1h 49m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/81e6fc890eab4776a874abec5a534adf : SUCCESS in 1h 38m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fa484413f27e4f66b1d84e2787d92dd4 : FAILURE in 1h 28m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb7b1e14f2554444b22d94b666498d84 : SUCCESS in 1h 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f130ad5d73391e07aecc51e650de8c2262234945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2022-01-31 05:09:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"00fee747e6c8cc28b29eeffec3dc025becc7921e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 07:10:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/57113207e20c45d490c0abaa440c30e9 : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/862f56fc0f034afebc0a796bd7ffc91a : FAILURE in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce3ad29055494d0e869745e1f91333b3 : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cec86f87f134ab3911fe313bd55f4d1 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/decbee7a62a346699c848181b0e8e40b : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cff00102ba54da59058e6d9b28e377e : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5cd2c1ca5b942efa882df8e4deea5f5 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275580323c2648f28dd350b5d271fc42 : FAILURE in 6m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/59399f37baf144ecb44b2cb400f53d50 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c9cf0cda9e794c2883f50595904498ee : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7b56ee5b45c848c3a1e48de1650fa8d7 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1df3349f70eb4d8ea49a844b884a743c : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f913c29131d4134b66e6a520f0ea237 : SUCCESS in 1h 48m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/154aff5772ae44f4a5544d90a066e666 : SUCCESS in 1h 56m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64599846e0804505ad949fd7a06ab691 : SUCCESS in 37m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7f6f11d05f144839868ec447fc58e89 : SUCCESS in 46m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/927a917d6f6c4a52b1eeec0cfa5243fd : SUCCESS in 35m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab6f4df745834079802b3614b7a99b75 : SUCCESS in 1h 21m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d48c915013b34274bec96227911e0b60 : SUCCESS in 1h 31m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a16d2d3ea071461c8137a309d200e62b : SUCCESS in 1h 58m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a1c12f436f174e36885ea2a5aa8d96e2 : SUCCESS in 1h 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2f10da87741d4b9c8d5500c5edc1cde5 : SUCCESS in 1h 34m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9851d68b9edb4db9b049195d9c2bd3ea : FAILURE in 7m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2e2362031df0fed0f04a74bf8b495ec57770491d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2022-01-31 08:06:45.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f6d4d0abf72451ccff30c7b21b931856a2657e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 10:28:28.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d1dd5953d174c42809914d79618e10a : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/643ed0a2a55244dc844f5c9997645f27 : FAILURE in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3871775928c1438aa25adb58e557ce28 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2ecb4271c34ce1a3e0a74663b31d3c : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da9819223439481cbb6bba01c54cfef3 : FAILURE in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a1630e163eb4dc0b3a301ceb9c28e64 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f59cc92404af4639ae34bdd97ea08945 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c31d1fd7bd424f8a92f56be301fc59 : SUCCESS in 13m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0a40057531824c1eb87fd465925880ce : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/66e044dd4a834332ad05db5c32fcb1c7 : SUCCESS in 12m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f5c8d8eac58a45649051b4a6bef880fc : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a7e1b1b5a034e34983a0ab015d146d3 : SUCCESS in 9m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e4aa675038f4223870fba2ca9c7fce2 : SUCCESS in 2h 14m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c52445737c434175b34760cbdecdd701 : SUCCESS in 1h 20m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60a788c2a0a7489482f180c730115650 : SUCCESS in 35m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1a26d6dace2c470aa8c54a337bf5a76b : SUCCESS in 47m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f343c04d9408412f966e1bc81f9530c0 : SUCCESS in 42m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b931eed1f0b84047a92e1caeea70f905 : SUCCESS in 1h 48m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3aa220551e2b4401b4b124dad7a6bf99 : SUCCESS in 1h 38m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d3affbf6d1ef4a84ba4aff1b494b109e : SUCCESS in 1h 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/834a0b1200154ddf9046091ac7f74b1f : SUCCESS in 1h 49m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cd2b0ed3cc9b4424b5744d6501901797 : SUCCESS in 2h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9966a594bc1e4dd991409f54d4ca1198 : FAILURE in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"046793dd133a2708dea97ef1b15a14fa583bba4e","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2022-01-31 10:29:47.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"01c70ab5f3c2b1cbba0732b5b51f249d1f069afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 13:41:16.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bd44019a4fe74cc0968bece6d21e1ae2 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39124760a5ea4b7f9ebe648105a206bc : SUCCESS in 14m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcdfc9f77c284bb08400b4d580614457 : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4edc8ff543a4f84ad63b5c5a3d7c1ac : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11c95a30bdf54c7a904d8b9e6c8f2feb : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b8394bcdce84702814cbde1784580a8 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/194cedb5135d4477a8d0860c6d19b0d3 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4848f74801438ca235be28b91d2660 : SUCCESS in 14m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/48fcc4254ad84bf7a46443c096be3476 : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5be0fe5fdbbd4d889eec68e78f047e27 : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/36a35c0940124ed5ab2bd0f93e83a552 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f49d22a8f30a4c6583fedaa8e4cf6dd8 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d4856ae3a6742f6b6d90fd15a29fe85 : TIMED_OUT in 3h 04m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ffe83f887874703b970da9091475a75 : SUCCESS in 1h 29m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6126eaac92ac40919ced8ca003575a1f : SUCCESS in 39m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3c3292c13db4fa79105313d72b48dd1 : SUCCESS in 49m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af9eb17e66444e638ea27ae1ba938380 : SUCCESS in 31m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/37a2268522bb44b7b0b380c5f08c3880 : SUCCESS in 1h 36m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da6f325b7d574061931cc37a39454247 : SUCCESS in 1h 54m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ccd56f55e4474b589c4d5e7d1c89dc8e : SUCCESS in 1h 46m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/355e910154be42bab7354e2c522f5a3e : SUCCESS in 1h 49m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a7e96e60994641cdb29208ba33351d5d : SUCCESS in 1h 30m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e129f85c6a04854a0060983d83bc6ee : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f4e50a72eeda0a73258e1b41ee0f492c93ece0c0","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-31 14:10:13.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"17dbee49a7a5a7ad1097529e6ac15736b16738a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 16:24:27.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c1537b8f286a4636afa6d2224828fc7c : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99e4dcc4163f4d74a17f46826a09d067 : SUCCESS in 9m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/483461eb33b0483ba488dd71988bdbc7 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c51472d9ab6d4a288954d6e154351a88 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23c7f26a31bc40b7b73c100faafc4a74 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f69eba620f8b476aa47aab7436f3d689 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4a8da195b0d48b19cfdc15320b93b00 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f02b1a1f264a51b0d830f7a4cd638d : SUCCESS in 10m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f4757d84efe9420caa42b9f8087d8431 : FAILURE in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/301057b896974cf79ac7472681161ee5 : SUCCESS in 8m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a43eff67124e425cb97728b91350d17d : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8bf2467789045c59c47ab33c389f383 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d0d47effa7c42e1aca13d92327b12d4 : SUCCESS in 2h 10m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0466a290df0a46158d24f099ece4b473 : SUCCESS in 1h 45m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa84d1145e144646a8b0fb904bb58290 : SUCCESS in 34m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dff0d94a6dfc413dbd756e771e4bcc6d : SUCCESS in 50m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c05e646e4b947298facee3bf06b3c44 : SUCCESS in 34m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1197670554554c57af302340614781f9 : SUCCESS in 1h 31m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df9e08abb6b646a3a78155c1c853ca1e : SUCCESS in 1h 53m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0e784c654b96414790dba15c222b1d7e : SUCCESS in 2h 02m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f5f02dc2dd2f46aa976a9088c96b4f4b : SUCCESS in 1h 52m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9f4437322a1d4f75b315a41293e03d0e : FAILURE in 1h 31m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/06ed63b3114b4d26830ae1b178304ff3 : FAILURE in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9f2b2631a579b66ee5f21eafccc46b43640c453f","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2022-02-01 07:46:51.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"291585d3875e931f902cf2a4b5dd3e230d2b02c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 10:46:56.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8da7a1fe711a4dfabb58dda5ec0559eb : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae58d0d0e3b84b3a835a066b8c5df0ac : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af539248f18743b4858e228efdcc27dd : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db8cc9d5fc24933b27b78d88797590b : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cbd67a13ddf48389cd9a7f89ab322a5 : FAILURE in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e399fe0b572d4d2ab4506b0c644b2d7b : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/706285eaef4f47a2b4cb2c54c98e5da7 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9f1bce1d364b2bb2f67d078fbb8d5e : SUCCESS in 13m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eafea0d1fa1c4a4d98e63cacc578e7fa : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/77fd08a25ae8492f890bd5c08e3b345f : SUCCESS in 9m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a35f22b0cd7a4c08b1cdd041549bb007 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a71e06cea6554f37ad859b78952767df : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a085561ed75a4271badc8894d39dc8cc : SUCCESS in 2h 52m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce3af04fb73f4fab9493a40af8e30a7c : SUCCESS in 2h 13m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa5c55c3f067430b935e24cffcf837ab : SUCCESS in 32m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55834179f39f4caca0817bdd0f680964 : SUCCESS in 1h 05m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/644b9e3b558f4b2287091bb81a047465 : SUCCESS in 43m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de63730105cc4d0392d9608792d9b91c : SUCCESS in 1h 48m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1ce5ab17c8245209a90d30d57a70b6a : SUCCESS in 2h 29m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9fc1b57d85d745ca9d877886c09ad3c2 : SUCCESS in 2h 26m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8989abc5aa5a4bafa141ce43b8193a08 : SUCCESS in 1h 20m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/98ac8de0b80a40408c5def9088c9e54e : SUCCESS in 1h 43m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a226d4df28f446c48600196a332ebfcb : FAILURE in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0848282680aff3d2ec0aa8797f4cab0dc1c4161e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-01 13:26:05.000000000","message":"Patch Set 16: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"9866381a573758387c368bd5ff22b1490ba0e966","revisions":{"200268d195933b88053b186e12242f9b42fbb025":{"kind":"REWORK","_number":1,"created":"2021-04-03 16:45:22.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/28/784628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/1"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:45:21.000000000","tz":0},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/200268d195933b88053b186e12242f9b42fbb025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/200268d195933b88053b186e12242f9b42fbb025"}]},"branch":"refs/heads/master"},"b0ad03750e34552594c52c0912e09a6e0074bbeb":{"kind":"REWORK","_number":2,"created":"2021-12-21 10:57:20.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/2"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-21 10:57:15.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0ad03750e34552594c52c0912e09a6e0074bbeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0ad03750e34552594c52c0912e09a6e0074bbeb"}]},"branch":"refs/heads/master"},"495e15b3a5afcd97f17860f6a176725e99ed5cf8":{"kind":"REWORK","_number":3,"created":"2021-12-22 09:08:30.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/3"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-22 09:08:16.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/495e15b3a5afcd97f17860f6a176725e99ed5cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/495e15b3a5afcd97f17860f6a176725e99ed5cf8"}]},"branch":"refs/heads/master"},"f7d60d8ab75fca1f206e297ee664000c47c4ff3f":{"kind":"REWORK","_number":4,"created":"2021-12-22 12:17:34.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/4"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-22 12:17:19.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7d60d8ab75fca1f206e297ee664000c47c4ff3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7d60d8ab75fca1f206e297ee664000c47c4ff3f"}]},"branch":"refs/heads/master"},"5eeaf4966114cbd9ed7b7a942c6a7b7e0ed0358c":{"kind":"REWORK","_number":5,"created":"2021-12-22 15:44:21.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/5"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-22 15:44:06.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5eeaf4966114cbd9ed7b7a942c6a7b7e0ed0358c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5eeaf4966114cbd9ed7b7a942c6a7b7e0ed0358c"}]},"branch":"refs/heads/master"},"3d2e2c21e52e37728d3d771f4b9e4034aa4ab4cd":{"kind":"REWORK","_number":6,"created":"2021-12-23 01:44:02.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/6"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-23 01:43:46.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d2e2c21e52e37728d3d771f4b9e4034aa4ab4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d2e2c21e52e37728d3d771f4b9e4034aa4ab4cd"}]},"branch":"refs/heads/master"},"7ea1d435b32247b3775125e94355d13b21f043c2":{"kind":"REWORK","_number":7,"created":"2021-12-23 08:27:46.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/7"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-23 08:27:30.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ea1d435b32247b3775125e94355d13b21f043c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ea1d435b32247b3775125e94355d13b21f043c2"}]},"branch":"refs/heads/master"},"f871a86e0b76223d9817eb09a3e608625d3b5d08":{"kind":"REWORK","_number":8,"created":"2021-12-23 08:38:57.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/8"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-23 08:38:43.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f871a86e0b76223d9817eb09a3e608625d3b5d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f871a86e0b76223d9817eb09a3e608625d3b5d08"}]},"branch":"refs/heads/master"},"e1d1c723f272e4f7498102781ebf8a5d97faf9c0":{"kind":"REWORK","_number":9,"created":"2021-12-23 10:25:12.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/9"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-23 10:24:56.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1d1c723f272e4f7498102781ebf8a5d97faf9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1d1c723f272e4f7498102781ebf8a5d97faf9c0"}]},"branch":"refs/heads/master"},"086e3ca8a2ccd5ea1ee66b42461ce1a3c55b6601":{"kind":"REWORK","_number":10,"created":"2021-12-28 08:13:10.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/10"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-28 08:12:39.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/086e3ca8a2ccd5ea1ee66b42461ce1a3c55b6601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/086e3ca8a2ccd5ea1ee66b42461ce1a3c55b6601"}]},"branch":"refs/heads/master"},"d908a822c5addd1c6c64bd5a942475cb9982ba9d":{"kind":"REWORK","_number":11,"created":"2021-12-28 12:35:20.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/11"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-28 12:35:03.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d908a822c5addd1c6c64bd5a942475cb9982ba9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d908a822c5addd1c6c64bd5a942475cb9982ba9d"}]},"branch":"refs/heads/master"},"189a6abad75ab683171f5d39505bb3e5ca740619":{"kind":"REWORK","_number":12,"created":"2021-12-28 16:03:45.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/12"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-12-28 16:03:29.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/189a6abad75ab683171f5d39505bb3e5ca740619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/189a6abad75ab683171f5d39505bb3e5ca740619"}]},"branch":"refs/heads/master"},"1e6aa2e739cc98e9f25f3af47d78626a61d3b019":{"kind":"REWORK","_number":13,"created":"2022-01-27 08:52:06.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/13"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2022-01-27 08:51:49.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e6aa2e739cc98e9f25f3af47d78626a61d3b019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e6aa2e739cc98e9f25f3af47d78626a61d3b019"}]},"branch":"refs/heads/master"},"5ce341167ad0f9d86e920342a5e6e93868f209c8":{"kind":"REWORK","_number":14,"created":"2022-01-27 12:13:29.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/14"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2022-01-27 12:13:12.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ce341167ad0f9d86e920342a5e6e93868f209c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ce341167ad0f9d86e920342a5e6e93868f209c8"}]},"branch":"refs/heads/master"},"7a13f4da283df84e6ae7b701d1b031efa9162210":{"kind":"REWORK","_number":15,"created":"2022-01-31 05:09:44.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/15"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2022-01-31 05:09:26.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a13f4da283df84e6ae7b701d1b031efa9162210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a13f4da283df84e6ae7b701d1b031efa9162210"}]},"branch":"refs/heads/master"},"9866381a573758387c368bd5ff22b1490ba0e966":{"kind":"REWORK","_number":16,"created":"2022-01-31 08:06:45.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/28/784628/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/784628/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/784628/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/784628/16"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-03 16:44:18.000000000","tz":0},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2022-01-31 08:06:29.000000000","tz":345},"subject":"Add simple notifications for load balancers","message":"Add simple notifications for load balancers\n\nThis patch adds the very basic infrastructure to send notifications\nfor the creation, deletion and exists notifications for load balancers,\nthe framework can be adopted for adding this for other resources very\neasily.\n\nChange-Id: I945f3b2383a46d0009af6a9f8cd6a666cf2c1ef7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9866381a573758387c368bd5ff22b1490ba0e966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9866381a573758387c368bd5ff22b1490ba0e966"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
