)]}'
{"id":"openstack%2Foctavia~784847","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ic63929d8864e5285baf70dd85e6362988bf2863f","project":"openstack/octavia","branch":"stable/wallaby","topic":"task/41491-stable/wallaby","hashtags":[],"change_id":"Ic63929d8864e5285baf70dd85e6362988bf2863f","subject":"Fix weighted round-robin about UDP and SCTP listener","status":"MERGED","created":"2021-04-06 10:14:25.000000000","updated":"2021-04-08 11:00:50.000000000","submitted":"2021-04-08 10:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784847-task/41491-stable/wallaby","cherry_pick_of_change":778132,"cherry_pick_of_patch_set":3,"meta_rev_id":"a4210b961f6c9d2fa44e2f2591e53505a6ee7e8a","_number":784847,"virtual_id_number":784847,"owner":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 10:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-04-08 09:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-04-06 11:09:04.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-04-07 15:57:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-04-08 09:21:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 11:09:04.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-04-06 12:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 15:57:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-08 09:21:30.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"48443ae4435be4ff61947dc3ff6d63f06c22126c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 10:14:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5a7b7dfe8563dc94e155bd04a53b4149785bb672","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-06 11:09:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f80c57441f9850e39bb345d7b83eb63ba56b07b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:27:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3392408034f84a0aa02c2e2f8c036eb0 : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8192ec1df4de4daf82dcfed5d7002173 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b85400effa4f4cc69f658bbf6cb42e2f : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee5047c7a7e24ccc96b847fabbb1b408 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0e9623aed39482185d1f0a3b65ae1fe : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bfee94829ae4ea0991fc19606ad67a0 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72bff5639d54ea5b2d9a40f1a6e1359 : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c16d57b69c154d889c04e393d5254b81 : SUCCESS in 6m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5adb938c2c9c47c6bdcec5a78016c227 : SUCCESS in 7m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8edfca64397342148cc0f50b51a29815 : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/45d008fbaeee49e68826f8d0fbe234af : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c3145d607fb4a2fb56e3a2f3cccbf12 : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/75870a3ea8d84032ba82798c5e20322c : SUCCESS in 1h 21m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7df3371507a241e683dc6165541f597d : SUCCESS in 1h 46m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a01f754645e24788b85bae41a3bf080c : SUCCESS in 42m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a04a1a152cbe48899d620f32c322138c : SUCCESS in 38m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0a1bbb5120a4e9f82f866a645a92810 : SUCCESS in 40m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4453d4810303420dbc7052ed4360e6fe : SUCCESS in 1h 08m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5bd3721b1fdc41f193b04cab41edec62 : SUCCESS in 1h 37m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d769ae7ab7b496c8654849ca2193ccf : FAILURE in 44m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/36a2567397fc471f99cf00864324c148 : SUCCESS in 2h 06m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f384d03900f048f5afb375d53a2aa923 : SUCCESS in 2h 09m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/80bb5d1c585649f0a2666832dd2645fd : SUCCESS in 1h 45m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8d20c6d4e31e43a6a172e505b019300f : FAILURE in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0a9354c1a84e26bb49f9d62ae02b01c538d50b2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-07 15:57:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"243f2e70a8c29dba592370cfb53966430742ec13","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-08 09:21:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46184c6f581bc5dd41d04ed7c0d442d4ecec7903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:21:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b54fc554a18c176ae2048d0751a62512c180fa7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:59:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88245145de2549c5932ba0077ff85812 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0533bc441942968b560b05ddcd7ef2 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c4d1ecc1786493d9d510c7f1d79b7d4 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd911d85f774529ace061639f0f2236 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0fc50c43a2847d89593d3dcd94e3db8 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1644af14fb44991ba2bfe1e7326f5e6 : SUCCESS in 5m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5dda7f0fd85477facd455870ef8e273 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb7db35e3cc34682ae4e08a4e062728e : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9dc5626634bc419681e52103e9efd791 : SUCCESS in 1h 07m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bda73fe369da46669e0966c1e94a6734 : SUCCESS in 1h 31m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/54a62a22c7db4ef7aa73dc56309d917b : SUCCESS in 43m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bf2b1c769447443f8e085941d91ca711 : SUCCESS in 37m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b868f69c967144c19f3b3cfc907957d4 : SUCCESS in 38m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d9defb74b2e4d59af39d78b1e0113f1 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"22405c82959ac6cc133e1f68e1fbf508f98f8ad6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:59:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4210b961f6c9d2fa44e2f2591e53505a6ee7e8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:00:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1878d55bb53c487fa26fc62feef547f8 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d82d353677394928a288ffb508412f02 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9603b3d21723fce3b7380b14d537d847088246c","revisions":{"d9603b3d21723fce3b7380b14d537d847088246c":{"kind":"REWORK","_number":1,"created":"2021-04-06 10:14:25.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/47/784847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/784847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/784847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/784847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/784847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/784847/1"}}},"commit":{"parents":[{"commit":"d463897551017f6c4329e7109649af87e1ec88a9","subject":"Fix pool ALPN compatibility with older amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d463897551017f6c4329e7109649af87e1ec88a9"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-04-06 10:14:25.000000000","tz":0},"subject":"Fix weighted round-robin about UDP and SCTP listener","message":"Fix weighted round-robin about UDP and SCTP listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work correctly with weight\nfor UDP and SCTP listeners, it should be \u0027lb_algo wrr\u0027.\n\n\u0027wrr\u0027 is superset of \u0027rr\u0027 algorithm, as round-robin is\na specical instance of the weighted round-robin scheduling,\nin which all the weights are equal. [1]\n\nAlgorithm in HAProxy is set to \u0027roundrobin\u0027, which also support\nweights, but in keepalived must be set to \u0027wrr\u0027 to work with\nthe weighted round-robin scheduling, as it is different to \u0027rr\u0027.\n\n[1] https://www.keepalived.org/doc/scheduling_algorithms.html\n\nStory 2008462\nTask 41491\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n(cherry picked from commit 5352a10f62f1a4763aeee81e2e638a9896b6ce79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9603b3d21723fce3b7380b14d537d847088246c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9603b3d21723fce3b7380b14d537d847088246c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
