)]}'
{"id":"openstack%2Foctavia~784848","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ic63929d8864e5285baf70dd85e6362988bf2863f","project":"openstack/octavia","branch":"stable/victoria","topic":"task/41491-stable/wallaby-stable/victoria","hashtags":[],"change_id":"Ic63929d8864e5285baf70dd85e6362988bf2863f","subject":"Fix weighted round-robin about UDP listener","status":"MERGED","created":"2021-04-06 10:15:09.000000000","updated":"2021-04-13 14:08:03.000000000","submitted":"2021-04-13 14:05:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"784848-task/41491-stable/wallaby-stable/victoria","cherry_pick_of_change":784847,"cherry_pick_of_patch_set":1,"meta_rev_id":"d795233a92052e7eebc28e7b39cc47606543eede","_number":784848,"virtual_id_number":784848,"owner":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 14:05:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-04-13 11:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-04-09 12:04:56.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-04-08 16:19:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-04-13 11:51:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 11:19:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-04-06 12:56:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 16:05:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-13 11:51:32.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"8dd2ecaed433b37b59b88da06c068af3eab92836","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 10:15:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* octavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n* octavia/tests/unit/common/sample_configs/sample_configs_combined.py","accounts_in_message":[],"_revision_number":1},{"id":"8a1f9d8e31c9a1918178c4461a31d1a2efcd00c2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 10:25:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7a12810cceb0b751c355e5f777c87d1359378e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 10:42:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21c3524a0bd364ac5af6c91f1207eb98be32ea3b","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 10:43:26.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"9ffd77b5c933d9fd343c88c728bb1669ef6485ad","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-06 11:19:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThanks for also working on the backports! We may as well backport this as far back as stable/train; I can help you if you don\u0027t have the resources.\nJust a small but important point on the release note that should be addressed to not mislead users into believing SCTP is supported by the amphora provider in the Victoria release.","accounts_in_message":[],"_revision_number":3},{"id":"e607c6924ad0e32dcf1ae75441e7ee4f96601579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:56:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cb9a5bd944647cd8c8a98d8b49f12c9 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a84240b7c114656b76f761e8b513291 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbb67bc315a34843886671beef21d1ed : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d19d1027f12d4e3781436d5d5fbc9dba : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ab4dbcf70f4fffb2c947200fba3d46 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33bde34d41964fa490e98ece0e57fcf4 : SUCCESS in 5m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4e7d1e30fccf4ef686b94cb52fc99b16 : SUCCESS in 6m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2e8a37098e7f44a782044b61b1191c68 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/50787eb90cf64ef096298020369dc00a : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1daba9726ee842e8a36dbda327fe96cb : SUCCESS in 9m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80918f9cd1cb4406a9d9ee3f7b2f5d05 : SUCCESS in 1h 14m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8527175f60c4964b32838954692625b : SUCCESS in 1h 50m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c22c3860500437b9fae577df1a37f95 : SUCCESS in 43m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0466665fae08450d9313a693f5ede42c : SUCCESS in 41m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81336acdc87346ca97077bfd545502a0 : SUCCESS in 43m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aeda7766ce4f4d538215c2dc6350aaab : FAILURE in 36m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1543d7496e4440d6a6d7b448306f13f1 : SUCCESS in 1h 53m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ffc2c145da9a4c688add269239891bb9 : FAILURE in 2h 05m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ffd957358e7a4a719cbae5421c2c09a0 : SUCCESS in 1h 40m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/560233837e6849c6af2fc9f014f71aa4 : FAILURE in 1h 56m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/de62731f3a9948709b8e282fb7191a84 : SUCCESS in 1h 48m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b0ab5356c331424698de4af0df0002e9 : FAILURE in 1h 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"56b10363bafbcc0dc004b42b63582609ca306a82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 13:07:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"53250c79b9e1e33c74db8eaaf9d02d08fa01204a","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 13:20:14.000000000","message":"Patch Set 4:\n\n(1 comment)\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Thanks for also working on the backports! We may as well backport this as far back as stable/train; I can help you if you don\u0027t have the resources.\n\u003e Just a small but important point on the release note that should be addressed to not mislead users into believing SCTP is supported by the amphora provider in the Victoria release.\n\nThanks, Carlos. I\u0027ll try to backport this fix from Wallaby to Train.\nAlso, I add a note about the conflicts and the change ID in commit message, removed SCTP in this patch.","accounts_in_message":[],"_revision_number":4},{"id":"142b56222b74f3af5abe3ab8de7ff36481e97d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 15:22:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0db4769b1ab447d5855235a31ba9ec11 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbcbfde104c64bdeaa65afc4f77d8b96 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb7685f510af457b9303fd357577e8eb : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a711f80f5fce4ef59b493f7708fb9bf4 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f991f15fba496b8f3f5c9dddd12602 : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91dabf5ed0994adfa7bfa7180e2936cd : SUCCESS in 6m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d9d9f40812a14b46b7ac3f7fd0f4ef91 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/128d7cea8afb4c21856c519645da0182 : SUCCESS in 11m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/95ac8f58b9034cf5b6ddacb9528a0a6d : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0459e1f69ad548578dca258670edb867 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09175d97c35a4ed6999a430937ca3f3e : SUCCESS in 1h 33m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2d2717caf9c4f8da66ff97a002323c1 : POST_FAILURE in 1h 49m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33ed44b13826487888ec746772e32d9a : SUCCESS in 40m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1c87e9fc515648908dede80a30f31b43 : SUCCESS in 37m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f3ae5bbf1a0427ab98387f6205252a2 : SUCCESS in 1h 00m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98a62af128da4c62940911fb3f33901a : FAILURE in 31m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7e59787de3394678aae0e3c789814866 : FAILURE in 1h 53m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81aed1dcc38a466099ac64be82ae8987 : FAILURE in 1h 47m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/51fe1eafa3524cda8732d22f6f458b8e : SUCCESS in 1h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6984220a01a84b3bb0a7db2ec78e8c53 : SUCCESS in 1h 40m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/34f08aadb27b4e1bb3c15e2ddbd5c44c : FAILURE in 1h 45m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d20de0b7468642f4a366046f7c7d3e2f : SUCCESS in 2h 08m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d559ffc27e4f269b0a501b47a2fb2cc62aa099ae","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-06 15:39:39.000000000","message":"Patch Set 4:\n\ndib failed:\nerror: invalid command \u0027bdist_wheel\u0027\n\nand without image, testing failed","accounts_in_message":[],"_revision_number":4},{"id":"52da26fe51b666f07a5343255716274497388950","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-07 16:05:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nThe code looks good, and the grenade failure might be due to an issue that was fixed late yesterday.","accounts_in_message":[],"_revision_number":4},{"id":"303b769f71305307e9ab8c1dae5a6202714d9d3d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-07 16:07:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9f01c1a6ce1957bb5f1f02ea8149d75c0f23530","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-08 10:11:59.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"41ae4e216056505bf3d5ad944407f95dbaca0fda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-04-08 10:17:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2ec7f0f99a253538ac4bb0fa8137435732f64a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:56:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39259e1d9ed44ca99bd48eeb63a33a7a : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c7570e7b9d244218d3a3f4900f298aa : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8729f338685648aaa0ebb790f7f62ce7 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c57067e00c6549ce81d0b06015b33ed7 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41e2a4fabffd4f959043f7dcb46d9222 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92331f04b73542468caa62b7c2e0a6aa : SUCCESS in 6m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2d04871b41c04e2c9f4383bb8fe22dc7 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/280eda3af47d4a678b044c027fd2bc07 : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ed7ba60537f546518182100a456a2bf0 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9ff9785824074a43883c2dc7ab572aef : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a99674bd4ac46afa08476949bb9483b : SUCCESS in 1h 18m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ec01317a43b48ae8d77ea70d7c3aaf5 : SUCCESS in 1h 39m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6dfe2a19db742dd998727e2051d0ac0 : SUCCESS in 45m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/81a3eb0c97244aba81478bc66e1a7e0b : SUCCESS in 33m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa7128bb4dd34615bb4226660f0bb930 : SUCCESS in 38m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e07abf645a124cdba8a5a855d3ac5c1e : SUCCESS in 47m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/18201904b16f46918328f23a6a049cd6 : SUCCESS in 1h 30m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/314ad55571a9469fa5f9067728b89103 : FAILURE in 2h 28m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/05cd2908fe8044b89038beaa5a2fbc09 : SUCCESS in 1h 57m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/47cb0cb11915411f8007ca26ff8b200c : SUCCESS in 2h 00m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/efeda97822c94b8289f7b989794ee444 : SUCCESS in 1h 50m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bc94f6723c4945f4827985cfad8b05cd : SUCCESS in 1h 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a1701914baee31d405aa35213addfbbb65ac2493","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-08 16:19:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6d7a7cf6304ede119b6c848a7d4424342468aeac","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-09 12:04:56.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"d85ffcc95f3713f64a44db9142d466919dc70716","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-13 11:51:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"07d2642b81611e578dc042ccba1febeeeea40f07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 11:52:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b0e36441371d60269a8a43cb31a23376e4efbff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 14:05:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8905aec7f3a94291b07834398d1ddaee : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5335499bc0d443469c7b99eb06fbf3c5 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f17feb89f014d238cde2d5e15099511 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b040acabea04ac883ff0207bc6a3bd5 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/706e79814a9e4e1bac1f5aa36ac13860 : SUCCESS in 5m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6be2516260f542e2aa96239a64e516e4 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4244bc3a15f4935a48d2e306d1dd5f6 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2e5f11ee3634c80b5a13e7582b0ff7f : SUCCESS in 1h 21m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdcc5f779986459fba765a90b7730914 : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8507cc248b9441e4b46419f57ce7051b : SUCCESS in 47m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5edfed367e1d4688a535046f1782a832 : SUCCESS in 43m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35eb3833c0064fda823f9f49d03c5768 : SUCCESS in 43m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/803979694b174c74ab2694faba777a94 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":5},{"id":"3795afa28056d70d9581b5242bdf90934712f834","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 14:05:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d795233a92052e7eebc28e7b39cc47606543eede","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 14:08:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c81e0413ef4c599f262a4efe187221 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c588cfeff574179b7cdf917644d839e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c7640b90ad549566e23a97fbcf2f2ad387bed7d4","revisions":{"76582df6a3d8418e5ca8cf415cd99a7f496536c0":{"kind":"REWORK","_number":1,"created":"2021-04-06 10:15:09.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/48/784848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/784848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/784848/1"}}},"commit":{"parents":[{"commit":"8b210bb52a6c44b3bd69acdf2202d1625e763fb6","subject":"Fix incorrect ERROR status with IPv6 UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b210bb52a6c44b3bd69acdf2202d1625e763fb6"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-04-06 10:15:09.000000000","tz":0},"subject":"Fix weighted round-robin about UDP and SCTP listener","message":"Fix weighted round-robin about UDP and SCTP listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work correctly with weight\nfor UDP and SCTP listeners, it should be \u0027lb_algo wrr\u0027.\n\n\u0027wrr\u0027 is superset of \u0027rr\u0027 algorithm, as round-robin is\na specical instance of the weighted round-robin scheduling,\nin which all the weights are equal. [1]\n\nAlgorithm in HAProxy is set to \u0027roundrobin\u0027, which also support\nweights, but in keepalived must be set to \u0027wrr\u0027 to work with\nthe weighted round-robin scheduling, as it is different to \u0027rr\u0027.\n\n[1] https://www.keepalived.org/doc/scheduling_algorithms.html\n\nStory 2008462\nTask 41491\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n(cherry picked from commit 5352a10f62f1a4763aeee81e2e638a9896b6ce79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76582df6a3d8418e5ca8cf415cd99a7f496536c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76582df6a3d8418e5ca8cf415cd99a7f496536c0"}]},"branch":"refs/heads/stable/victoria"},"b263287209ab6889aedc9251eede0b464351a43f":{"kind":"REWORK","_number":2,"created":"2021-04-06 10:25:24.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/48/784848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/784848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/784848/2"}}},"commit":{"parents":[{"commit":"8b210bb52a6c44b3bd69acdf2202d1625e763fb6","subject":"Fix incorrect ERROR status with IPv6 UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b210bb52a6c44b3bd69acdf2202d1625e763fb6"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-04-06 10:24:51.000000000","tz":480},"subject":"Fix weighted round-robin about UDP and SCTP listener","message":"Fix weighted round-robin about UDP and SCTP listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work correctly with weight\nfor UDP and SCTP listeners, it should be \u0027lb_algo wrr\u0027.\n\n\u0027wrr\u0027 is superset of \u0027rr\u0027 algorithm, as round-robin is\na specical instance of the weighted round-robin scheduling,\nin which all the weights are equal. [1]\n\nAlgorithm in HAProxy is set to \u0027roundrobin\u0027, which also support\nweights, but in keepalived must be set to \u0027wrr\u0027 to work with\nthe weighted round-robin scheduling, as it is different to \u0027rr\u0027.\n\n[1] https://www.keepalived.org/doc/scheduling_algorithms.html\n\nStory 2008462\nTask 41491\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n(cherry picked from commit 5352a10f62f1a4763aeee81e2e638a9896b6ce79)\n(cherry picked from commit d9603b3d21723fce3b7380b14d537d847088246c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b263287209ab6889aedc9251eede0b464351a43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b263287209ab6889aedc9251eede0b464351a43f"}]},"branch":"refs/heads/stable/victoria"},"481937eceb9f1ee04a7c577c75347256149ea59a":{"kind":"REWORK","_number":3,"created":"2021-04-06 10:42:54.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/48/784848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/784848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/784848/3"}}},"commit":{"parents":[{"commit":"8b210bb52a6c44b3bd69acdf2202d1625e763fb6","subject":"Fix incorrect ERROR status with IPv6 UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b210bb52a6c44b3bd69acdf2202d1625e763fb6"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-04-06 10:42:38.000000000","tz":480},"subject":"Fix weighted round-robin about UDP and SCTP listener","message":"Fix weighted round-robin about UDP and SCTP listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work correctly with weight\nfor UDP and SCTP listeners, it should be \u0027lb_algo wrr\u0027.\n\n\u0027wrr\u0027 is superset of \u0027rr\u0027 algorithm, as round-robin is\na specical instance of the weighted round-robin scheduling,\nin which all the weights are equal. [1]\n\nAlgorithm in HAProxy is set to \u0027roundrobin\u0027, which also support\nweights, but in keepalived must be set to \u0027wrr\u0027 to work with\nthe weighted round-robin scheduling, as it is different to \u0027rr\u0027.\n\n[1] https://www.keepalived.org/doc/scheduling_algorithms.html\n\nStory 2008462\nTask 41491\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n(cherry picked from commit 5352a10f62f1a4763aeee81e2e638a9896b6ce79)\n(cherry picked from commit d9603b3d21723fce3b7380b14d537d847088246c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/481937eceb9f1ee04a7c577c75347256149ea59a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/481937eceb9f1ee04a7c577c75347256149ea59a"}]},"branch":"refs/heads/stable/victoria"},"cc2fcad795578211c48a53fe531440c9d53823a5":{"kind":"REWORK","_number":4,"created":"2021-04-06 13:07:52.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/48/784848/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/784848/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/784848/4"}}},"commit":{"parents":[{"commit":"8b210bb52a6c44b3bd69acdf2202d1625e763fb6","subject":"Fix incorrect ERROR status with IPv6 UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b210bb52a6c44b3bd69acdf2202d1625e763fb6"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-04-06 13:03:35.000000000","tz":480},"subject":"Fix weighted round-robin about UDP listener","message":"Fix weighted round-robin about UDP listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work correctly with weight\nfor UDP listeners, it should be \u0027lb_algo wrr\u0027.\n\n\u0027wrr\u0027 is superset of \u0027rr\u0027 algorithm, as round-robin is\na specical instance of the weighted round-robin scheduling,\nin which all the weights are equal. [1]\n\nAlgorithm in HAProxy is set to \u0027roundrobin\u0027, which also support\nweights, but in keepalived must be set to \u0027wrr\u0027 to work with\nthe weighted round-robin scheduling, as it is different to \u0027rr\u0027.\n\n[1] https://www.keepalived.org/doc/scheduling_algorithms.html\n\nStory 2008462\nTask 41491\n\nNOTE(Xing Zhang): The first file does not exist in stable/victoria\ndue to unit test structure was fixed, file renamed from test_lvs_jinja_cfg.py,\n(patch I6d84047b3481a2bf6bf9bd17d482fb504dbc752b) was introduced in stable/wallaby.\nThe second file has conflict due to the SCTP is supported by the amphora provider\nin the Wallaby release. (patch I30997ae6cc6b8ec724f0e9dcfdfe49356b320ff4)\n\nThis backport removes SCTP in commit message and releasenote due to SCTP\nwas introduced in the Wallaby release.\n\nConflicts:\n\toctavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n\toctavia/tests/unit/common/sample_configs/sample_configs_combined.py\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n(cherry picked from commit 5352a10f62f1a4763aeee81e2e638a9896b6ce79)\n(cherry picked from commit d9603b3d21723fce3b7380b14d537d847088246c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc2fcad795578211c48a53fe531440c9d53823a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc2fcad795578211c48a53fe531440c9d53823a5"}]},"branch":"refs/heads/stable/victoria"},"c7640b90ad549566e23a97fbcf2f2ad387bed7d4":{"kind":"REWORK","_number":5,"created":"2021-04-08 10:17:54.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/48/784848/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/784848/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/784848/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/784848/5"}}},"commit":{"parents":[{"commit":"8b210bb52a6c44b3bd69acdf2202d1625e763fb6","subject":"Fix incorrect ERROR status with IPv6 UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b210bb52a6c44b3bd69acdf2202d1625e763fb6"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-02 04:55:35.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-04-08 10:15:58.000000000","tz":480},"subject":"Fix weighted round-robin about UDP listener","message":"Fix weighted round-robin about UDP listener\n\n\u0027lb_algo rr\u0027 in keepalived won\u0027t work correctly with weight\nfor UDP listeners, it should be \u0027lb_algo wrr\u0027.\n\n\u0027wrr\u0027 is superset of \u0027rr\u0027 algorithm, as round-robin is\na specical instance of the weighted round-robin scheduling,\nin which all the weights are equal. [1]\n\nAlgorithm in HAProxy is set to \u0027roundrobin\u0027, which also support\nweights, but in keepalived must be set to \u0027wrr\u0027 to work with\nthe weighted round-robin scheduling, as it is different to \u0027rr\u0027.\n\n[1] https://www.keepalived.org/doc/scheduling_algorithms.html\n\nStory 2008462\nTask 41491\n\nConflicts:\n\toctavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n\toctavia/tests/unit/common/sample_configs/sample_configs_combined.py\n\nNOTE(Xing Zhang): The first file does not exist in stable/victoria\ndue to unit test structure was fixed, file renamed from test_lvs_jinja_cfg.py,\n(patch I6d84047b3481a2bf6bf9bd17d482fb504dbc752b) was introduced in stable/wallaby.\nThe second file has conflict due to the SCTP is supported by the amphora provider\nin the Wallaby release. (patch I30997ae6cc6b8ec724f0e9dcfdfe49356b320ff4)\n\nThis backport removes SCTP in commit message and releasenote due to SCTP\nwas introduced in the Wallaby release.\n\nChange-Id: Ic63929d8864e5285baf70dd85e6362988bf2863f\n(cherry picked from commit 5352a10f62f1a4763aeee81e2e638a9896b6ce79)\n(cherry picked from commit d9603b3d21723fce3b7380b14d537d847088246c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7640b90ad549566e23a97fbcf2f2ad387bed7d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7640b90ad549566e23a97fbcf2f2ad387bed7d4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
