)]}'
{"id":"openstack%2Foctavia~785567","triplet_id":"openstack%2Foctavia~master~I6f200561af19af072c92a61ac2e92442d93239a9","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6f200561af19af072c92a61ac2e92442d93239a9","subject":"DNM testing octavia-lib","status":"ABANDONED","created":"2021-04-09 07:35:10.000000000","updated":"2021-06-22 09:53:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a9ccd91b1019145f0760e37ba6f2db055d651ae","_number":785567,"virtual_id_number":785567,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-09 07:55:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 07:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f701570f05ad737d9c92f0f5de7106cf1c1de79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-09 07:35:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3851150e2e21ef1d5a175068dd4ebd3639c6f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 07:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a819ef2c7f4f40b89afbfcf8bac38d15 : SUCCESS in 11m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3822120190254fd49157937eefa95634 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afc183836744cdc93916e816e479ce7 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/951884cf72da490bb8fa9f0a02f83d74 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aef8f5001ece424b85c487e605a446ea : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ce60f7c24d341f8bc18df535622c94d : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7fb1366afa54dde9987596586135dee : SUCCESS in 11m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b805a9ce72204b4cb809ec912f1b0604 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a496f8d328c644a5aae9105a62f477a4 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3e97b4b1229f42d89e0271a779a66352 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80a75fa4a0754b399f41d4dec281c6f2 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7a9ccd91b1019145f0760e37ba6f2db055d651ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-22 09:53:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e548ba0b21b8ae8d0e16ede1bb4b5b8740fa0e1","revisions":{"9e548ba0b21b8ae8d0e16ede1bb4b5b8740fa0e1":{"kind":"REWORK","_number":1,"created":"2021-04-09 07:35:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/785567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/785567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/785567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/785567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/785567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/785567/1"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 07:34:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 07:34:51.000000000","tz":120},"subject":"DNM testing octavia-lib","message":"DNM testing octavia-lib\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/785559\nChange-Id: I6f200561af19af072c92a61ac2e92442d93239a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e548ba0b21b8ae8d0e16ede1bb4b5b8740fa0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e548ba0b21b8ae8d0e16ede1bb4b5b8740fa0e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
