)]}'
{"id":"openstack%2Foctavia~787190","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Idb3ad20b8539bfdb788981a8634317257d83b238","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Idb3ad20b8539bfdb788981a8634317257d83b238","subject":"Fix empty Batch Member Update to unlock objects","status":"MERGED","created":"2021-04-20 11:58:00.000000000","updated":"2021-06-17 22:43:13.000000000","submitted":"2021-06-17 22:40:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787190","meta_rev_id":"271ff43504b24b782b61342d28ba4dab3ac967b3","_number":787190,"virtual_id_number":787190,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 22:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-17 22:43:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-04-29 12:37:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-17 20:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-17 20:38:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 11:58:00.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-04-20 14:23:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 09:19:24.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-04-26 14:43:21.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2021-04-26 14:52:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-29 09:42:12.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2021-06-17 20:38:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5765856c0552d945858e05ae403a9a8e8a2e128a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-20 11:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"183aaad2cc7494ca263bd5bfd6f1a1227535ee20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 14:23:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebf3ab2a2b9c459081d878264e920f08 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1eee4807804d0fac8f15714f08cb8c : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef1e8be2b95844e68793bde66611e431 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a1e4d5a5d314afab2a961e7045b4e8b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f32ebc192d414d469d90f3efcceac1e6 : SUCCESS in 15m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f93c4467d464c94b72c9f48b4b95a87 : SUCCESS in 5m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5aa5a809dbdf4371a1921f94f3f0fdad : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/817857f04ae2469da4c1c169aa071af2 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c09468baafb4d5eaa637086bd1ca94d : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/515ecf70206846eab8431aa4b0e735ea : SUCCESS in 1h 30m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92f1bd5a717740c391c1c7aa8f7be479 : SUCCESS in 2h 03m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a1e9583d240403394e74938abf41850 : SUCCESS in 48m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/39ba12603ce945428defd40b7ac9604b : SUCCESS in 49m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cf6c80406694f57bea5feace887cf94 : FAILURE in 51m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c67b075b5c5b4a1090f47d642986be40 : SUCCESS in 1h 10m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e9f18ff17eb48a2a80773a9f0951161 : SUCCESS in 2h 12m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4baa92e4648648c2bd33a7f0fae5eeeb : FAILURE in 53m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0b238d7b136f4db29936e6e8b0fe437a : SUCCESS in 2h 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d3b82178ad71533ca3262317165c027fbd5e5f1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-20 14:33:39.000000000","message":"Patch Set 1:\n\nrecheck\n\nsome random dib mkfs error? passed all other scenario jobs","accounts_in_message":[],"_revision_number":1},{"id":"07a9e4328c85e6f22c4c4061d87ca5673ec7b6d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 16:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d0d01dc08d1429e8d653aa19c94bb46 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77c6e6d02dd4d7882fb9e86290eabb5 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3117d98ef36f44f783f4c9a942f74919 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82b8bf1be959468ab974a9072e2f328a : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3805c5f7ecfe4cb09e1f29175a07188d : SUCCESS in 16m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d0af21833b64134a93e28d1b780b0d6 : SUCCESS in 5m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d985cfbae3604988a5784046af95f041 : SUCCESS in 7m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/63eba5640fc14c1ea46f1631dfd4969d : SUCCESS in 9m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68f5b7db7ac74f94b269dedb38f5c966 : SUCCESS in 8m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/360ca8f51009439490d6f490c6ccfbdc : SUCCESS in 1h 21m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63151f9bc16a41a4921f979c50f76614 : SUCCESS in 1h 51m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec8c19a3f6f0419e8d6c38440444854c : SUCCESS in 36m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a1abe38c36524db08ef0127de00e067a : SUCCESS in 37m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94e90f47222b453b9c7882ca36dd19b3 : SUCCESS in 45m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fdc2a6b5868d4ae9904f56782d45690a : SUCCESS in 1h 34m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e4fbf22873ab4b138ab1cba3678b097a : SUCCESS in 2h 03m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a0426964f184f849476cec93e189fff : FAILURE in 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c4d74aa27f6e46da8e832048188ce17d : SUCCESS in 1h 56m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcd2e6625620886be832d8a8d500fe2921bc5fae","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-21 09:19:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de212cf47b0de83543e53397b27a357ed2aedc00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-04-26 14:43:21.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8f3057ecefc9789e6e81119a14c44810c78c73fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-26 14:52:37.000000000","message":"Patch Set 2: Code-Review-1\n\nThe stable/victoria change needs to be proposed as well, I only see this change in master and stable/wallaby.","accounts_in_message":[],"_revision_number":2},{"id":"8a8e8a4d9d047530c160e7ed7e7fb65002141d93","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-26 14:56:02.000000000","message":"Patch Set 2:\n\nDone: https://review.opendev.org/q/Idb3ad20b8539bfdb788981a8634317257d83b238","accounts_in_message":[],"_revision_number":2},{"id":"091112eca21d1f04b58e5cdd7aa4a65628f3115e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-26 15:00:30.000000000","message":"Patch Set 2: -Code-Review\n\nI now see a victoria change...","accounts_in_message":[],"_revision_number":2},{"id":"71749d573bafabb7827fc158e69977a09db08fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:48:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a379e668ec5f4d8fa7b219d237a209a4 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f140d0a117c4fd485ced40a4567ac40 : FAILURE in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4adf4274e77a40acb148cb961a7692a7 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/279492a7d241449f9d1d3879edc555b1 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16b5fc693674f00a1cef7cf42605ccc : SUCCESS in 14m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a05f8ec0b86648709eda71da84c6c875 : SUCCESS in 6m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b6852659f054bca9c33564876480f75 : SUCCESS in 7m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eda9aebdda964bbd9637e0fc45903bcd : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00bf0361042b403a8d6b9036b23889a2 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3ed705bbeb041379582ee0eca35986e : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8dfbe9cd460d4cd6b531124489fb04d1 : SUCCESS in 1h 40m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ceacee5532d4cfd9e48f477f3f94715 : SUCCESS in 38m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f92c7b5bcf7e4c55ae7f418d5df1c262 : SUCCESS in 41m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20a4e1561b904829810cf168096c542f : SUCCESS in 37m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69f1f17efb494350ae2e48e79aa37b40 : SUCCESS in 1h 08m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cfc2a795bbdf436bb25cf7960ff9115c : SUCCESS in 1h 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3b87bdf31524d25906d061681015a12 : FAILURE in 52m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b0c91ebccc0f48dfb15dcdb12d811001 : SUCCESS in 2h 02m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_updater.py","accounts_in_message":[],"_revision_number":2},{"id":"f1f21251467ce4fffad671eba6c639e84fe570b7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 09:42:12.000000000","message":"Patch Set 2:\n\nrecheck\n\npep8 fix has been merged","accounts_in_message":[],"_revision_number":2},{"id":"9f8a3e50244b3d019abf462acb2330d003130124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 11:49:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/554944d3212e431bb18520804caa70d6 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b4b5365b8f43a9bbf02bbfbe6e1866 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/060152dcdcf84ad4af89d53bab8751c2 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cf677fbf93a428695b80fd4ec10f74e : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d20dace545a4a398809f1d9f23be0d1 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f3bdaaa2819414a8b5d8ef3212c9579 : SUCCESS in 6m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e36db1e5f217439e88a5658c4aa89407 : SUCCESS in 6m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8537f09398454f9aae84909abbe297ca : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6829ae6e68884e008af028795232d496 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29d187bb711b49a9bb0707c379505b66 : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a720d560d454c659c7781a47ae9f43d : SUCCESS in 1h 44m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b7cc84718f84701aca91516b8f9dc3c : SUCCESS in 38m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7344fa9e21324a75b7eef3826b5e2304 : SUCCESS in 33m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84a7a06711e348dc8945daa5fb3f0639 : SUCCESS in 36m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bedd64a78f5f4d38a6bd7eadb7aa4184 : SUCCESS in 1h 06m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9e39ccd1713f4872997af234ffc63020 : SUCCESS in 2h 00m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2f24dd167f34d1fbfc97f6210575cc3 : FAILURE in 45m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/18aa8946e5a34e3a8a06487342540081 : SUCCESS in 1h 58m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d2a61a39e47e8f37520cf0052c91624c4770832","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-29 12:37:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"10579a93684c153baeb48e114e900da5d56f3924","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-17 20:38:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"ac554ee5fd667a473cb721beb14b7ea20db73c77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:39:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d47bedf8bcf434f97bbc7b14f5aca35d923ac034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:40:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7fa00532c3b47839fc8562358c766ff : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80cefe177b4643eaa02621596a5bc5db : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6e66660c82243889ca428eed2cbb7f9 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2342061853624851a6c0a94d5942e8ad : SUCCESS in 14m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cd87b0b265444ebad0ba0b5ee3274f2 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/adcf4cdc1f3448c290012ad417169d66 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f0f956e3dca415782ddde2f7e97442d : SUCCESS in 1h 22m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08bdcd5b7cab42e8971ae933234d2f39 : SUCCESS in 1h 48m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0119877acbf548a98478d86888d7b7c7 : SUCCESS in 54m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2dff1a7b9e104b748d509bff15c75d91 : SUCCESS in 34m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d0eba72089a47319bcc4b8290bd2d1c : SUCCESS in 35m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b73f4c0b31bb4d059296a56e917fb7ce : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":2},{"id":"cf42cb5d539f50043b43ac39bdab43b95c5abaf3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:40:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"51369aa075d574fa4dd41a62e620bed27326f6cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:43:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1210bae4af7845ec9549e83417cbe6d4 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69c25c1f18934b70974729d2ad12e1d5 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7367311cadb958f1806abbdcb1abb5f848d2b486","revisions":{"1b1cdfa3a935a0b5fb74b6d833c3b3d55964adaf":{"kind":"REWORK","_number":1,"created":"2021-04-20 11:58:00.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/90/787190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/787190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/787190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/787190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/787190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/787190/1"}}},"commit":{"parents":[{"commit":"a8266a9e4f89c0c2120994e3d6b98c1ffdac4877","subject":"Fix weighted round-robin about UDP listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8266a9e4f89c0c2120994e3d6b98c1ffdac4877"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-03-23 00:14:53.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-04-20 11:57:31.000000000","tz":120},"subject":"Fix empty Batch Member Update to unlock objects","message":"Fix empty Batch Member Update to unlock objects\n\nMove the decision up one layer to the API controller.\nThe amphora driver should now just do as it is told.\n\nChange-Id: Idb3ad20b8539bfdb788981a8634317257d83b238\nStory: 2008731\nTask: 42083\n(cherry picked from commit 00e3f8349d353b1912a994b26926de4e0ed88258)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b1cdfa3a935a0b5fb74b6d833c3b3d55964adaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b1cdfa3a935a0b5fb74b6d833c3b3d55964adaf"}]},"branch":"refs/heads/stable/ussuri"},"7367311cadb958f1806abbdcb1abb5f848d2b486":{"kind":"NO_CHANGE","_number":2,"created":"2021-04-26 14:43:21.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/90/787190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/787190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/787190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/787190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/787190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/787190/2"}}},"commit":{"parents":[{"commit":"a8266a9e4f89c0c2120994e3d6b98c1ffdac4877","subject":"Fix weighted round-robin about UDP listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8266a9e4f89c0c2120994e3d6b98c1ffdac4877"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-03-23 00:14:53.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-04-26 14:43:20.000000000","tz":-240},"subject":"Fix empty Batch Member Update to unlock objects","message":"Fix empty Batch Member Update to unlock objects\n\nMove the decision up one layer to the API controller.\nThe amphora driver should now just do as it is told.\n\nChange-Id: Idb3ad20b8539bfdb788981a8634317257d83b238\nStory: 2008731\nTask: 42083\n(cherry picked from commit 00e3f8349d353b1912a994b26926de4e0ed88258)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7367311cadb958f1806abbdcb1abb5f848d2b486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7367311cadb958f1806abbdcb1abb5f848d2b486"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
