)]}'
{"id":"openstack%2Foctavia~787286","triplet_id":"openstack%2Foctavia~master~Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9","subject":"Add Octavia performance guide and 2vcpu flavor","status":"NEW","created":"2021-04-20 19:40:59.000000000","updated":"2021-07-29 22:17:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"f10243e1efb05d2ea1900dfbb05e67bc3b436f9d","_number":787286,"virtual_id_number":787286,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-23 23:43:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2021-05-06 09:03:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-1,"date":"2021-04-23 21:42:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 20:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 16:35:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-04-23 20:07:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-28 08:21:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2021-05-06 09:03:47.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-07-29 22:17:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"227c37cd22bff739bc900a18f400a1cee9e35153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-20 19:40:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce1adec51424cdd2bb33cf67eafc38ae4dde40dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-20 19:41:16.000000000","message":"Patch Set 1: Workflow-1\n\nWork in progress","accounts_in_message":[],"_revision_number":1},{"id":"36778cfa9ea310133e07e67d9718a5dee055ebc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 20:28:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b889ad1d43c14505b7722abfb6a97cec : SUCCESS in 13m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80053cb263e34a3a9b9d100fbd0e97ed : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60bcab56062e49228120a0ad1d97b6ee : FAILURE in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fdb294422c14072919ae6bb169d15e9 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47fbbd6f543e40dcb9ed989829660b52 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f999668bdd764945bc14c981620d9abd : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d434ed8001480389f1a02defd7f7c1 : SUCCESS in 11m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/20df6dacddfc4610ab768aa89af446fd : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49556a02a6614bdf807e9996be0f1af7 : FAILURE in 11m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/47585cfd32674114bcdd3bf8a1f7661a : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f4e82116686448eb1f0a60dcaf7d09d : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f04cf7cddd9d40b7aa3a24b287d89e35 : FAILURE in 22m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a13ea9e1adaf445da722c28c93733b2c : FAILURE in 24m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cad7bb0835a145f3beb926eef913b48b : FAILURE in 36m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/94c7c62fc29c4c06b761744d1cf33ae9 : FAILURE in 24m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6672c762a5945e4ab3c504d94b07c47 : FAILURE in 32m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e38b1b45e8de468e99f8c3b1f8baba43 : FAILURE in 42m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/69a3e13d149b41e69e3f98f10eaae8b4 : FAILURE in 29m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/11c468104c104f38b2c57b5315c67c49 : FAILURE in 24m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/af97fab4bc8c4714a219a4f4522a8982 : FAILURE in 24m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6ef7a36ce754458898ff3f85b93be17d : FAILURE in 24m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bb8cee4cbc934d71be3e21f4b2929740 : FAILURE in 44m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1604b3c1c7954aaf8d85f453cdbe1c00 : FAILURE in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4ab24f5f8ad04a3a261892ac8e0b31d7a3d7588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-21 21:01:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b3481bb392f677821f9eb286647ff478e40dcb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-21 22:03:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d081e8966ed0a44ecbc2a08aed909d716b2dc314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 22:56:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0559d81798aa4b009797b4be8523992f : SUCCESS in 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f807d4e3a69f486da60553da0b389055 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30de2c679704a25b24fa62d997706c7 : FAILURE in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d3805b97b3b4ef7bbbc124cd371c95a : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ffa4ea5a6dc4b70b0dcc16cb7992463 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7103a482e2b485dbb29a5eeef7302d4 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cd71691d6444cdaa5ea15f76fedc53 : SUCCESS in 11m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cc6d19f9c3c743678d1e888b4e12bf49 : SUCCESS in 7m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e95523263e164ba699986af8dc10851d : SUCCESS in 10m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/738789227e3e432c86e794860c362dcd : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da521f5b93134164b92c7ad6da9bcf09 : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/712c7f6349d549ef8f62ad456c8c5f53 : FAILURE in 43m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/726d662ce78649179efac4c463058773 : FAILURE in 31m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e024b08e260149158a31d966f82c52bd : FAILURE in 26m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c91dc55b34564b52804635a0c8f4f640 : FAILURE in 30m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31e30ac7af414aa0868e8ef01b0dbe12 : FAILURE in 29m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eac68f242e3c42519ab221df07b268e2 : FAILURE in 51m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6071035461a04600a25b16ea6f356dfd : FAILURE in 27m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a1c5748ec29d4da681e1dccaf00f9413 : FAILURE in 28m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d7e760970c144487854f124872ff58eb : FAILURE in 26m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c0ac3603d6124826bd224d2f9a411128 : FAILURE in 26m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/61fec498ee5148778a4b477e036b927c : FAILURE in 26m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/06896e3062c54d49a09044d7522cf296 : FAILURE in 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-2vcpu-amphora https://zuul.opendev.org/t/openstack/build/3e0cbd1da6344861a4b499312037a3b7 : FAILURE in 39m 35s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":3},{"id":"9874520010f86938797a20c762350232d1133199","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-22 16:35:31.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nLooks like all the failures are related at least","accounts_in_message":[],"_revision_number":3},{"id":"cfffa0bcda43aea90e449420703c784a35ccafa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-22 19:46:29.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d287a8d78492224a3534d9146a45558c2ea6cb69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-22 19:50:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9150d2fb55f69c218c033a84523e52a171922185","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-22 21:33:21.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"0fd3fad03ca7739a7f2e6ad50949dc5b1ce23a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-22 21:33:29.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b407649878d212fcef92177199fbfa21d7450001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 23:27:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5be643a6d0d84500a7b2287c9812385d : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fac72b9ce374f818e3bab4f7a1be262 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c833b7e1a884fe3809e0dc7f59eb004 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/258aba7e853c4fcd9e36099cb9da5cf0 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3fc2b4cbfeb4fb0ad11324980fe7025 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18af16b60a8d450fb75b1fb71f04a3e6 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a9a2d7f04d4aa7a755d568699741cc : SUCCESS in 10m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf4015119504486a9ad0222aaa4d8fe0 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d83b400d983f40b0b8e9788b8b53fc43 : SUCCESS in 9m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7b1e737d31e343779aab9038ccfa6fcc : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b5d3da70da646e2bedc5f69acb17e16 : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/32ac67002af048afbb474507b16053bc : SUCCESS in 1h 25m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3eeb734a64f4952a35673355c98278a : SUCCESS in 1h 51m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92da36040c804a87a46a1a36f7e6aa5d : SUCCESS in 33m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ed447bcdbf24462957b06d5c6ce36a3 : SUCCESS in 33m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b8e6e2505e3498a8f31c1716cf9ad9c : SUCCESS in 31m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/359277696a6343b5b43c69e4377fceba : SUCCESS in 52m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/501285f46fd242fbac3daa46d641640e : SUCCESS in 1h 24m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1175080858c141039ffe7c832fd6ba6e : FAILURE in 49m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5537e45eb9ec4cffafb00ce2aaeb10f9 : SUCCESS in 1h 46m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f11914a20f6a450091ea50942ca7aa21 : SUCCESS in 1h 38m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/05ad5b294b72490193c2ea44b94d5107 : SUCCESS in 1h 35m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5bf11b478c64481eb2f84d9427d0d34f : FAILURE in 25m 32s (non-voting)\n- octavia-v2-dsvm-scenario-2vcpu-amphora https://zuul.opendev.org/t/openstack/build/9a3b7e769fb24983a51ad58c11a0fb22 : FAILURE in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2b0af2daa01b383843edf8f13ed5462a06996327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 00:06:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5586650f88c17a4d95006c522841157f677fb421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 00:52:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bccdfe05fd9320f1c9d41d4cf66de1f33ebbf83b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 02:51:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3842f3b2799041b6960e828e8ca60ecb : SUCCESS in 11m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11782a22d20f456aab3557d3bef20235 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58d81db9f7a4e089dd50573aec6969a : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bac7cc754af48158f98b9901e9e06bc : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/508e8317520b4436a1046cda8591e468 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35292a6be8dc4dc0aeb5cd4417b5107f : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a79f9dad15d4f1795f2ce9f2d76c454 : SUCCESS in 11m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eed95a39dfdb430fa6c20c5215144e0b : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9e40a80f0e894dfe82ec02fa2bd0e267 : SUCCESS in 8m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/27193eff733640eaac294e3cf5c36a3b : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8120f31534843c788a7d8e4974a61d4 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a056e51d6c244bd9a90091629ed52d4a : SUCCESS in 1h 29m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2026e2115ee642de98fa3ac922fc9235 : SUCCESS in 1h 13m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78cb2252c31a45d2bf2832a48ba5832d : SUCCESS in 57m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9794d53472be4818bf2e0bd82f585ec2 : SUCCESS in 42m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2576f50f30f4aaa948d07e51858dca7 : SUCCESS in 56m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/514ab35371b6455a987f53bff0bb2f6d : SUCCESS in 52m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a97961e060c04b688c8ca049e4c0512e : SUCCESS in 1h 55m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/292dad19b804482ba22aa62dfa1f362f : FAILURE in 1h 12m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d0a223fdf7847819b78ccc8d98ecb1f : SUCCESS in 1h 28m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e0753246cccb4d289212d8814efaac79 : SUCCESS in 1h 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7f0d6da92e5646f688173058384fe954 : SUCCESS in 1h 44m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/56bc5a0665e94fb9bbe213f4adbdfc6d : FAILURE in 32m 06s (non-voting)\n- octavia-v2-dsvm-scenario-2vcpu-amphora https://zuul.opendev.org/t/openstack/build/c42e3cfba2b64945845b2dffcb21270a : FAILURE in 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3396a3ee610a9ddf771cd4394deb727d8f7da749","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-23 20:07:36.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)\n\nCan\u0027t tell if the failure is due to devstack/settings","accounts_in_message":[],"_revision_number":9},{"id":"beea39ab2a8d3a7bb042aed85e67a497292d4cf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:24:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"234a47722b098d3e09297169a08182f27d4cf946","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:25:35.000000000","message":"Patch Set 10: Workflow-1\n\nJust in case you were not in the PTG session. This is a stub patch for a new contributor to share their finds. It is WIP until that is complete.","accounts_in_message":[],"_revision_number":10},{"id":"0ed2aa36769d5083e5d890fb172e0c287e042639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 21:42:23.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5259e73ea59d62c3ae681efb4f6e5552bb404d7b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 21:42:45.000000000","message":"Patch Set 11: Workflow-1\n\nJust in case you were not in the PTG session. This is a stub patch for a new contributor to share their finds. It is WIP until that is complete.","accounts_in_message":[],"_revision_number":11},{"id":"05de853752f08a2642333c0d5f40492569dcd7b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 23:43:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5eff7b18b06c4236bb87ca0b606c384b : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09aae6a3194e4ea6adb614e5da1b53b8 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0feee7af74e43b68edb410a062117c3 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c03867e26644645aafbf871d9de2314 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50e20ade5ff4b1faa35beb545bf0a1d : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cab596fc23c4dd38ef9be89e252edf4 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89852731c444dd89e95a3f6f6a19acb : SUCCESS in 14m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fc38fa35dd474f419f006020ffd20fa3 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0b36dcdaf13b49218df84a902774a3d3 : SUCCESS in 11m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c33e28a65ee44fed9c641601e122f3a8 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/167a59cf6e144c038e25b420cf933a64 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45a3986a9eb14b2894bf2cc079351ce0 : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c0c05ece6114594811dbff95d53f3e4 : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe660324514b4ffe9d6bc1b894fd97d8 : SUCCESS in 40m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/75b2f716ebd74bb1ba550c8b146ffe27 : SUCCESS in 33m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c81b91aa8b44d0bb9baa7b9985fa5e7 : SUCCESS in 41m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5849a428587441cdb10bcb1a794f2293 : SUCCESS in 48m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d3da3982ff91434293c1435037c15d41 : SUCCESS in 1h 48m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3c6861e6534477d8a570c919fb98c14 : FAILURE in 49m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3cd2bac06f4648cdb00fc72b33986a8b : SUCCESS in 1h 58m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5eba60fc02074827a7326fbb237c6ba2 : FAILURE in 1h 25m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/703416b1ab1442e8a8f976c9a5b432e3 : FAILURE in 1h 54m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/146b1cca1bc445419c85305d24863d8e : FAILURE in 7m 39s (non-voting)\n- octavia-v2-dsvm-scenario-2vcpu-amphora https://zuul.opendev.org/t/openstack/build/a76697a0b2464e10a7095180f1bcb29c : SUCCESS in 1h 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"66e754e9403d8d91b59931041cb44395c8ead98e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-28 08:21:41.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0f8da86149ce801ae0f6d0e898db61d2ef45bfb2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-28 16:18:02.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"89f230395014add9b5b8564e8cd7466d2c841e5d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-06 09:03:47.000000000","message":"Patch Set 11: Code-Review+1\n\n(2 comments)\n\nI\u0027m happy with the answers to my comments","accounts_in_message":[],"_revision_number":11},{"id":"f10243e1efb05d2ea1900dfbb05e67bc3b436f9d","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:17:53.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4614672f255ca57097c5b109d72bbeae0ce42d76","revisions":{"ea8e7dd45ac8936d0cf620b4815596be9dc4537d":{"kind":"REWORK","_number":1,"created":"2021-04-20 19:40:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/1"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:40:56.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea8e7dd45ac8936d0cf620b4815596be9dc4537d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea8e7dd45ac8936d0cf620b4815596be9dc4537d"}]},"branch":"refs/heads/master"},"af995fecfe4187e662f8251106e10cfe39036bb1":{"kind":"REWORK","_number":2,"created":"2021-04-21 21:01:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/2"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-21 21:00:57.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af995fecfe4187e662f8251106e10cfe39036bb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af995fecfe4187e662f8251106e10cfe39036bb1"}]},"branch":"refs/heads/master"},"3f92c4808ea348ec13c8aaed8370393dfb41b7e2":{"kind":"REWORK","_number":3,"created":"2021-04-21 22:03:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/3"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-21 22:03:07.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f92c4808ea348ec13c8aaed8370393dfb41b7e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f92c4808ea348ec13c8aaed8370393dfb41b7e2"}]},"branch":"refs/heads/master"},"f927a150b88df87e08962ce49627f46b3756aae9":{"kind":"REWORK","_number":4,"created":"2021-04-22 19:46:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/4"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-22 19:46:24.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f927a150b88df87e08962ce49627f46b3756aae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f927a150b88df87e08962ce49627f46b3756aae9"}]},"branch":"refs/heads/master"},"2f857696baa7b9d90430a4913a0981bc3f09a1a4":{"kind":"REWORK","_number":5,"created":"2021-04-22 19:50:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/5"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-22 19:50:02.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f857696baa7b9d90430a4913a0981bc3f09a1a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f857696baa7b9d90430a4913a0981bc3f09a1a4"}]},"branch":"refs/heads/master"},"6d60f05f3447ab591394542c3f548a02beaad0e1":{"kind":"REWORK","_number":6,"created":"2021-04-22 21:33:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/6"}}},"commit":{"parents":[{"commit":"bd01ad94dcea3eca7636f1286ceed0753b563a28","subject":"Merge \"Fix weighted round-robin about UDP and SCTP listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd01ad94dcea3eca7636f1286ceed0753b563a28"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-22 21:32:55.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d60f05f3447ab591394542c3f548a02beaad0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d60f05f3447ab591394542c3f548a02beaad0e1"}]},"branch":"refs/heads/master"},"8c11dfbd74bc9038f37a567af9b73bef03afefb0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-04-22 21:33:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/7"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-22 21:33:29.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c11dfbd74bc9038f37a567af9b73bef03afefb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c11dfbd74bc9038f37a567af9b73bef03afefb0"}]},"branch":"refs/heads/master","description":"Rebase"},"9253da1af7239c83a0b028b08338a34b09874fe4":{"kind":"REWORK","_number":8,"created":"2021-04-23 00:06:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/8"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-23 00:06:07.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9253da1af7239c83a0b028b08338a34b09874fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9253da1af7239c83a0b028b08338a34b09874fe4"}]},"branch":"refs/heads/master"},"09dcf12427669c357c00e13c2319157394deeb24":{"kind":"REWORK","_number":9,"created":"2021-04-23 00:52:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/9"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-23 00:52:44.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09dcf12427669c357c00e13c2319157394deeb24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09dcf12427669c357c00e13c2319157394deeb24"}]},"branch":"refs/heads/master"},"b7d451e41c8b33403025e1b3a230b994dea2fb14":{"kind":"REWORK","_number":10,"created":"2021-04-23 20:24:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/10"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-23 20:24:43.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7d451e41c8b33403025e1b3a230b994dea2fb14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7d451e41c8b33403025e1b3a230b994dea2fb14"}]},"branch":"refs/heads/master"},"4614672f255ca57097c5b109d72bbeae0ce42d76":{"kind":"REWORK","_number":11,"created":"2021-04-23 21:42:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/86/787286/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/787286/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/787286/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/787286/11"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-20 19:38:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-23 21:42:16.000000000","tz":0},"subject":"Add Octavia performance guide and 2vcpu flavor","message":"Add Octavia performance guide and 2vcpu flavor\n\nThis patch adds a performance tuning guide to the Octavia\nAdministrative documentation and adds an Octavia flavor for two\nvCPU amphora as an example.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/787449\nChange-Id: Ia51f97fc5c0e67072a50ea5ee57673112e8a39a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4614672f255ca57097c5b109d72bbeae0ce42d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4614672f255ca57097c5b109d72bbeae0ce42d76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
