)]}'
{"id":"openstack%2Foctavia~787528","triplet_id":"openstack%2Foctavia~master~I6251cd17bd6fa9faf5109e50c2190dda3614908d","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6251cd17bd6fa9faf5109e50c2190dda3614908d","subject":"Update nr_open limit value in the amphora","status":"MERGED","created":"2021-04-22 10:35:06.000000000","updated":"2021-08-19 17:51:49.000000000","submitted":"2021-08-19 17:49:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787528","meta_rev_id":"431f803955883ddba116761784b2ac81e81bca72","_number":787528,"virtual_id_number":787528,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 17:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-08-19 17:51:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-08-19 17:51:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-08-19 15:36:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 12:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 16:12:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-23 20:57:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-07-29 22:18:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"},{"updated":"2021-08-18 14:21:21.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"a7f47b84cda801bf0360a42f41870966cea11a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-22 10:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caf47f1b1e25e11e6958a37e40b22b83382ebeed","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-22 10:36:18.000000000","message":"Patch Set 1: Workflow-1\n\nWIP I need to dig into the octavia-amphora logs to check if this patch fixes all the possible issues","accounts_in_message":[],"_revision_number":1},{"id":"eb6cb75d0d8739afc808cca8ce78f4d618694c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 12:45:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ecec6ecb76f46f5958667a51466e2ab : SUCCESS in 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70601242c42745a79119171908976773 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b4b7d699518434687a975e5e4712c95 : FAILURE in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50a9c925b4a443608ab5d45eba9ff32b : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dbc225f961f438c9d380d07688ac090 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c585700a09c24215a1cb6bcdfc643a40 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1969e073ab6f4b4b8c886345aaa5e0f2 : SUCCESS in 14m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11971b5acff54e488112b39ff47dea84 : SUCCESS in 5m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8efe0cd4f9d446c3bb3b39eff166a41a : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7251e20d161d412e945cd2a4f94948c5 : SUCCESS in 9m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aed9a736f72041099c3b5dd7b5146a60 : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7d3daa9da6c434bb41690dcbd05c517 : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f84c899d7d294cea8d45ca8c7c937b12 : SUCCESS in 1h 14m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a79c7736ce2d45fb80084f8abb31f7c0 : SUCCESS in 1h 46m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/638aa46342274ff2a4a2a0190a5a66f7 : SUCCESS in 47m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/58ffb7102e394a5f9104c1df29a73f77 : SUCCESS in 48m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1fdf59a941449ccbc042ad272588012 : SUCCESS in 31m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c47ebcc7a7cb4c7f9ceb8c747b70b35e : SUCCESS in 1h 07m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/775212cbd09c45de9bcc1807bcf5714c : SUCCESS in 1h 45m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d56fcd0efdbd4f679436244424d5e180 : FAILURE in 46m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e7235e7d9d8d41f7b202b5f50d23f02e : SUCCESS in 1h 38m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19670340c9304438a5adae08e3d72487 : SUCCESS in 1h 35m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ed0c0995716e46568aea5f866d42dd71 : SUCCESS in 2h 02m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d32991c4bb9b45ef988b46252b952b84 : FAILURE in 28m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"faf08d94279ba87a9603152ab2c36514008c941f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-23 15:07:20.000000000","message":"Patch Set 1: -Workflow\n\nit looks good, I didn\u0027t see any \"FD limit \u0027 warnings in the logs","accounts_in_message":[],"_revision_number":1},{"id":"7bd95c5ed89ec0a1b086fb6d69dae6f0f47ca397","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-23 15:10:16.000000000","message":"Patch Set 1:\n\nrecheck\n\npep8 fix was merged","accounts_in_message":[],"_revision_number":1},{"id":"7006459558d5fab6abeb3e96f59178d324d9f0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 19:15:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b12587617bd4803ba324e5ae76957a0 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e6d20f54995465187bdc75d08360791 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50f0fb74d4c34754ba0bb35168db62e0 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f9a9c212ce54b91a93d5aa3b7c55fb0 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf889cff1a304b44b015fab1a48960ec : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3301ffb8d3874b55b9d8ada8a22ea022 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a6ee6d91884297bcce242b798092c8 : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c78a5e1a14e406fba0c3aa635087370 : SUCCESS in 5m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/88feda42d0284e4e8d083d3fec7491e8 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7cc9f112d7e445a5a8e3038861a80275 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b737f7904454047922c461958f9ec98 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85d0b57e2fa84bbea854abeb90a54ab9 : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a136fe4cb0f54c7692656e9ba1197022 : SUCCESS in 1h 21m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7790cc0a526b48dab901673cbbe6488b : SUCCESS in 1h 31m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/969f52794c7e421aa4e7bcd7a2c3d612 : SUCCESS in 37m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a334163c2a2641ca9ef2da7f9158a174 : SUCCESS in 50m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81f7abc951304c0dad5b26409264b10e : SUCCESS in 37m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd49d15337ca4ebb9799180e4bbecd6b : SUCCESS in 1h 06m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de2fa34ef2e441fcbdd41f503d1e601c : SUCCESS in 1h 14m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/928883f1ce2348d9922ebb94e656a1a1 : FAILURE in 57m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/491085462d694e4da3f723935e6dd8f6 : SUCCESS in 1h 31m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/02185fce48ab41d887ba79009b085ef2 : SUCCESS in 1h 23m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f175cd9f4844430484cfd2122016b7f2 : SUCCESS in 1h 25m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/80267a7ada50483b9b935474ef1b120a : FAILURE in 7m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb6a9790ba97484021e3c27feaea4483cd4a84ea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:57:53.000000000","message":"Patch Set 1: Code-Review+2\n\nThis looks good to me, but I found a similar patch open:\nhttps://review.opendev.org/c/openstack/octavia/+/764448\n\nWe should align these two.","accounts_in_message":[],"_revision_number":1},{"id":"b9e6118913afc282c90c31d7435c723ba35d4a64","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:18:03.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":1},{"id":"fc5a622c611a639c4856f700ec5516b9f41412f9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-08-18 14:21:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05ea69b064dd4f23628c4e0c22aa977d3b747a22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-08-18 14:21:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"273f84f73c8386ba748c21dba7ec50c4dee64e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 16:36:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed7bbbda7a774c94b1302bda4b7de5d5 : SUCCESS in 12m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed68684953db487e83c034fc15f63a2e : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f39cb581a4242ab91426d70342a4036 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ff1a91f64b545c3b631eefea948d516 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61e9635b2df34f508e7d6fc93292b287 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9e95b0efeb044f395abc78882eebeeb : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e50627cac24180b451b516d094b9ad : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a73ddc4c0fa9479db137af9db5704cb4 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/60244ab4d5b04bd9a04d36e2020b5e7b : SUCCESS in 6m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bffb220b289f4f35b0bcdaa1db8f0c3a : SUCCESS in 9m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/79d24eb8240748999e6b35126b1be44a : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8eac1adf06d54e1497c9e3b88a53efcd : SUCCESS in 9m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a573e2db035a4b1984619de168887386 : SUCCESS in 1h 45m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a81e5374038148648c30034027349445 : SUCCESS in 1h 44m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/696445030e0f4e069e86a7bd57709d73 : SUCCESS in 43m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbf13a002634428886be1834068f83d2 : SUCCESS in 53m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0105df8fd7694a35bc9a047cee3a492a : SUCCESS in 50m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7e3e6e3414f347bf9e0c1ab274f6c33e : SUCCESS in 1h 41m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd71dde8e9d1408c92212d1d87ffc712 : SUCCESS in 2h 11m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/42d9954d1f044c12998c4f0a8778c5df : SUCCESS in 1h 27m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/41c7637e9d8848febac2e3089d6236ef : SUCCESS in 1h 49m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a17a217557d94764ba40a1aca5fa07db : SUCCESS in 1h 30m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/21c2bd1085c04ea6ac6f9fb18ab494b1 : FAILURE in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e10cdc7358b8ba203cae6f4ad23557a021a0a953","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-19 15:36:04.000000000","message":"Patch Set 2: Workflow+1\n\nLGTM, this one has a release note","accounts_in_message":[],"_revision_number":2},{"id":"a587a79b9bd1d7fa38987c0567e800c3770667d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:36:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2f3b54e732189fe9e0601a3ef64e712b8c4abdd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 17:49:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33395d179e664153ada6ee242bf1da0f : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f0bb2ed9124e6a844a019b652f3d84 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87adc758a1e64232b8196a8b3ece8083 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a16087e8e7424c96b1c0fbdb0f2ded5c : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7b4ec4f46f4cd8a6dfc26c74673f25 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/517f2d48a3144e4495e2f3fed19494b3 : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/41e6b59bb0ba43a5b3e518309fba667b : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c746b915caa4c7cb7337f2212370716 : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a92fb410eea40959737923356c3dc3d : SUCCESS in 2h 07m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68dade57604044b6bed64ef4fba1cf71 : SUCCESS in 2h 00m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8ec8014bbc44894b96635f642052df5 : SUCCESS in 35m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ecd895d8eef4858abb51ef1470b126b : SUCCESS in 37m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c0b346199ca465b83a1218ce463aeef : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":2},{"id":"891f75916de9e05541a8ddfecd0dd25c3a500c1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 17:49:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a4dd92a23eef7d67d18239df0681a775316730d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 17:51:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de7b09f384cc494da8e9e27c10c3fdd0 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e92f52ae74654e558e862888f893e69f : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4174f4a5a47a208d7cb7fa6d736f5bb20e995b64","revisions":{"f4f6356d9bd9b1a7fb314159bd7f360d3b89c135":{"kind":"REWORK","_number":1,"created":"2021-04-22 10:35:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/28/787528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/787528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/787528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/787528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/787528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/787528/1"}}},"commit":{"parents":[{"commit":"de352cac203b8e545adf1318ce53b99dc0bd404b","subject":"Edits for \"Operator Maintenance Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de352cac203b8e545adf1318ce53b99dc0bd404b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 08:58:08.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 10:34:26.000000000","tz":120},"subject":"Update nr_open limit value in the amphora","message":"Update nr_open limit value in the amphora\n\nUsing haproxy 2.x, the ideal rlimit value for nr_open is close to\nconnection_limit * 2.5 (see compute_ideal_* in src/haproxy.c).\nSet this limit to 2,600,000 in the amphora to support a loadbalancer\nwith maxconn 1M.\n\nThis prevents the following warning messages when launching/reloading\nhaproxy:\n\n* \"Cannot raise FD limit to 2375058, limit is 2097152.\"\n* \"FD limit (2097152) too low for maxconn\u003d950000/maxsock\u003d2375058. Please\n  raise \u0027ulimit-n\u0027 to 2375058 or more to avoid any trouble.\"\n\nChange-Id: I6251cd17bd6fa9faf5109e50c2190dda3614908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4f6356d9bd9b1a7fb314159bd7f360d3b89c135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4f6356d9bd9b1a7fb314159bd7f360d3b89c135"}]},"branch":"refs/heads/master"},"4174f4a5a47a208d7cb7fa6d736f5bb20e995b64":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-18 14:21:24.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/28/787528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/787528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/787528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/787528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/787528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/787528/2"}}},"commit":{"parents":[{"commit":"b89c929c12fb262f59ba320a37f2a5bf4109df98","subject":"Add caps for pip-extra-reqs/pip-missing-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b89c929c12fb262f59ba320a37f2a5bf4109df98"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 08:58:08.000000000","tz":120},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-08-18 14:21:24.000000000","tz":0},"subject":"Update nr_open limit value in the amphora","message":"Update nr_open limit value in the amphora\n\nUsing haproxy 2.x, the ideal rlimit value for nr_open is close to\nconnection_limit * 2.5 (see compute_ideal_* in src/haproxy.c).\nSet this limit to 2,600,000 in the amphora to support a loadbalancer\nwith maxconn 1M.\n\nThis prevents the following warning messages when launching/reloading\nhaproxy:\n\n* \"Cannot raise FD limit to 2375058, limit is 2097152.\"\n* \"FD limit (2097152) too low for maxconn\u003d950000/maxsock\u003d2375058. Please\n  raise \u0027ulimit-n\u0027 to 2375058 or more to avoid any trouble.\"\n\nChange-Id: I6251cd17bd6fa9faf5109e50c2190dda3614908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4174f4a5a47a208d7cb7fa6d736f5bb20e995b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4174f4a5a47a208d7cb7fa6d736f5bb20e995b64"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
