)]}'
{"id":"openstack%2Foctavia~787667","triplet_id":"openstack%2Foctavia~master~Idd9021c8a73d1be87f5df40d3f27c0fba9067e3c","project":"openstack/octavia","branch":"master","topic":"amphora-failover","hashtags":[],"change_id":"Idd9021c8a73d1be87f5df40d3f27c0fba9067e3c","subject":"Fix amphora failover flow in amphorav2 driver","status":"MERGED","created":"2021-04-23 06:45:47.000000000","updated":"2021-04-29 17:52:24.000000000","submitted":"2021-04-29 17:50:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787667-amphora-failover","meta_rev_id":"24f07aac293b635c8adec882b5cfc848a4ee8d21","_number":787667,"virtual_id_number":787667,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 17:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-04-23 12:55:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-04-23 19:25:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-04-29 13:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-04-29 13:58:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-04-29 13:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-23 09:02:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 10:17:33.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"CC"},{"updated":"2021-04-23 12:55:56.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-04-23 19:25:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-29 13:58:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9ab9d216b1f94bf285b25de6a3951077c7f42820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-23 06:45:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ac75a72d266da5d4813cc458a3f317635ab9ebb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-23 06:46:50.000000000","message":"Patch Set 1:\n\nIt looks like a partial revert of I5ed6288b2776bd7f1c9b67e9cfd9a8f05b1196be","accounts_in_message":[],"_revision_number":1},{"id":"7f63e9fe443979f2330cd15ec8c4187bcbd1377b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:02:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0fe3c4ece584442b47b5789e84180e9 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd17d81a338e4abf869e72000d30bd42 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa32b2d44cb4ba4a60a8fec90d4b92a : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27169929bfd4416fa63b02d10117f09d : SUCCESS in 14m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfbededc5b2d44f1bf1d592f612ed80c : SUCCESS in 12m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f100d1d4caf4fdd914f4493fc13a148 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43a77fb9c6a40f78e3d701399000a8d : SUCCESS in 14m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6414ee68ea742e6aaad5c92a40f4c41 : SUCCESS in 5m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d88c2e4dd807481d9a12afb9e5304d52 : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6fe1348f0a62488c8011536ffd28cc21 : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75e3200fc5934c948cd0477e3de31643 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6c697f2a73c4487a2b2bcd9d19e4ac1 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2fe58ff2b424d1085014f091f76bd1c : SUCCESS in 1h 52m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c44463245854863b8e0aaa035a2f909 : SUCCESS in 2h 03m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eeb36e3f37484fa3bb0214dfe144f290 : SUCCESS in 45m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/654f0f5786144217968369f1a046eaca : SUCCESS in 41m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fbf2ff277194571b9b404939be037c9 : SUCCESS in 44m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eec8a51ff63a4f828a3b21863718a5c1 : SUCCESS in 51m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/63a443af93c841d2a9ea6f816676eadb : SUCCESS in 1h 21m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/79c1b7e079034ce1bf8b9c519e866a2b : FAILURE in 53m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e57afbe459c944d183935f374a47f080 : SUCCESS in 1h 25m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f9d2dbc2bfc8437fb5314e30d247b309 : SUCCESS in 1h 51m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/816e409492e840fbbcf57cc4ede66d0a : SUCCESS in 2h 00m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fda053a5f1fc4f4b804cc74fe1dac19f : FAILURE in 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9c0e07fa43c9cc7ae9d93fa4a7a15efa205711b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-23 10:17:33.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"bd5bd0857446ad801ba3393ae89f2abd93adeb0a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 12:49:50.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/d4b6ccdc5b964ef4aa429b9e01607f6e : SUCCESS in 1h 33m 13s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/554e286e57b04d62afed53acbf9e6608 : FAILURE in 1h 42m 14s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/302d796e3f1643b6acce854c061184ee : SUCCESS in 46m 43s\n- octavia-v2-dsvm-spare-pool-amphora-v2 https://zuul.opendev.org/t/openstack/build/2bda317356584719b5bdaf971137d531 : FAILURE in 40m 50s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/90ea70aa4e984b74835dca68725a8dab : SUCCESS in 45m 38s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/de25710bfb0040bf9e10a58205b7ad12 : SUCCESS in 1h 09m 58s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/8fc1c445d6a9488998b507aeccf98cbc : SUCCESS in 2h 11m 48s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/c03e1f82ce6d4d9091845e2d447ee83d : FAILURE in 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"03a9c825108d34cd64c30ce6c4e828ae55bdcacf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-23 12:55:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05513bc9a65413928b0fc1c0940f97dc8b2e0c8b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-23 13:13:28.000000000","message":"Patch Set 1:\n\n@ataraday this patch reverts some changes you did in https://review.opendev.org/c/openstack/octavia/+/760380\n\ndo you remember under which circumstances you got the \"Mapped argument \u0027amphora_id\u0027 \u003c\u003d\n\u0027amphora_id\u0027 was not produced by any accessible provider\" error?\nI\u0027m also removing spare pool support, it might be easier to catch missing attribute in the tasks.","accounts_in_message":[],"_revision_number":1},{"id":"2a068e372c642ba442cfd755c30e8704db253d40","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-23 15:38:03.000000000","message":"Patch Set 1:\n\nSome of the tests were failing and in worker logs this error was present. I\u0027m not sure if there was some special scenario for this. I checked logs and there is no such error now, so this will be fine to partially revert this.","accounts_in_message":[],"_revision_number":1},{"id":"28cbb14315d8bf7f7dc4d4cb741510621653d1ba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-23 19:25:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d6497e3704ff2b2e3fe92719ddb70f852b77d53e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-29 13:58:29.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"f5a895aa66c182fac066b6b773c794237a6bbd09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:58:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea9229f25fc327a3fe95f5fbbeeb387b64e7c0b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 17:50:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c80e330693340b9b6f19e7a906f62ab : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f63123b693470198accb1d6d8211b4 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11a911dc03842ba8c950bc0f54d00bd : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76ed02b5caf54e0b8d26177fd81851a8 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a29207d9c1e4f9894b7318ad8779329 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a060dad478964d8ba48c4f1c6095662e : SUCCESS in 6m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/60f5ef5bc0cf4a3196bdabc717c19295 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0444dae222dc4f788cecfb2e9ac6ee01 : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/813e593da7674ddfb2114423147cefcc : SUCCESS in 1h 54m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d97161d1f984d2ebee4f58fcdca85d3 : SUCCESS in 1h 51m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f259fcd790143aaadfdb24a62347327 : SUCCESS in 39m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5abaacbf489b4ef282f18fa8dc3a3352 : SUCCESS in 39m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68c110537deb47e28d27aa23633dad13 : SUCCESS in 32m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc9934009d4d4b8eb0bc88cfb3746eee : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bb9fa78170a40bedce1b5aa39360d20fb7f6026a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 17:50:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"24f07aac293b635c8adec882b5cfc848a4ee8d21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 17:52:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af6e46daa5ad444cb7d00f204ae14085 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64ef116560484fb288dca0ec2ef8bc13 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e2dab64a04db857058cdff16ae5a2432bd2f8c8","revisions":{"7e2dab64a04db857058cdff16ae5a2432bd2f8c8":{"kind":"REWORK","_number":1,"created":"2021-04-23 06:45:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/787667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/787667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/787667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/787667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/787667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/787667/1"}}},"commit":{"parents":[{"commit":"de352cac203b8e545adf1318ce53b99dc0bd404b","subject":"Edits for \"Operator Maintenance Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de352cac203b8e545adf1318ce53b99dc0bd404b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-23 05:09:09.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-23 06:26:40.000000000","tz":120},"subject":"Fix amphora failover flow in amphorav2 driver","message":"Fix amphora failover flow in amphorav2 driver\n\nRemove amphora_id from the failover flow parameters in amphorav2, it is\nunused and it hides in a later task the amphora_id of the newly created\namphora, breaking a simple amphora failover.\n\nStory 2008848\nTask 42360\n\nChange-Id: Idd9021c8a73d1be87f5df40d3f27c0fba9067e3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e2dab64a04db857058cdff16ae5a2432bd2f8c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e2dab64a04db857058cdff16ae5a2432bd2f8c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
