)]}'
{"id":"openstack%2Foctavia~787722","triplet_id":"openstack%2Foctavia~master~I830c6a4c49fa47105f788cf99a0f775e5dbdcaea","project":"openstack/octavia","branch":"master","topic":"spare-pool-removal","hashtags":[],"change_id":"I830c6a4c49fa47105f788cf99a0f775e5dbdcaea","subject":"Spare pool removal","status":"MERGED","created":"2021-04-23 12:15:54.000000000","updated":"2021-05-05 12:36:21.000000000","submitted":"2021-05-05 12:33:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"787722-spare-pool-removal","meta_rev_id":"c9713d1b0f4e3a020b94827fe359b053eb26d41e","_number":787722,"virtual_id_number":787722,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 12:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-05-05 10:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-04-29 21:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-29 13:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-05-05 10:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-23 15:13:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 19:32:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-23 21:11:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2021-04-23 21:28:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-04-25 09:07:13.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-04-29 13:52:46.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"CC"},{"updated":"2021-05-05 10:48:03.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"d223c12f1d4f40e8e32cf629ad1a1d4065c63e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-23 12:15:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a5bdaaabb39bade634e340a32285db7f05f91a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 15:13:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa8e3ddaf46d4f4fb4f04e7217dc93b0 : SUCCESS in 11m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5b09e9737994902a314ea24190ff8b8 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1553bcb75945afa17ac6d9005850c0 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccfc0d093eca42fd9688ac989e7c33d0 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33a6cb039f540e1b06cd326a2e49303 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b27f6db2d764165a8db53128e0cfe6d : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959e18bfa61c4f10a64bd5cc86c55465 : SUCCESS in 15m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71f63ae069804274a18a7728394d9a7c : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ca1ff90ff9f4423b8f7840f1aeb36b62 : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c3a0398844424ab58eba9c5048a0012d : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f9a22aa740d844778f569a30983abf86 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8b8835c02ff4d6dbc1bf5a2753e991f : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d750252b2174a1d8ec0f39b4b4cf715 : SUCCESS in 1h 23m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e169485988174589a37d3223cae93a04 : SUCCESS in 1h 17m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/94fa6d532f1f4751baf5fae07dcd274b : SUCCESS in 37m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5d3e0b324774088b349b8ad4ced43af : SUCCESS in 44m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd88b5ee5e4143f68c5713486a0c4dfa : SUCCESS in 54m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/82e4a94a20d9423ab99a45d23df14d8a : SUCCESS in 1h 30m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56b716ec98834260931e145515057d94 : FAILURE in 47m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5d46913d6d24ad98e53db962d3e11fe : SUCCESS in 1h 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/60531f9c250b4ee08a2579ccc31a728b : FAILURE in 2h 06m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9d1c4944d3454f669320f8160c6e3714 : SUCCESS in 1h 49m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8b0cc802151041dcb671377dbaa78821 : FAILURE in 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d158536b9567fd5b0353324d5b2f0f02828a4dcb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 21:11:06.000000000","message":"Patch Set 1:\n\nI would not consider removing the DB table, at least not for a long while after this patch.\n\nDoing so could cause issues with upgrade scenarios. It shouldn\u0027t hurt anything to remain.","accounts_in_message":[],"_revision_number":1},{"id":"a6a0a2132643f3507804d8943c0a52edd9143919","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 21:28:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nNice job!\nSo glad to clean out all of that conditional code around spares.\nI have one comment about the commit message, just to make it clear for folks that this feature went through the required deprecation cycle.","accounts_in_message":[],"_revision_number":1},{"id":"7524e180159f59ec73912bbd7ffaae886d5d3fe1","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-04-25 09:07:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6aca1c0a6fdef7b17cb8e41598106685cd1673dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-26 13:49:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"39eec59c719547a94633b18f54c3a082b2cf7f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-26 13:50:11.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"18d1afd72bef06529b4761631fc7f6aaabee4c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 15:46:44.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e919d6b1de704fddb430e0479674eef4 : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/880934043f4e4e2e9b00cc190479ae1b : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57857e1ae9104244be6c506d785887bd : FAILURE in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7cac1d714c48c8848921f2ecbe904c : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ec88d3e82f4a9f905f6765507ad0a4 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af5dd2215e2341f0aa221c5fb19f7824 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93dd7bee4e5943bc95858350a8fd113b : SUCCESS in 16m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf69892faa1f4b349d6aee97fb7c3314 : SUCCESS in 5m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2be60d81209e46de8855c9bce7f09a96 : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/afa5a1ecd2dd4fe7962c8882f21263e3 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5fbdb3f94f5d4c989deb23839823b371 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25fbde1596324169a5fd7d15499458f5 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e95c63211cbd443c84eb19b4f68d6fe5 : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80b0e16bbddd4af4bd0ad855e3c2ad35 : SUCCESS in 1h 33m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36d42f148bfe40d0b696a823e20d0e47 : SUCCESS in 36m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab1e5fbb42e74116820e5be98920acfe : SUCCESS in 41m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da848b02155647ddb45e702a22c22417 : SUCCESS in 52m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/36a8611f1b274070824fddce2941cdbf : FAILURE in 13m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dab01f348a29495ab76f35812c706082 : FAILURE in 45m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c878edf47b4d44179953b54efac837e7 : SUCCESS in 1h 49m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d1e99b1a14d47308e917989f0680c14 : SUCCESS in 1h 52m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/84547a58fbab4109af73cab159eb041d : SUCCESS in 1h 26m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/539a625b07e646e5808358e815354725 : FAILURE in 29m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/controller/healthmanager/health_manager.py\n  Comments left for invalid file octavia/common/base_taskflow.py","accounts_in_message":[],"_revision_number":3},{"id":"65f2762a02170d3c39243b032543a5ec6b153274","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-27 18:05:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a97681cb39dc87853a7c00137202dd18b96bad19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-28 07:15:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"00c9046e834833049a8682615fe1d34ed5b8d067","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-28 07:16:14.000000000","message":"Patch Set 4:\n\nPS4 removes a unused class from database_tasks","accounts_in_message":[],"_revision_number":4},{"id":"a82cd08705fd8d58930b1f34074490b29f9904e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 09:12:43.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5edcbbcdb6764ea6a81746da0ce56dc9 : SUCCESS in 12m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3477a7983fa44e99e115e6fbbad6d79 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e51fa65adbb47a5a90757bcf8cf9302 : FAILURE in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83d9826412314ab1af7732beec7924c7 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17ed268bfce24f029049ecb83b733dca : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51242b39ecec414a80a4f7ce5719be44 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77539d2c0de842598dfa91a537bff1f6 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cd46e172d444acd921ba22e5cb227e4 : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a26aa7b75f834e8282aa698872e0048e : SUCCESS in 6m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4eb1393f55a438990509e10cad496df : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf5e82c7b8b947dcb7a838ca9e7743ad : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de7532fe09b24ba9ac6b38bafa8df65c : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/573a971080424ab8bccb16bd75ec967c : SUCCESS in 1h 07m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9830189269b741e79048d4ff127f7fdb : SUCCESS in 1h 17m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f95408fd42094b41b810af82570b9069 : SUCCESS in 40m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd2d6d854c3c4b1ba144fd954f07fb37 : SUCCESS in 33m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c20cac93dbdd4d6cb016bda8301651dc : FAILURE in 40m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/faa874304235480eb7e98336b81008e1 : FAILURE in 10m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b6abfcf6705649f694516e1774d660ac : FAILURE in 40m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a57b8d1ac576438290b821f311eae689 : SUCCESS in 1h 35m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3116a3b193564358a81255d9e2d9513b : SUCCESS in 1h 36m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dbc4f2ec98f9438e9dde6f1185dc7c52 : SUCCESS in 1h 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0d8f3b3b39a14c888662af71089ffe73 : FAILURE in 10m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/controller/healthmanager/health_manager.py\n  Comments left for invalid file octavia/common/base_taskflow.py","accounts_in_message":[],"_revision_number":4},{"id":"c6f95aded98dccb2d20cdf216a671b8a6ac66107","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-28 12:11:54.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2d3ba160567850d1c0380f2f20b2f881a9c2cb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 14:30:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/165ff47282bd4101a5103110d6f4c0fe : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca53ea4f22bd417c8a4672add5790374 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c450a0c7a842689e96b17b1759cf62 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec7255a7c08a48b88b829841f08c6dcf : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/847303e253dc4bd4b9a42b0ce6be10dd : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d168876fd72942ba90ac476d7a3dffa7 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3274945da74028ad075af39394488d : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/267da5fe1e734d4c9c3cd833065b7571 : SUCCESS in 5m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4191fe35f3794a32bb4c13054d68a6d4 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/34bac652f4d64f448cd629bf5cfe6970 : SUCCESS in 8m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/08f1d2ebf8054e4daf5166b36f59a0d6 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d9216fd7b324eb2b3ea97f71567d66a : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b7e7117d20d6451b91c93313f1cb574b : SUCCESS in 1h 27m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24116446857a42f9892714589108c6ae : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6658ac96a5e6441982138a6a51a62f77 : SUCCESS in 39m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8963145f936640b59e0497acb2c0ba17 : SUCCESS in 40m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68dacd8b126c4eefb045c183d9f90dfa : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c87131cb585641d8ad50373e377b2ca0 : FAILURE in 14m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4751797894f94075a785558a71c56886 : FAILURE in 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e070df7b12594d0ea4f711f851a65505 : SUCCESS in 1h 58m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9a37ba19bf7a490096cbb00ca73b072d : SUCCESS in 2h 14m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9e83de7cfef84860885f258c499676f3 : SUCCESS in 1h 59m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6c920e6eea424477902de6cd3608c4d2 : FAILURE in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"440f0c82cb5cdaf42e9fd5a29bd8111c7d0e2199","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 08:51:21.000000000","message":"Patch Set 4:\n\nrecheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"1a353d1858f1a6b085b5ff6a9b0e24ef4c4a41b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 10:52:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ffc84e77dc447b5acfe12038cd32531 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/642543931a1d439bb661ae02f9e2deda : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02d9b0a8dc3d4fa380a85caec1b4b178 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cafa3f02dd5142b1b95f0c0657fd7c21 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06bc8fb91304445d90908d95d6f51d7f : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d74352349944e87941260169e9fec7b : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4a444423fa4d5784a2c03ef1990ee3 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdc6822489894b128d85b54912c2a041 : SUCCESS in 6m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7ca8f1f55b994a9aa18f9789d7166fdf : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e3efbd82f0a64011b976a7e7059c245e : SUCCESS in 7m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/beab526afa55453a8e9a78c2e82f1613 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/294e8c7ac414437897345dc3ea3b1bfa : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc7b58b93d62464b95ffa99e7cf53d1a : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6021780695da43a98f6da37deb024581 : SUCCESS in 1h 32m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e18365771064ab2b586d2a933da8c27 : SUCCESS in 34m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/580dadfd3f044897a921c445474a1a7e : SUCCESS in 33m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67021f27c71d4466bcc82ec88c2a4fbf : SUCCESS in 1h 04m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4af189f41d30436cae135a4f117016a0 : FAILURE in 1h 45m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/69f785680f6849b792287d35504887a0 : FAILURE in 49m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4e4cb4adaf0e48b883a47b53acb358a6 : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dff101695870451eb28f63af3eb941a6 : SUCCESS in 1h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0e85dbdfb2154f6cacb66a569ade9f8a : SUCCESS in 1h 36m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/32947470e23349fe8eaf93bc7549e077 : FAILURE in 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7f5c7ff8e86bc223ba743c45512ee39b3fab413b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-29 13:49:22.000000000","message":"Patch Set 4: Code-Review+2\n\nStill LGTM, thanks!","accounts_in_message":[],"_revision_number":4},{"id":"321054c836e981dd09bde5c9fd7b979fc653b747","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-29 13:52:46.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"e94dfd2e0a4b314a3e578b542f84b4a88fff34cc","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 16:03:56.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/401de047b0d34786921e5337078530e4 : SUCCESS in 1h 50m 00s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/a1eb539ab1fa4bd08704254662e6299e : FAILURE in 1h 33m 05s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/1aef5433c42a4ca0a60e9f073e72cfb6 : SUCCESS in 34m 53s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/960f00928a7f413db87ec7c5553802d1 : SUCCESS in 38m 11s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/54cfabcf08eb4bcc94e8aa93e5615372 : SUCCESS in 58m 41s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/78fc64bfbbd647fd8295fbe52b3f3baa : SUCCESS in 2h 04m 01s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/caaa87e232d84af7928e29aeabad4734 : FAILURE in 45m 36s","accounts_in_message":[],"_revision_number":4},{"id":"59fa16707e7631e79c33ea39863066c78af34ec2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-29 21:18:47.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82911e2d7513ec547202ec1287973d8a952bc349","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-05 10:48:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1605c9707ae93cd2d10a936e01a0902a330feab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 10:48:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"29e2e69b04c2bce84c9849fecd2441fc85e861a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:33:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4948711e18f24d3d9e07578e3faa4a02 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114f74554227479c9e5a9a21864767dc : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc89ff2adf43467ea84c530c324c0efd : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91d2e6fdf3824b24afc496b3bf2f8cca : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c974cc07d18a4d29abd8dd5197a0aa76 : SUCCESS in 21m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b73988f177e43ddb2dbb0fb223f5d59 : SUCCESS in 5m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/20d7d69b04244c66a5ad9f24f069a85d : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12b7116ead774622b0325b0fb0289aa4 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/887b0003bfcb4dbe96080b8c3d30001c : SUCCESS in 1h 02m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01f344fcaf9844749de2cad74bfb951b : SUCCESS in 1h 34m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1c9d7e1f845458a80f0766132b1ffa9 : SUCCESS in 38m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/578ebda76a7048cf90d04b09943ff307 : SUCCESS in 54m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b41eca88863d45b8872d3bba3d12464e : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":4},{"id":"154570796a450ab0ed6498c1c96c76b472b57224","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:33:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c9713d1b0f4e3a020b94827fe359b053eb26d41e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:36:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a1c2acdedd460b820ae043aaf104ec : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de77f04179f64e788989209d9f7c7d33 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"815a283823121d563562474eb9048673a95f6e1f","revisions":{"1dd6a0ca80e06a8657643314bb0dbf80900af0c9":{"kind":"REWORK","_number":1,"created":"2021-04-23 12:15:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/787722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/787722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/787722/1"}}},"commit":{"parents":[{"commit":"042a6ea925644099ba83e822f47a7bf52fecb5ef","subject":"Disable spare pool jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/042a6ea925644099ba83e822f47a7bf52fecb5ef"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 16:43:42.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-23 12:13:28.000000000","tz":120},"subject":"Spare pool removal","message":"Spare pool removal\n\nSpare pool feature was deprecated in Wallaby, we decided to remove it\nduring the Xena release cycle.\n\nWIP items:\n- Should we delete the spare pool table from the DB?\n\nChange-Id: I830c6a4c49fa47105f788cf99a0f775e5dbdcaea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dd6a0ca80e06a8657643314bb0dbf80900af0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dd6a0ca80e06a8657643314bb0dbf80900af0c9"}]},"branch":"refs/heads/master"},"c089e1a8a3b31ba4162eccceee1beca4cdf07f9b":{"kind":"REWORK","_number":2,"created":"2021-04-26 13:49:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/787722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/787722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/787722/2"}}},"commit":{"parents":[{"commit":"042a6ea925644099ba83e822f47a7bf52fecb5ef","subject":"Disable spare pool jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/042a6ea925644099ba83e822f47a7bf52fecb5ef"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 16:43:42.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-26 13:48:36.000000000","tz":120},"subject":"Spare pool removal","message":"Spare pool removal\n\nSpare pool feature was deprecated in Wallaby, we decided to remove it\nduring the Xena release cycle.\n\nChange-Id: I830c6a4c49fa47105f788cf99a0f775e5dbdcaea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c089e1a8a3b31ba4162eccceee1beca4cdf07f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c089e1a8a3b31ba4162eccceee1beca4cdf07f9b"}]},"branch":"refs/heads/master"},"6707f0c284b7457fac06467691d66911d69c85f5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-04-26 13:50:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/787722/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/787722/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/787722/3"}}},"commit":{"parents":[{"commit":"042a6ea925644099ba83e822f47a7bf52fecb5ef","subject":"Disable spare pool jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/042a6ea925644099ba83e822f47a7bf52fecb5ef"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 16:43:42.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-26 13:49:50.000000000","tz":120},"subject":"Spare pool removal","message":"Spare pool removal\n\nSpare pool feature was deprecated in Victoria, we decided to remove it\nduring the Xena release cycle.\n\nChange-Id: I830c6a4c49fa47105f788cf99a0f775e5dbdcaea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6707f0c284b7457fac06467691d66911d69c85f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6707f0c284b7457fac06467691d66911d69c85f5"}]},"branch":"refs/heads/master"},"815a283823121d563562474eb9048673a95f6e1f":{"kind":"REWORK","_number":4,"created":"2021-04-28 07:15:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/787722/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/787722/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/787722/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/787722/4"}}},"commit":{"parents":[{"commit":"042a6ea925644099ba83e822f47a7bf52fecb5ef","subject":"Disable spare pool jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/042a6ea925644099ba83e822f47a7bf52fecb5ef"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 16:43:42.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-28 07:10:09.000000000","tz":120},"subject":"Spare pool removal","message":"Spare pool removal\n\nSpare pool feature was deprecated in Victoria, we decided to remove it\nduring the Xena release cycle.\n\nChange-Id: I830c6a4c49fa47105f788cf99a0f775e5dbdcaea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/815a283823121d563562474eb9048673a95f6e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/815a283823121d563562474eb9048673a95f6e1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
