)]}'
{"id":"openstack%2Foctavia~788018","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Idb3ad20b8539bfdb788981a8634317257d83b238","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"Idb3ad20b8539bfdb788981a8634317257d83b238","subject":"Fix empty Batch Member Update to unlock objects","status":"MERGED","created":"2021-04-26 14:55:46.000000000","updated":"2021-05-11 13:03:32.000000000","submitted":"2021-05-11 13:01:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788018","meta_rev_id":"14b1e90bf549cc8b9a1a8f100a097256a5286796","_number":788018,"virtual_id_number":788018,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 13:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-04-29 13:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-04-29 12:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-04-29 11:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-04-29 13:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 14:55:46.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-04-26 16:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 09:42:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2021-04-29 11:53:51.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-04-29 12:36:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-29 13:53:20.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"7bd5d7cedcfcba936ce1520bfbf3716856dfe0a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-26 14:55:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19c1c0cc3abc2c77505d7e6e1f3d508b7cf3c322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:52:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d3f8be28e140d5b7aca74b4ce76212 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/686bdb2194644b5892ef276aacab4d73 : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc2298071b26436c9c361baf5740adda : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/001bcc6dfe3a407c8d580f7f9fc80149 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84810c1f53794f208624d9d132b8b7a2 : SUCCESS in 16m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84006ca496734faf892266774ed17bd3 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a814e1af67ea4842989968045e807205 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/31e14ce45d504884b4075824208b2174 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e01520649b0140558eaa1d18b045bc2f : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5880c056f6484f41a120ed5208490263 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6de830c069c84a408ded2e6bb43469ae : SUCCESS in 1h 33m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2d9df80f9bc485a95ca05e23c771a2d : SUCCESS in 1h 54m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0bed66f8a8a4839aa44541b89e0eaeb : SUCCESS in 38m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6112d4384d2d470285184cac423429d2 : SUCCESS in 29m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c449df4bd8a4a5db39d96e6265e3c83 : SUCCESS in 47m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/34b5bc61fb884eeda668c1f817c5ee74 : SUCCESS in 53m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e1ff11b7f0d408fb25da84f31ae9910 : SUCCESS in 1h 36m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6508da03f85a4bb1a214947cf334ddc3 : FAILURE in 1h 42m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/433f511aa21a463f995fd21babecb991 : SUCCESS in 1h 28m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7886ead9dae64968a4cfdb07c0a230ac : SUCCESS in 1h 49m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8b95f3cd354349b28a47930943d3dc33 : SUCCESS in 1h 34m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/00672b11c4014e2094f47d17ae27d479 : SUCCESS in 1h 36m 00s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"34a9e4c4fb098280e4cdb51355142c316ceedc95","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 09:42:05.000000000","message":"Patch Set 1:\n\nrecheck\n\npep8 fix has been merged","accounts_in_message":[],"_revision_number":1},{"id":"8ff1bac9f9fc48f796e370b8ad682bbcf27822ae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 11:53:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4ae1a32211210de5d3544b936ac62a8f256ee2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 12:13:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b31135033263452cb0e02c2aedf5e49e : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e9b022ded645c8be331128319ff27b : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be0a546b77ba45eba3f7283eb509eb2f : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df270d007a03428794d73fe612f78cf9 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb09871c5cc47fda01d0e0d2ad63916 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bca736a36f52474ba48aa9aa98fd1bac : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3384e6947406469ba1cf6b6bd6eb1beb : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ea0a6684911b4e22a23ae581c8f15267 : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6dad18798d3f48e8bd043aca05ab63fb : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3c4c39c355f432f869f533a2de45758 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8358193ca89d48c89b9ef193c67d9223 : SUCCESS in 1h 37m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13154e9623234c2ea89e8082264c1021 : SUCCESS in 1h 49m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eca5a83c64344caab1041822ab159df2 : SUCCESS in 54m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/43bb54f8ff1142938f93650655f5a2fb : SUCCESS in 33m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/372648b1a73e4ff1a784a316e1c82a18 : SUCCESS in 40m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d25db24abe194c96a7ba3860168891f6 : SUCCESS in 53m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/302df9c277454222bf77eefc50d615d2 : SUCCESS in 1h 52m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0951d55d492949d38640e115ce61780e : FAILURE in 2h 26m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fc6b9b5f5484b0fb888bf280be6a659 : SUCCESS in 1h 39m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/504d7beab59440698ef911baffa7bf86 : SUCCESS in 1h 29m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d17814bdd82342f4841aa518f6626c1f : SUCCESS in 1h 49m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/93f4764c2cca4c9687bf69e3802edaea : SUCCESS in 1h 42m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"174dbfd48de85bf88aa6ae78d5c2014e313d0083","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-29 12:36:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bad98affcb7b1a27ec887dabcb642e4078106e0e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-04-29 13:53:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcbb148a7de874d4aacf734aa4e16819241463ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:53:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"628c407bbf22ae05ba8660f1f19cc8fe24ab7578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 15:49:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7870363c56524cf9b7a3bdacc010215e : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c07176106e543fbbe303eb98472c4ee : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57e0730c45734ec688d402f4f6c0ba8d : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9532fd601b394ff4b852ecc077c9e69a : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32057fdcdb6a4516809424feb0baa4b0 : SUCCESS in 5m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0bdaff0877bd442a8df0d51532131d7e : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bfa14ec5a6e43b7923e267217a03138 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5e9113b200c41d6a4183ad400ac1937 : SUCCESS in 1h 38m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67ec77449b814e1eb084d03c5b2d61be : FAILURE in 1h 50m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/488dffa435754c338d2c4e942915daea : SUCCESS in 40m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/27860687a03a482691d3f92280038eb2 : SUCCESS in 37m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba5be02d7bf341e79c2aef5f4429cd47 : SUCCESS in 43m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aca941db27204be798b81373a7df56db : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dfb291da11b74ae05c1814795314800554a37354","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-11 09:10:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce62e3762d8a4c07a417ec29a2f9269118f63010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:23:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dca7200eda544e54b873dc08bc143f9d : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38c3a2fa855f4c46991964136033e73e : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f91641ab1e642b1886ba5ae2ddc4c41 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ab65fb911724b99b1dec633ece2b505 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9088f44ec86e4439a1e4ba80511c1035 : SUCCESS in 15m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7f144afb0c14c5b9d261791a18aef50 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/64b7e79321fa4c95840ef22c4d69b1c1 : SUCCESS in 8m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f360f0a42487442bb8c736a38882748b : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/131a034663914ac5b2fbcd7dac48c316 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b98352462144c1193eaaf718db6dfda : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48af00d166a04ac4ac322138e9053070 : SUCCESS in 1h 33m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0b4015e56c74a16a34a076775f27dc3 : SUCCESS in 2h 01m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b83083b9ff484835b9b36647f7fb6207 : SUCCESS in 44m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/20ec6aa692fe423185b26f3359ada886 : SUCCESS in 37m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b07a0a7727cd4da3ad007624ff852c63 : SUCCESS in 35m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a1947450ad044f1498b2a7ed32e350c2 : SUCCESS in 56m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ece33fb2d17b45dda031d624e18e6458 : SUCCESS in 1h 41m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c76c1ae850341fea3d43770c1439b63 : FAILURE in 1h 53m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6d6a272bdb2742398243f68f817bea71 : SUCCESS in 1h 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dcbced87e88941ada48618f32c2db1d1 : FAILURE in 1h 30m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4db9b2134b9a4c5dabc0f27996575ded : SUCCESS in 1h 40m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d1358c61b4e94dea91c8a6053a6afa36 : SUCCESS in 1h 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff7abc03e3c8a47d7714417b3a2465f3b81066d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:24:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82f6771c3a2a7509784f05da50b43a0e76b6c885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 13:01:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4632fa80c46f4d269643cf91a054061e : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/713983bd69b641429ac2443f854822a2 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08d76c569de447d19763afbf370fe820 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708da4f218f04e2187fb24d45554450a : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/182e521996554984b568a77c68b5d1b1 : SUCCESS in 5m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c3546f7d32264deca119ea1e8424736f : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/151c9a0925f44b3c9efa33e00b7eb9b3 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fec11caed3be401da5b69fa8e4f1d82d : SUCCESS in 1h 22m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e272ea15bed742ae8425f143d24d86d1 : SUCCESS in 1h 35m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/027626a1a81b40c894bbec6071f4a4ad : SUCCESS in 41m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a92cfa5df7e44d429b3f05bc44fc31a6 : SUCCESS in 47m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6214ee5df414ddfa54921dbf6d5dae3 : SUCCESS in 35m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e18e41ab247452b922b7ede993e13cc : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"730d9d60311536c9cac95e4cf12ceadf1d561b96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 13:01:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14b1e90bf549cc8b9a1a8f100a097256a5286796","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 13:03:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7ae3feb8434517bcda0558673f0811 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecb617c1ec3d42a6977888dcc7cfa8ef : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cb53e1492da132e23bd20b890648159ac962e13","revisions":{"1cb53e1492da132e23bd20b890648159ac962e13":{"kind":"REWORK","_number":1,"created":"2021-04-26 14:55:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/18/788018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/788018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/788018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/788018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/788018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/788018/1"}}},"commit":{"parents":[{"commit":"c7640b90ad549566e23a97fbcf2f2ad387bed7d4","subject":"Fix weighted round-robin about UDP listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7640b90ad549566e23a97fbcf2f2ad387bed7d4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-03-23 00:14:53.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-04-26 14:55:36.000000000","tz":120},"subject":"Fix empty Batch Member Update to unlock objects","message":"Fix empty Batch Member Update to unlock objects\n\nMove the decision up one layer to the API controller.\nThe amphora driver should now just do as it is told.\n\nChange-Id: Idb3ad20b8539bfdb788981a8634317257d83b238\nStory: 2008731\nTask: 42083\n(cherry picked from commit 00e3f8349d353b1912a994b26926de4e0ed88258)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cb53e1492da132e23bd20b890648159ac962e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cb53e1492da132e23bd20b890648159ac962e13"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
