)]}'
{"id":"openstack%2Foctavia~788124","triplet_id":"openstack%2Foctavia~master~Iebffbb041dd2d8c7418b054c865c45f0eac8459b","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Iebffbb041dd2d8c7418b054c865c45f0eac8459b","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-04-27 03:00:25.000000000","updated":"2021-04-29 13:58:19.000000000","submitted":"2021-04-29 13:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788124","meta_rev_id":"abadd8abc41cfa2be5de55bc2961ccb388044143","_number":788124,"virtual_id_number":788124,"owner":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 13:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"date":"2021-04-29 07:56:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},{"value":2,"date":"2021-04-29 08:00:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-04-28 13:06:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-29 05:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},{"value":1,"date":"2021-04-29 08:00:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"all":[{"value":1,"date":"2021-04-29 07:56:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},{"value":-2,"date":"2021-04-29 08:00:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 05:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 13:06:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-29 05:34:37.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-04-29 07:56:28.000000000","updated_by":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"reviewer":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"state":"REVIEWER"},{"updated":"2021-04-29 08:00:59.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"1a05ab61acf4260a9c33b7a12c3a6184b96744c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"date":"2021-04-27 03:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"287d4eded94e199eda6d0ed7b89ee577d398767e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 05:09:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8dcc4f65b8e145e68789630037cd3294 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba80de1e98fd481b8ae205cb9eb35f90 : SUCCESS in 10m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d44e17fb5fed424cb3e8f1c51cb47676 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e3724194154106bfe205d924ea0d4d : FAILURE in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bccca3e1b377474ca031121934831985 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/189509fa7fbc439ebf707498a1205c3b : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42380a33357b428d8736679b28ff5d71 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa19e6e69d614f0d836011f248b514d4 : SUCCESS in 12m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a95e3d55f8542e6ba18229b698a32c4 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ee4df2c9a8fb4f5db418fbac14bc2496 : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e35c3d0fd8c64bb7b59632416e19fcd4 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7614f7da542a48acaf532e5f6726b772 : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9900eccb527649a8a1049c099df2965c : SUCCESS in 1h 17m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbc1e79cd8744b98944c77cb2204fe7f : SUCCESS in 1h 19m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6cd226c517cc453292fa38f609fc10a3 : SUCCESS in 45m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9da1840c014248f78a67dcf32617daeb : SUCCESS in 36m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d5246a43b5343f8845f8864fc7b9a7f : SUCCESS in 35m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/417ae87f803546e888c1e72ed3e35fa6 : FAILURE in 10m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eb9e0e14d0624b8ba6af2cb27f52fb4f : FAILURE in 38m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/028ceb3d63d647a89b7d45159665a1c2 : SUCCESS in 1h 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e05fd2446b04cf9a7badcfa0d850a1f : SUCCESS in 1h 50m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/066c485ead18423e96b2c71156aa1f6c : SUCCESS in 1h 21m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1dbe1f87f8ce4dc9866018733d51b2c2 : FAILURE in 20m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/controller/healthmanager/health_manager.py\n  Comments left for invalid file octavia/common/base_taskflow.py","accounts_in_message":[],"_revision_number":1},{"id":"5ee10dd33f56232ff026ab0cd6f1df46f2487214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"date":"2021-04-28 00:44:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7653cc61e7abd9f8522dedbc97ce4bb8e0222426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 02:15:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5664959c23414813b98ad8da9995a772 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bb81f209b374461b4fedffdebc9d095 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa27fab5b0de4c1e97d8d98c6ba7cd5c : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e1556cf3f740efabcee538829ccb7e : FAILURE in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bd7ce0c3962476891679bbe53396d08 : POST_FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89123750ff8f49fea329b53cbb41547d : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d72974b46aae4a90a0b7dbb887a3a5a2 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d6ce30e3364320b7fb3d5624e438a0 : SUCCESS in 13m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/34cf37d133e841ba9d0002254cfbe733 : SUCCESS in 8m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/79a4206b187b44b58a4be3b72a907f3f : SUCCESS in 9m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fb049f6920bd41b2bfe96548cf885342 : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3f56075806d431e811e8d955a6d3913 : SUCCESS in 9m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4812964d0c4b4bfebdaa79efd2175337 : SUCCESS in 1h 28m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64e1972b83484b09b8bfee3013f0d3eb : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4fdcb858f71b48bba7a15dcd561d02d4 : FAILURE in 38m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0c61a0cd81e04e828794793b5113485e : SUCCESS in 33m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/969aa73f2e9c431891d883a0e4fca847 : SUCCESS in 33m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c2ae522045ef4dd2ae95b21ae0902a7f : FAILURE in 18m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f21bf60f3084b0da9f05b0695b47590 : FAILURE in 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5c812a1207484f56ac0a6739f593aebb : FAILURE in 34m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f901babf33024469a6cdf235d525ccd9 : SUCCESS in 1h 26m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ab2d387feb8e4271813555248be32b26 : SUCCESS in 1h 16m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d6a510a184f14073b22ee053c0e7a3db : FAILURE in 30m 07s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/controller/healthmanager/health_manager.py\n  Comments left for invalid file octavia/common/base_taskflow.py","accounts_in_message":[],"_revision_number":2},{"id":"48e0feec098c649b737181ab36d4c49bf80ecebb","author":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"date":"2021-04-28 03:14:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"db1082b22c04a2bdc2b5cb5b92602d2c9f5a89b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 04:54:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12131f787ce0487894272010abceb9ad : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f493baab105c472aa110ee98d8c44557 : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ae0bc78f154ecb934dfa99a5d560e2 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c5974eea554c658296acb7e14d3abc : FAILURE in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d97dcb018d642f9bc252e88cd49c47f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10f6ac4360e946a3b377bfa86953d40e : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f3634c28e3e4637a79ee01adbfb7387 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fca39b7e346436da3c39e4e8e2ea66c : SUCCESS in 11m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d0934c645bdc420fbc2c6c836372e11d : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/09b0855b483d4518ac841b3b8e32798d : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c8daf9ea90fb4741a4a0d99035bb4955 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa81c307734a4d078bb49b7e056bb029 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7adb95f51dc94f6aba49469d9fd09bc5 : SUCCESS in 1h 35m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13e032d2b6124991ad8d582aec711cea : SUCCESS in 1h 24m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd6f49c7399f41b6884f2a91bdfbabba : SUCCESS in 40m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b6769d66fb6f43c389df5bf6c1ed5364 : SUCCESS in 43m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e3a643d5c1a4df283e9c501558d77c5 : SUCCESS in 30m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/821244556c2a45a4aef9216f82a9ddb3 : FAILURE in 13m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71880366467d49a29f7f3e25f9bc3e86 : FAILURE in 55m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/758611f8e8644b12be27fbf471c75ea8 : SUCCESS in 1h 27m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/582e7bf012ad49978b72c69a106b5b0c : SUCCESS in 1h 37m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a28c20aa214a4054831cce16a803ff74 : SUCCESS in 1h 35m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/99b8bbbaf1c344bdbaa61c4f5826559f : FAILURE in 24m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/controller/healthmanager/health_manager.py\n  Comments left for invalid file octavia/common/base_taskflow.py","accounts_in_message":[],"_revision_number":2},{"id":"7308f47d449b7bceb0686fdf6aba71e29a05e48f","author":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"date":"2021-04-28 05:56:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2bf16fcc5e54c9a305ba75fcff357a5b01a6479d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 08:01:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fa802edf73a24ec3bfdae2eeda129dbe : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58a8708b1f914794a27edcf459acaee7 : SUCCESS in 10m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/871f5cbd87f9491a8798036af83cb5e9 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ac15e58ce9468d9c1c9a33dbe5e850 : FAILURE in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e15099e82934b35b91534a2673f6f44 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aba32687d574d92942dc0a3c7b543a8 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af19880ca27642dea9ac427165b668f0 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d649923eca4a415bafcfe32174076115 : SUCCESS in 11m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aa7394fb47574ab184e6c47193f5a0be : SUCCESS in 6m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/df5215c58238407faf01c02bf268e693 : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0adaf6937b7c4433b3733690ea001c30 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2d8d56ab0d04e49a47d9571f4a6292d : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fe0c129bb23d4ee19126d066111c79c1 : SUCCESS in 1h 28m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b02951f0c424358bb291fbea374709b : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81bd9975e70c41e9a3463069190fb356 : SUCCESS in 40m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7464faf4d5f44c14a463ce3a7d804018 : SUCCESS in 38m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/350a3b521fdf4ac191d2a48c83cfce23 : SUCCESS in 40m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c7d65ec2b74a49748ed4fb35e2e352fa : FAILURE in 14m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eb65b972073c448a85295d8f310fcd08 : FAILURE in 53m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0b14bb9014b948c9a1a662b2c0659496 : SUCCESS in 1h 52m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8331a49bb4f248b8aaf35266571a4280 : SUCCESS in 1h 58m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1c64c94846964996a350139c21d41d38 : SUCCESS in 2h 02m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9a5da6076aab44a689656e63c794a5f1 : FAILURE in 36m 23s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/controller/healthmanager/health_manager.py\n  Comments left for invalid file octavia/common/base_taskflow.py","accounts_in_message":[],"_revision_number":2},{"id":"c045ba1079c08c82f2dfc2aeb796d4a2ae297eca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-28 13:06:35.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a0fdbb129c29d32fade59d690ad8e775a24cf803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 15:49:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a39956e92c5f47e8b9d42d0e8a48ee0e : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ce0484aab64531ada3e1648f39ac8d : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84f62ac921504a8691e0070958cfb066 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b6d7659ec044cf8ab5b5809d699896 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e0efffb2bf54a2a8a314f9b9e6926a3 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf13626fe96f4a6cb5893df553963396 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37a8949aab964ae98dbe193de2e8b86c : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13db5f2b8f8c4ae3a3ea04e0a796a150 : SUCCESS in 12m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/930eee076cd64fcb874100a3e2297df9 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/af0682c51e2c4e9e9135e16de97a6efa : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3b93f34934444ef8bf814dec97ca7325 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f5f6a6dce5a42b894e75d4591eb2042 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97f8bca36dfd428494d255b9aa366c15 : SUCCESS in 2h 07m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68bc5ff3ef54471f8cb782983e73f73c : SUCCESS in 2h 25m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36fcd22af1c3440db47e6d54cdc7f075 : SUCCESS in 47m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a6b7625f40a94aafa6785fd2641a368b : SUCCESS in 35m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a36179b4b75042429f105d77d92804fb : SUCCESS in 31m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a29db66b2f5c4e2284d6939233ba0926 : FAILURE in 10m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7401b6b6938d4d1aa67f394a3074cb55 : FAILURE in 45m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/57d0d0452e0444019d77482d331767fc : SUCCESS in 1h 38m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fba186928bd94b7c80631f5c7a351009 : SUCCESS in 1h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ccc11e99917846488b6f15af935adac9 : SUCCESS in 1h 39m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c7bff805216748c7ad255ba03624f1cb : FAILURE in 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e86b21ad3097a0c9e9f2808954d26be50b1cd34","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 05:34:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61516e3a30993174d9f7f8b698d2487b2f7de910","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2021-04-29 07:56:28.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"a158b6737286dd8794cc2e58b5daf1f37dd10577","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2021-04-29 07:56:44.000000000","message":"Removed Code-Review+1 by Xinxin Shen \u003cshenxinxin@inspur.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"28472df8de655792dbd1e1b709c4f101f8954ccd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-29 08:00:59.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1\n\nI see no need to backport this to stable/wallaby or older stable branches.","accounts_in_message":[],"_revision_number":2},{"id":"d123d5a241838b14e906a69c7c93d34ec56a39a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 08:01:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26adf954e706fccc0b66db56c13d01ee88c4ef45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:37:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e570b40e34a4790ba3e345da9b8da5d : SUCCESS in 2m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4e49b0f6ae74503be97f2106918b30c : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e27e540d89d4f0cbbcc23c15a87d466 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/590d17b4dd994e96a6053595847b5490 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b6b99ede5064dedb96df9a55d0d22db : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76ba953284e415c9a10150bd364e990 : SUCCESS in 16m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/81f25ed195e8413995adf7caddf1946d : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad7d86e7b2024be6b0d8cb7e649100b8 : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/13119da777d44de59084749927b5d92d : SUCCESS in 1h 32m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95dfd4b898534ea494f22c86316dea5d : FAILURE in 1h 34m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7dc830256d124a5dad81b74ee9748513 : SUCCESS in 42m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3ea345abac8346ad8a632e35ef59fbe9 : SUCCESS in 38m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/217b4aa020cd4a62b224b57a79f9464b : SUCCESS in 33m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d2e2fafe124748f2bd9624e9b73def5d : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":2},{"id":"6d0c778332ebc51c2496af8bb997f4c6467d0394","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-29 09:39:19.000000000","message":"Patch Set 2:\n\nrecheck\n\ndispatching issue in test_least_connections_tcp_traffic","accounts_in_message":[],"_revision_number":2},{"id":"8376ac0f53a130aa9c68bac46735ad28d09bba73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 11:54:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2b415f2d2304e8a888e7b8d9c6da810 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bde15b8e96fd478696acb86cfd8b1c56 : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a239554902b548bbbaae18a573053c6b : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7f169afe6740b9b884c2e732bce2c3 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f6fa5ad0e704841bef10e073269ddc3 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6a50e17b27746dd80047ae469cd5997 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5b20582384040ef976b7ee4e5a61889 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f879a4044248b98b15711332accce0 : SUCCESS in 12m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/92f8fce174ea454cab721ec4f3273651 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cf8b45a028b41d7bd76bedcb69f27b3 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dfbe2855806b492b984cecec09bbb841 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be74ccf322b54faa945a34425a9fd731 : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a0c478cfdd764999a9fedf2faff65faa : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cbc35239c374f75b8b59604d526b14e : SUCCESS in 1h 33m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c498d10367f4a32a8a3f7dd9a355e90 : SUCCESS in 35m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/81c83104033043f08254e3142a3b61e4 : SUCCESS in 35m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42bd90c699d54ab6abadf89572d83492 : SUCCESS in 48m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c500596c3b8f4047baba8e3df62f519d : SUCCESS in 1h 53m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd1a5c2d2fda451f9d609f94f42dfbf6 : FAILURE in 55m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d4667e508d8349128a82b695f968f5e0 : SUCCESS in 2h 07m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5576b1323f684d3b8f20050f2a047345 : SUCCESS in 1h 42m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fbc48b685aa54e7e8fecaa59c4aad220 : SUCCESS in 1h 39m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/200cc7b8f2644295962cd7c0ff6dff67 : FAILURE in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c887cf8aac2b18479af2b22138dc76d43b7ce6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 11:54:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef35d9f3d29c68eb46634409caa6d3a9a66c11b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:56:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c9b9ddc752048ec8353d951205e9bb3 : SUCCESS in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3629369165ca4e4cb82f90419a92c312 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f7b70a1bef4647992596b50fbd433b : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3d201d713864aa4ba65524058efd59a : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29d5b31c55014e59ba28cc2a28af80fe : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb24272aa584464bdadcab632d60a10 : SUCCESS in 11m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/61316f6aaff7452f9a855b140663dd70 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af81c83c0a6249109cc9dfbb62a3fc2b : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd85d21c51db4d22a180145cb6206133 : SUCCESS in 1h 13m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69a129c1848142048f6bee0621a354b9 : SUCCESS in 1h 59m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6aedc4052994ef9ac705d94070e2b7f : SUCCESS in 34m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6a0d566236ef41b2a1c547820f8167d7 : SUCCESS in 35m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b539c5811404381bdd79c5391959ff7 : SUCCESS in 50m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9ba2d5e78764f03915da2645de5a32c : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a8662fde755c5ada2abd2e50e2013142f88650b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:56:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"abadd8abc41cfa2be5de55bc2961ccb388044143","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:58:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75a5758d9f3493cb9431381d4ee4aef : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2f49250d1cafad9fdbf7d807dfc53c7f603a98d","revisions":{"eb100a5c60cc3499888d48b344b26f17f6dd91bd":{"kind":"REWORK","_number":1,"created":"2021-04-27 03:00:25.000000000","uploader":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"ref":"refs/changes/24/788124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/788124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/788124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/788124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/788124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/788124/1"}}},"commit":{"parents":[{"commit":"725e1c531345dabe103d1a6d1f5cb15615d812a1","subject":"Merge \"Fix rsyslog configuration when disabling logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/725e1c531345dabe103d1a6d1f5cb15615d812a1"}]}],"author":{"name":"YuehuiLei","email":"leiyuehui-s@inspur.com","date":"2021-04-26 08:46:36.000000000","tz":480},"committer":{"name":"YuehuiLei","email":"leiyuehui-s@inspur.com","date":"2021-04-27 03:00:22.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nResolves warnings like the following:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\nChange-Id: Iebffbb041dd2d8c7418b054c865c45f0eac8459b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb100a5c60cc3499888d48b344b26f17f6dd91bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb100a5c60cc3499888d48b344b26f17f6dd91bd"}]},"branch":"refs/heads/master"},"b2f49250d1cafad9fdbf7d807dfc53c7f603a98d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-28 00:44:57.000000000","uploader":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"ref":"refs/changes/24/788124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/788124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/788124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/788124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/788124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/788124/2"}}},"commit":{"parents":[{"commit":"725e1c531345dabe103d1a6d1f5cb15615d812a1","subject":"Merge \"Fix rsyslog configuration when disabling logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/725e1c531345dabe103d1a6d1f5cb15615d812a1"}]}],"author":{"name":"YuehuiLei","email":"leiyuehui-s@inspur.com","date":"2021-04-26 08:46:36.000000000","tz":480},"committer":{"name":"YuehuiLei","email":"leiyuehui-s@inspur.com","date":"2021-04-28 00:44:57.000000000","tz":0},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Iebffbb041dd2d8c7418b054c865c45f0eac8459b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2f49250d1cafad9fdbf7d807dfc53c7f603a98d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2f49250d1cafad9fdbf7d807dfc53c7f603a98d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
