)]}'
{"id":"openstack%2Foctavia~788307","triplet_id":"openstack%2Foctavia~master~I5341949513e6eb814655ee8f14a01b061894fc34","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I5341949513e6eb814655ee8f14a01b061894fc34","subject":"Fix task_flow.max_workers with persistence in amphorav2","status":"MERGED","created":"2021-04-27 14:10:18.000000000","updated":"2021-04-29 15:52:29.000000000","submitted":"2021-04-29 15:49:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788307","meta_rev_id":"e288ed25ab8c043f8191f458c6de465ddead7701","_number":788307,"virtual_id_number":788307,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 15:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-04-28 16:20:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-04-28 16:17:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-04-29 13:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-04-29 13:51:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-04-28 16:20:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 15:54:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-04-27 16:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 16:20:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-04-29 13:51:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d53d6db1841cd4b724082dad52312f2a8ad9b6ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-27 14:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af4f1e1c13a42d930c7303ec7110277cec88114f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:55:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fab9923feaaa4ba49ffd3b66ebddbcbe : SUCCESS in 9m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e94f76ebfa4b769a2da480f0129ddf : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b838016669406cbf46b46af469642a : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afb8709f545c44ffbf84f891e7cbdea9 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7548288a759d48f1909f26d82915da0e : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70c94a597ba640cd9f3acd5dbf7e48f7 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a54377069b4fb4bdd94158eb279fa0 : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8470d429492a4da4ad3a07dd5e794e69 : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5ec4363daa94474bb3c05aa40bf97442 : SUCCESS in 9m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c433c05b1ac74794878bb45ae74fc9c1 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bb22526308e461986053d7541722de0 : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca8999f3a2ea42dea7e5743ed5187e6f : SUCCESS in 1h 29m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9bb282988d847eb87f9851f7bfd2bad : FAILURE in 2h 27m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8fd2cf76eb1c48248c779a77f4aa57a2 : SUCCESS in 42m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b28fdf268c55422a8c5afcfcb63d80b4 : SUCCESS in 42m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08ca23a837494578b2a09065722a523a : SUCCESS in 38m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c46f9bd4dff4cd9844573bc079933a7 : SUCCESS in 1h 11m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5dfeb994541741ce897351ec22892daf : FAILURE in 13m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b9941f66de1f427ea840b0fec2a2e086 : FAILURE in 47m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a053044c8cd049109b5d21f1abf8bb82 : SUCCESS in 1h 36m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c193751a9bf546c695a7fe037dc44db4 : SUCCESS in 1h 57m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ba5adb0221664533b47196e51de07127 : SUCCESS in 1h 50m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/052a03a6596e4d9980922bed75564bf3 : FAILURE in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb26ab50446aa4b13ba5abbe0306bd646374c7b1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-28 12:12:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5efbdb2c95d89ff497029964b42086d0bf403094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 14:20:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d797946a34104516b07a15bf19957ce2 : SUCCESS in 11m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42016f5aa12c4d548492c2b526aab776 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e609a90f4cc54d65b4a7481e7b374a65 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06022fbfee5548a4b796c0121e84be20 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d149bd45dc754a4382138ebd01f4b9f5 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6799c90161c142178668f7b8f41570cc : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c2bcd1926043f189867699dc9bc799 : SUCCESS in 12m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/05468f1c4e0243f6bec7432fe3dae18b : SUCCESS in 6m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b90eec12a9a6457c91d33919c292adc8 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f58813d530234e79901de621bfcce090 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f66c78710ad2464983367348fd45ced8 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da10a5a6bc2c464cb4e684904d8e55a8 : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f5a71816f294016ab1b917f7d6157c0 : SUCCESS in 1h 32m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8601677bd63e49f68055a1efcef898a1 : SUCCESS in 39m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/158e3c38069b46328b6ac08417b477a1 : SUCCESS in 39m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/662e8720845e4a82a1df754a1ec8e297 : SUCCESS in 42m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd459694e4c84f4daf096542eadd18a9 : SUCCESS in 1h 08m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0eaf5673193848188e72ba2ab994bb58 : FAILURE in 24m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3847fd937bfe43f598059c529ea1963c : FAILURE in 1h 12m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/940437b8f0e349b990ffd752bb0445f4 : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c57acf6517e94556b20b160276e9102a : SUCCESS in 2h 05m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c9fee3b70ffc4fe887bdbaaddf9442c8 : SUCCESS in 1h 54m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2eb4fcedbf104639b0f56aaea5a8de74 : FAILURE in 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66b7928e3216c3d77b413ff84f43481c17ef1f5f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-04-28 16:17:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8c19273d93a77d54cda8dc72df1f1806813aed58","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-28 16:20:14.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68aa714e73736d6bc7993c6289f9b62b871e2081","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-29 13:51:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"89a1a4271d75876c6ec7e20568906be09ff0969f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:52:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab4cb0287914a4ca274c19582ad790858da489a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 15:49:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c1c9690e9ef41a888af298685640891 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1bb09c285aa4e7da796af1f2070bccf : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/942c1f012da34e059f36832383c07cf0 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f18bcbed4d45fb8b0eae2da1a2268f : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ddffd9d19349ff9c77bc7ea208a41b : SUCCESS in 13m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fc9a4aed7ef04ecdb16e0970f9a8e940 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a21ab849c9c3462d8271a58b0c16c931 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f64c827aaaf748adbff5a2f3c13babfc : SUCCESS in 1h 43m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/429aab661a294d229911e72a2509aebf : SUCCESS in 1h 52m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aaae58dc324e4262b48e3d8fae9c5d00 : SUCCESS in 1h 05m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7e6ea55cbb0d4a009fca62d9e50e1662 : SUCCESS in 33m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9dba5429e7b4d15a54167bf1271e0c8 : SUCCESS in 39m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/30409d3ee10946859d72fb5308f1405f : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":1},{"id":"862dd33136256dd00b3d7ca0e5dcb68eda258640","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 15:49:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e288ed25ab8c043f8191f458c6de465ddead7701","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 15:52:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2e95aa3a6845f189c3f8fe06624e40 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5287e1b08c8ffc16bcfcb7d878dc6a652ff2834","revisions":{"c5287e1b08c8ffc16bcfcb7d878dc6a652ff2834":{"kind":"REWORK","_number":1,"created":"2021-04-27 14:10:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/07/788307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/788307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/788307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/788307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/788307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/788307/1"}}},"commit":{"parents":[{"commit":"4e53676b2a4f8e97b8c2412436813d890b67672f","subject":"Fix pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e53676b2a4f8e97b8c2412436813d890b67672f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-27 08:08:36.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-27 14:09:35.000000000","tz":120},"subject":"Fix task_flow.max_workers with persistence in amphorav2","message":"Fix task_flow.max_workers with persistence in amphorav2\n\nWhen jobboard was enabled in amphorav2, the [task_flow]/max_workers\noption was unused, updating a LB with a lot of children could have\nfailed because of dozens of concurrent connections to the database.\n\nStory 2008860\nTask 42381\n\nChange-Id: I5341949513e6eb814655ee8f14a01b061894fc34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5287e1b08c8ffc16bcfcb7d878dc6a652ff2834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5287e1b08c8ffc16bcfcb7d878dc6a652ff2834"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
