)]}'
{"id":"openstack%2Foctavia~788945","triplet_id":"openstack%2Foctavia~master~Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","subject":"Optimize CountPoolChildrenForQuota task in amphorav2","status":"MERGED","created":"2021-04-30 12:54:23.000000000","updated":"2021-05-05 12:34:43.000000000","submitted":"2021-05-05 12:32:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788945","meta_rev_id":"d0af0fe769b576cc99cbd82fe7e2121f1cb61d94","_number":788945,"virtual_id_number":788945,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 12:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-05-05 10:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-05-04 22:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-05-05 10:43:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 14:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 16:39:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-04-30 19:48:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-05 10:43:39.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"a3ff79f771a0dbf19e59210865a3323de9aaf7af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-30 12:54:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"012a50117bcc2eb5f7d3d0764c7a0118b1512c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:45:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b1725e8592e41bda7571d80d07c1f08 : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d30ea33e820e4f30bdd994d52f87f215 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6a3304205147f4b4bb74562c49dc98 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82714fe32a0a44728a39d14b0867defb : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1df1d5743def47df86351cc5b948130c : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0150cafe9ca46dc8336e180ce2a2474 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65fce21dbec744d884b8f48228d9d6c0 : SUCCESS in 13m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b7c9c16af3d2425aaae83e29204d1fd7 : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a535258746a34dc988412f3e71c7b9da : SUCCESS in 8m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/647d7d119fae43f6adf79feea8b5884a : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27be2bafcb5c4b5fb70705d94ca4539f : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e071de9ecf5a4ae8b1664c6bf37daa1d : SUCCESS in 1h 34m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6429db1de0694d838790557ea773adf3 : SUCCESS in 1h 41m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/594d73352dbc423ba0f486385fb12d8d : SUCCESS in 33m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49c0e4c42963452c8030738f68bea0ef : SUCCESS in 38m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2882bcc25f4452e98b7eb31459e161e : SUCCESS in 53m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/55b300512eca48839c07a3465caeb206 : SUCCESS in 1h 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3e40cf0542749a79ff0bb0eb53ee5be : FAILURE in 40m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e49a7f42fddf477288cd14d562fd91fc : SUCCESS in 1h 48m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4681db1a642c4f848744eacb5d46286d : SUCCESS in 1h 42m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/23a5e2ea46d140b5ba0b564ccf9aa356 : SUCCESS in 1h 41m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4e3d9716ed1542f099fd81a4a3a09040 : FAILURE in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ac30d6b1a390275a5ab5e5eff430bf34eaf266c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-30 16:39:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI think we can do this better.","accounts_in_message":[],"_revision_number":1},{"id":"ae55accd741bc3390242331394af1e8e7ce6b5db","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-30 16:43:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1a0811431f31bd2e01a23a3ef781f98687c90169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-03 06:57:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1bb6bbe462dbb0914177ab99252277e52e04c4e6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-03 07:01:37.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"523a2cd41a554eeebc620b571dcfac48e53c36f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 09:23:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/955eaeb9e70041bcafd751bf032689e1 : SUCCESS in 10m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39eeda45a87c44d49195d6765d3e704d : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb275c091754656b74e9833543e73cb : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6707870a513d49a9bb20728e1f52da1c : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89575896e3384aa69c601d3f7e939418 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91f56d5b2af0437c88f52f8b54d39dd9 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60cae16c6674a529c2e804c2ec095e6 : SUCCESS in 11m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9cc5ce99da194058bc5cb7da03df2e53 : SUCCESS in 6m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/048e2a41d4124be5a8c0ec3826a8e05b : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/484acaebe4fa48d9b029cc414f9ae029 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6666fa7816046fc859dc709a83338ea : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1eb6c67c5d4d4a9bbe1b0778322635e2 : SUCCESS in 1h 38m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77775306fb0248689de992cf3b0292d9 : SUCCESS in 1h 39m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/225f8d9111bb4a29bbad89fd31b4d635 : SUCCESS in 42m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7dd5ec01236f49c485daf58fe0321a12 : SUCCESS in 39m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c689b5dac3447a68df50a135df88e69 : SUCCESS in 57m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c633566fb1844aab5e928fe8dfa883b : SUCCESS in 1h 32m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7535f188a3754dfb907aae1097937a8c : FAILURE in 46m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dec836773cb642aaa6ec00663731584e : SUCCESS in 1h 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/98b1da92e9934b999ce79b0c983776be : SUCCESS in 2h 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/949184da585041168e83ba48110eb835 : SUCCESS in 1h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/467d31872c4149479c45adc28982eadf : FAILURE in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f53dd167f45d420f184a19d6495d05da9d20c5fb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-04 22:08:45.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"f72e0c575dbbf33e0a6c5f8ecfb5d1ecc1359365","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-05 10:43:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a913de41a5a6d48c5c0b418dc6a4089f1ae68f75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 10:44:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7dbc015420acd833ea635d2db4a03a52664adfcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:32:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56abc377f22f4bc5bf35d62b478debd1 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd2b997f7f5417caf1af5640021306a : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c522e3f502d344779fcfbeb927712c02 : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82f47012968649d1b5b00126a78dfc39 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5b76f6d4d34febab2dea5eece3916d : SUCCESS in 12m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e8f5d9048e854abcab84110544cd34ea : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a9dd53a5a46c4a4ea15479b838d68330 : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9dc54b0c24e446ab98c21fc970cdd8ae : SUCCESS in 1h 13m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b562348851c7432e9cdbd01cab54148a : SUCCESS in 1h 45m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14ee1ff6be504d5494215246cfd779cf : SUCCESS in 40m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b7de5f5bc2a4c8a9748f51f1f53b0f6 : SUCCESS in 35m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bbddbe0e0a4f4c92aa4396f67c6fbd80 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b075f2ffc46cb04afd6e60454f5e82ceca653882","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:32:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d0af0fe769b576cc99cbd82fe7e2121f1cb61d94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:34:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891cfbba30bb4e2285c139206ec0179c : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25","revisions":{"39e7ef872c8e83a39860cfa8fd3b72e6db84134d":{"kind":"REWORK","_number":1,"created":"2021-04-30 12:54:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/45/788945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/788945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/788945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/788945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/788945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/788945/1"}}},"commit":{"parents":[{"commit":"c55e68a3bef94b481e23c2530b79826092a88db9","subject":"Merge \"Disable spare pool jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55e68a3bef94b481e23c2530b79826092a88db9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 07:21:45.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 12:53:19.000000000","tz":120},"subject":"Optimize CountPoolChildrenForQuota task in amphorav2","message":"Optimize CountPoolChildrenForQuota task in amphorav2\n\nThe CountPoolChildrenForQuota task takes 3 or 4 seconds to execute in\namphorav2. When cascade deleting a load balancer with many pools (60),\nit might take minutes.\n\nWe don\u0027t need to get and to convert all the fields from the object to\ncompute the number of children of a pool, we can optimize this task by\ncreating a new method in the PoolRepository class.\n\nStory 2008873\nTask 42414\n\nChange-Id: Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39e7ef872c8e83a39860cfa8fd3b72e6db84134d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39e7ef872c8e83a39860cfa8fd3b72e6db84134d"}]},"branch":"refs/heads/master"},"7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25":{"kind":"REWORK","_number":2,"created":"2021-05-03 06:57:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/45/788945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/788945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/788945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/788945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/788945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/788945/2"}}},"commit":{"parents":[{"commit":"c55e68a3bef94b481e23c2530b79826092a88db9","subject":"Merge \"Disable spare pool jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55e68a3bef94b481e23c2530b79826092a88db9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 07:21:45.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-01 15:00:45.000000000","tz":120},"subject":"Optimize CountPoolChildrenForQuota task in amphorav2","message":"Optimize CountPoolChildrenForQuota task in amphorav2\n\nThe CountPoolChildrenForQuota task takes 3 or 4 seconds to execute in\namphorav2. When cascade deleting a load balancer with many pools (60),\nit might take minutes.\n\nWe don\u0027t need to get and to convert all the fields from the object to\ncompute the number of children of a pool, we can optimize this task by\ncreating a new method in the PoolRepository class.\n\nStory 2008873\nTask 42414\n\nChange-Id: Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
