)]}'
{"id":"openstack%2Foctavia~790909","triplet_id":"openstack%2Foctavia~master~I258b9b49b4a876a2bf9772b19f2100187ea83007","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2023-01-04 12:00:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I258b9b49b4a876a2bf9772b19f2100187ea83007","subject":"Remove i386 amphora build support","status":"MERGED","created":"2021-05-12 07:30:59.000000000","updated":"2023-01-04 12:01:50.000000000","submitted":"2023-01-04 12:00:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790909","meta_rev_id":"4edd9dde8217e8be50d58ce5c6888fb9732a7ca5","_number":790909,"virtual_id_number":790909,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-04 12:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-06-07 17:05:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-06-04 22:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-29 09:36:40.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-29 09:36:40.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2021-05-12 11:15:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 09:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 13:40:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-06-04 22:54:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-12-29 09:36:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"643bdc8e749648060e95155c2a9dbad392e7ff1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-12 07:30:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21025e7e9c975e379a96e351fa05b85de3cdab8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:54:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d9bab94ad5247a29b9ff237a6e69528 : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e93b70540f4439caa12cabb7152937c : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cd1c936d57047438a57dbbff23b429e : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f8f8fff27f4bea990595ae5d03749e : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43d3c1fd11d949f59a1ff4ea718148bb : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16a30167723b4ea99ed9b0f23d4e7812 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db8b666177443f78fded335ab57eefb : SUCCESS in 13m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/86dd34321efd4a72bcaf119f42b287bb : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51395dbbcf994a179ae06f525df06db9 : SUCCESS in 11m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eb543fdc096c49a5bbf50eb246263776 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/156eb9449eec4a2a95903e2c9b5eb096 : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b413b058bb944f428d99124500c1a47b : SUCCESS in 1h 20m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/105ca632cae24294bfc609cbf2d4e426 : FAILURE in 2h 19m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f5f8e4147cf48e69cd5a248f7f2c346 : SUCCESS in 29m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/566190edad514082ab32ec168f0f0542 : SUCCESS in 40m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d287c540d3484250afc0d38cceeebe99 : SUCCESS in 1h 10m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7d3717173df645f1b13783e5de98f3af : SUCCESS in 1h 36m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1659c577b8304af78a7375128a5066d8 : FAILURE in 41m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5a09996a594849d0b9b3827410599e3b : SUCCESS in 1h 51m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/91ba9524602342b78ed8ce8c40d6119d : SUCCESS in 2h 20m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e819d6bbe98745cb922410188f7d305f : SUCCESS in 1h 51m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9e26ea5e74d0404d8ddfc33e23e3793e : FAILURE in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c080da0256541ce32e2b0c9d78edbaba72f136a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-12 11:15:12.000000000","message":"Patch Set 1: Backport-Candidate-2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0046cd94ddda604986ed26509497e043a30725f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 12:59:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69ac1933b57d4daa9bde64c011649da5 : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5c15f36bcb4f2ab94a68b0f3850cd5 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db63461afbf49d2a3baa2a63168dc2a : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a2657d249ff4949a641477d8526bb60 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/804bbf52fe5442288887f578891a7879 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe06a872ff1d4838a23e743758d166fe : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee582257d4446d4ac011ded8200facf : SUCCESS in 13m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/27b511e90ec44cd58f231990dabf5183 : SUCCESS in 6m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5ffb1a98df7a4aebbaac8e5bdc9e5919 : SUCCESS in 8m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d55e4c95fc064836a0a9ee530e627cc3 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/14eaeabc491840a79e3580e5b4341c95 : SUCCESS in 10m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/669a919c463a45f2985db6e840ba5155 : POST_FAILURE in 1h 42m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6322c4e14ce44e8a8cc8452bdfcf329 : SUCCESS in 1h 35m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c3a7fc505174edb9499b1730eca5ef1 : SUCCESS in 55m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fa92f65c62843ff88eb39cb095b5801 : SUCCESS in 33m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af3650d142ec464e9991bd42c54e9f0f : POST_FAILURE in 56m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4c6ceaf8e8024692ad84714e60d29c02 : SUCCESS in 1h 32m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/77a6b27b3af74afe936d3fa6d2bc41ad : POST_FAILURE in 43m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a0fb8c73f7844e9b900722abcaf62fe3 : POST_FAILURE in 1h 34m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e08abe5b9acf49c2a0d98777801c664d : SUCCESS in 1h 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/46af2b056a2d445d9710914abdafd416 : POST_FAILURE in 1h 26m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/eda11e9842824d0c9122f849e9492c19 : FAILURE in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"657e504ec8af2082b47ab18fe8695260bc6de1ab","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-12 13:10:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"63cfc5dda6650cf18ca39f0e5fdc977a28a4278a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:15:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2957b58a8a0a40139d4de0c55145e6af : SUCCESS in 13m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddb0809dec1e4b35b5d0b778e5e606cd : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5aaa79b2df4a859f8440113b73e967 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c679ecf034d3424ca36fb7630cacd739 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe86db1ffd7d406cb0003ee6fa56df7b : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df6b7bbc482a49f681151ba9a565353c : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7547431d4bf54083aa3a852b2ab270b8 : SUCCESS in 13m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/166ddcb6e41e42048a0b33c9bb24a243 : SUCCESS in 6m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e552bdbe84894f2283989279b1efb15b : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cea7f33933f54d6b93f149696758f4a9 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86c7ef77fd7f4d3ca12bae759c5beab0 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a419d47edce4fb09299d3fa17153e83 : SUCCESS in 1h 54m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3593c76eb144486eb020c2c4fe879153 : SUCCESS in 1h 47m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f04a172a138e48fdb5f6723ae2d4cee1 : SUCCESS in 41m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33465815c8634354a2517e48607c12f5 : SUCCESS in 42m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/18fce7c38bec4af1960e818c2a8bbe44 : SUCCESS in 1h 26m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8a56bc9eb594cdf9e6ab74112e6f563 : SUCCESS in 1h 31m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b07d17203c8e436eb109f0b8bc440ec6 : FAILURE in 45m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98930301f5a44e1ca7ae155a3837795a : SUCCESS in 1h 43m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eb11e0a92b194d5b9f1cb4167a989bd9 : SUCCESS in 1h 53m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cbaf19cbe42f4a8a8a8bb4d23018f9a5 : SUCCESS in 2h 00m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c089599b4f764168810253f08a9f3663 : FAILURE in 28m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22a45b341799f49d1c56922f05c6ca3b7176e4d1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:54:38.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM\n\nI have been working on this project for a while now.... lol","accounts_in_message":[],"_revision_number":1},{"id":"206d9a629d09492ee9674a383e65234523ba0e72","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-07 16:34:59.000000000","message":"Patch Set 1: Code-Review-1\n\nThere is also an i386 string in doc/source/admin/amphora-image-build.rst since it has example diskimage-create.sh output.","accounts_in_message":[],"_revision_number":1},{"id":"428d84437557ac9cb2e952380ddc2e97b1dd01c0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-07 16:52:09.000000000","message":"Patch Set 1:\n\nWhere? I don\u0027t see it.\n\nhttps://c1bb961e4e3d587ff63e-cf5939cc3c93cbe14b6358ccf9ca6414.ssl.cf5.rackcdn.com/790909/1/check/openstack-tox-docs/7547431/docs/admin/amphora-image-build.html","accounts_in_message":[],"_revision_number":1},{"id":"6b74c4a1ae56c499119fad37a3ba541577465d13","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-07 17:05:20.000000000","message":"Patch Set 1: Code-Review+1\n\nStupid me, it\u0027s a symlink:\n\n--\u003e ls -l doc/source/admin/amphora-image-build.rst\nlrwxrwxrwx 1 bhaley bhaley 36 Jan  7  2020 doc/source/admin/amphora-image-build.rst -\u003e ../../../diskimage-create/README.rst","accounts_in_message":[],"_revision_number":1},{"id":"24c0e100f21e22358282c86a5b88753a45dc26a9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-29 09:36:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9d848edf455dc0d3afae1b53e1edaebb7e26cea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 09:37:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b2efd99e63c5672602e5df4d2389b84c2abde91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 09:41:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4151ad7cd7b4e33ba79ecc9cd06c9f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639445139a2a427fb33103013c7b7c05 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e188a8b2660348bd93295887f9ce8350 : FAILURE in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbb563c1f33b45d8a4c56a668d508d75 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/361f769c13784ff3853c898eea25dec6 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f4d4228c94f45a897a1f76a1c025e15 : SUCCESS in 3m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/99b157c1f2b3476d95eef00e6c0cd958 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00802a4fed864da69a863f74077bfcdd : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd42d52c2b4c4dd6bfdf2863cda1d9cb : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6824150399604aeb98002d4c2908cb12 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b45641bc0674f11b5f26f03298ad7a5 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"32c130c38cfd8b954dadd5571771ff3fecc5983a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 07:44:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daa2f433000901d42e7369fec9f949fc0cfa3026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 10:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34f50e9ea79740129184f3e8cb07222e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27f05c389c384c9eb950ae90d3e85ce7 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c65598e5c14a40acdb97c1a2f40d99 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782407a066a84b8686978b6ea5f3d299 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f41b267d275b477b9856abb1c5f03b54 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d422088c9c4d3fb8bcb549174a4c07 : SUCCESS in 10m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8e051f09a0b641f4a14b30f67745e212 : SUCCESS in 4m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0b8490ee9efd46d0a16d9f2a96b31649 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c38adf5499c248fda77e3db1b92c08c1 : SUCCESS in 3m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c4363b67ed64230a825e0c782d9afc6 : SUCCESS in 6m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31ab4ac1809f4bb8b8e9c9bec9bf1817 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/44732d58203a434eb7b8df2b1f9d03ea : SUCCESS in 7m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f82038d976c47878157bb9a9db257d5 : SUCCESS in 1h 19m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae59e6a12f8342a885acc27d5bc8610e : SUCCESS in 1h 52m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e58e221dc704968a4d28e3a22c1297d : SUCCESS in 34m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2942603442bf466abcfd55fd2e7285a1 : SUCCESS in 49m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7dfacb4c4a704fafbf3e2362412da16b : SUCCESS in 1h 15m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea37c066e4024f879c5dc972198962ce : SUCCESS in 31m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b66a22a0bfc48c8bac74bd19f77cc94 : SUCCESS in 1h 34m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7b6364ba78045e1912e2410d10bed6e : SUCCESS in 2h 02m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/834cf0527aff4d70a7d1cbcdf82e1d07 : SUCCESS in 2h 11m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/88f9ef809f424d68b04e03c9096b7560 : SUCCESS in 1h 45m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/81e7a20cab48437abcd2a367a737b238 : SUCCESS in 1h 44m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3405f7382ad94f5e94324a1c42a723c5 : SUCCESS in 2h 29m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae1ce6d63d487f0d902dda422ac87fa20538cc8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 10:16:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c99b97aff0b8869a101eea19300ea211a81efd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 12:00:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe925245b35e4f74966f3a63fc776eec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/775382e6b0d543dd9a5997e111fb3a34 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/053d97df9c1b4106890414641089b7ed : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d985dc429f5b413c89c4c3e1f8dcee85 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4b494980a447c7a2cfdc1b226f2447 : SUCCESS in 11m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c4db35b00e4478ca0e182f951ffa774 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d58ad259b26941c088bc1418020d0211 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/119d5f95caa649e9aa4aace1b16b9f73 : SUCCESS in 1h 37m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25275d1922874d9f8aeaf18a4720e519 : SUCCESS in 1h 43m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f28250e294744917bdd737b41d855a27 : SUCCESS in 33m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37bd41de66484fd9b37d0cca3f670568 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cc3f8a711f0b92377627e9d0934d3a32dc0f8df3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 12:00:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4edd9dde8217e8be50d58ce5c6888fb9732a7ca5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 12:01:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86b377cb9cf643a29d49dfae3bd22e4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f50a9bcfe64dbcaa4340b6312e7ade : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4333881cc0173ca2d778c7575775048657437e3c","revisions":{"4333881cc0173ca2d778c7575775048657437e3c":{"kind":"REWORK","_number":1,"created":"2021-05-12 07:30:59.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/09/790909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/790909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/790909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/790909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/790909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/790909/1"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-12 07:15:27.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-12 07:30:49.000000000","tz":120},"subject":"Remove i386 amphora build support","message":"Remove i386 amphora build support\n\nDiskimage-builder has removed support to building i386 images, and major\nLinux distributions have already end-of-lifed 32-bit host support.\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/790878\nChange-Id: I258b9b49b4a876a2bf9772b19f2100187ea83007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4333881cc0173ca2d778c7575775048657437e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4333881cc0173ca2d778c7575775048657437e3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[]}
