)]}'
{"id":"openstack%2Foctavia~791056","triplet_id":"openstack%2Foctavia~master~Ieeb08170198126e8e2b2d3c62c2e2308fe458c26","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ieeb08170198126e8e2b2d3c62c2e2308fe458c26","subject":"Fix config loading","status":"ABANDONED","created":"2021-05-12 22:32:14.000000000","updated":"2021-05-17 16:06:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7754e031e4f8fe8e6b0d6b7bac9849ec332a07cb","_number":791056,"virtual_id_number":791056,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-13 01:08:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 01:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 01:45:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"99279cb2b85390219e9335d275016c032deb0c7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-12 22:32:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0831e490da2b5b06aa7a60d3e41f1a66d7aca12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 01:08:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0c02341b12443eae6a67c7ce59bb25 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d59e82092a294d39a94c18eb03e0adc7 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b31dec7aea47b0a66f02ffbd2e352f : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3ae6c13eee4990b0d0436fc798aee9 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da8208fd823645128b1cd8fb6579a3a8 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e21614c28f7041b99681179ceb3b0bf7 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2e7b208ed343ebadf71fcf1a48c6d5 : SUCCESS in 10m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e0819fcc93904feabc900bbff25708af : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1464b3142a194f9396308766981abd61 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d0a1f847da14f6f81655b59883b4457 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b025f86afa3b44ffab507517ca5102e4 : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eefa88e0fb72462597cef0d598565ef9 : SUCCESS in 1h 04m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b490090c56314f50a3b15a28ca89c0e8 : FAILURE in 2h 33m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f0ead2aabde4dca871f0383eba38400 : SUCCESS in 34m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2afe45770ed4dd58cf2b3d5bbb4b40b : SUCCESS in 42m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a411fcf3e18470da00935a715b62a7c : SUCCESS in 52m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fca61a2f51ab4d81ac1c7b9b5badc028 : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81126636582b43e299924cec279301ad : FAILURE in 20m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ddfa7f77affa44e59afa9596b906e5bb : FAILURE in 2h 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ee198938dba94745b16d20779c08c6cf : FAILURE in 2h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bd69448fd5934434908572e5c9b10d3f : TIMED_OUT in 2h 33m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2502d95a410545f095f969516ca91417 : FAILURE in 7m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7754e031e4f8fe8e6b0d6b7bac9849ec332a07cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-17 16:06:32.000000000","message":"Abandoned\n\nIn other testing, this seems to make no difference.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d79a2a7a16bdf5f1e915dd76e9bc2543ad1d94ab","revisions":{"d79a2a7a16bdf5f1e915dd76e9bc2543ad1d94ab":{"kind":"REWORK","_number":1,"created":"2021-05-12 22:32:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/791056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/791056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/791056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/791056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/791056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/791056/1"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-05-12 22:29:20.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-05-12 22:32:11.000000000","tz":0},"subject":"Fix config loading","message":"Fix config loading\n\nThis patch corrects a few areas in the code that was incorrectly\nimporting configuration settings. This patch will reduce the health\nmanager memory usage by 27MB for branches with the amphora v2 driver.\n\nChange-Id: Ieeb08170198126e8e2b2d3c62c2e2308fe458c26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d79a2a7a16bdf5f1e915dd76e9bc2543ad1d94ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d79a2a7a16bdf5f1e915dd76e9bc2543ad1d94ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
