)]}'
{"id":"openstack%2Foctavia~791087","triplet_id":"openstack%2Foctavia~master~I42e82cdf9dd2fccc3e69689d14a686667431a9a9","project":"openstack/octavia","branch":"master","topic":"update-bionic-jobs","hashtags":[],"change_id":"I42e82cdf9dd2fccc3e69689d14a686667431a9a9","subject":"Drop Bionic support","status":"MERGED","created":"2021-05-13 03:34:37.000000000","updated":"2021-05-21 04:29:57.000000000","submitted":"2021-05-21 04:27:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791087-update-bionic-jobs","meta_rev_id":"a7016148161468121a6b00fd9294921c360c1b1d","_number":791087,"virtual_id_number":791087,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 04:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-05-19 07:50:06.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-05-19 16:04:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-19 16:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-19 16:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 03:35:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 07:48:11.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-19 07:50:06.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-05-19 16:04:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-19 16:06:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-20 12:58:18.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"}],"messages":[{"id":"7bd29c89d2d39355a5247c1f37167f25825692a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-13 03:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3482fdd388f6ef1c00c1e11679d7374aba0ee296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 03:35:00.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  The nodeset \"octavia-single-node-ubuntu-focal\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: octavia-grenade\n      parent: grenade\n      nodeset: octavia-single-node-ubuntu-focal\n      required-projects:\n        - opendev.org/openstack/grenade\n        - opendev.org/openstack/octavia\n        - opendev.org/openstack/octavia-lib\n        - opendev.org/openstack/octavia-tempest-plugin\n        - opendev.org/openstack/python-octaviaclient\n      vars:\n        devstack_localrc:\n          DIB_LOCAL_ELEMENTS: openstack-ci-mirrors\n          LIBVIRT_TYPE: kvm\n          LIBVIRT_CPU_MODE: host-passthrough\n        devstack_local_conf:\n          test-config:\n            \"$TEMPEST_CONFIG\":\n              load_balancer:\n                check_interval: 1\n                check_timeout: 180\n        devstack_plugins:\n          neutron: https://opendev.org/openstack/neutron.git\n          octavia: https://opendev.org/openstack/octavia.git\n          octavia-tempest-plugin: https://opendev.org/openstack/octavia-tempest-plugin.git\n        devstack_services:\n          s-account: false\n          s-container: false\n          s-object: false\n          s-proxy: false\n          c-api: false\n          c-bak: false\n          c-vol: false\n          cinder: false\n          octavia: true\n          o-api: true\n          o-cw: true\n          o-hm: true\n          o-hk: true\n        tempest_plugins:\n          - octavia-tempest-plugin\n        tempest_test_regex: ^octavia_tempest_plugin.*\\[.*\\bsmoke\\b.*\\]\n        tox_envlist: all\n        zuul_copy_output:\n          \u0027/var/log/dib-build/\u0027: \u0027logs\u0027\n          \u0027/var/log/octavia-tenant-traffic.log\u0027: \u0027logs\u0027\n          \u0027/var/log/octavia-amphora.log\u0027: \u0027logs\u0027\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 31, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e079e3caa3e53dd7746e10f030ad53469431ee10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-19 07:48:11.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"29bb5dd38cec1a6e1f7195fd50714bd7e3c64190","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-19 07:50:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2b895087b67654d89287395d25ce3588320dd9a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 10:01:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61397d23090c471ba2395214fa499e2a : SUCCESS in 11m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0ae4e5281684117828a5e6215babb26 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1aa1d487a2841e092820a2dceb8af90 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/633fccdc6a034bb6a144a80b8037397d : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c18fad71dc446c9987e1c66b38a3e58 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a4a941f1da46618a7a9914224e2413 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abde98c4189c4e5db4fcbbbfaad28dfe : SUCCESS in 17m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/45a85cccfaa24bc4a55dbfc020bfe924 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/41ddb4c766fb480a8c9a23537df7c48c : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2938a8b5b445425897d5e209c1657a8e : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5d63e69ad154c5e99d1e86f8783c267 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a999ba542234794897fea70cc19e3ea : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f643d952c3f446985729224ce43429f : SUCCESS in 1h 35m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee2baa548e8c4ba9b8b33808ae6d2b5b : SUCCESS in 34m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1029e661f0f143389069d9b121295388 : SUCCESS in 1h 01m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14485bd137014ba28807916098a3f629 : SUCCESS in 58m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cb4027e29eb04e9aa3ae195bf0027f50 : SUCCESS in 1h 41m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d1dffaeaac44f3dbe462938a8eba1a4 : FAILURE in 43m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f5a0b5ac3bbd4140bc2061eda63fe786 : SUCCESS in 1h 54m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c14d5166309d453fa4540262b451c2cd : SUCCESS in 1h 53m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b50cd38725244720a13adc7b0a4a7229 : SUCCESS in 2h 08m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4f851ad98fcf45219360f53b8f41788a : FAILURE in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"931c7810b1c13f27d155a7ea6566ff2df23824d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-19 16:04:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7ff260857611b6b18b44a9b0d338815704ffd44","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-19 16:06:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"797baad4dd8d3d8bd755cfd3733b087bed531faa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:08:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8afb51faafb11ab2f1d78d02ccc9585e3a0d995d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:22:18.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f2318e0342d4d5890cae8193f6a20ca5361a6845","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-19 18:23:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6633520eae7d51695fef99f1165e1493f6276b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:24:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/046985bc879c4794be4cd5d7539830ed : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa02e67ac7dd4e878312476280d5b198 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09eee4c0c6974a95a1ed5f81ca14e326 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c6f9d2a2b0743e7bb7d3128beef6d19 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc106e3f49974b02acaf5a7ad1c104c4 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65c64099b0ad413da5f3ee056e5a1e6c : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc7fec148b2f45e19e518d280e36d4e8 : SUCCESS in 9m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/052b58304fd2423599e408f307aef608 : SUCCESS in 6m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aeb425c1644e4caeb8220714141be514 : SUCCESS in 8m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b4252ca6d22c49f0ad7bbe9786386350 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d433bbe80ad94e7eaed882a947cadad4 : SUCCESS in 14m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28deb9e494d84596ae80187f0510c02c : SUCCESS in 1h 25m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed3fda6647dd4ac8886e7d14a73ca19c : SUCCESS in 1h 59m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c769a6db9294476ab99329d2d6aa0b4d : SUCCESS in 37m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8596df100e584ab08e1b9cab85fa68ad : SUCCESS in 32m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8db4e222280540bc8bd1db664f3b7590 : SUCCESS in 1h 42m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a4a585bdaa24db3aaa3f4e8e585d880 : SUCCESS in 1h 31m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42132ebb92204d8bb23e518b26f2154f : FAILURE in 46m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7e8ba6f308e94dcbbdc15a5b3e4b02d3 : SUCCESS in 1h 40m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4b1844ff0a5845ab9067845792af1f27 : SUCCESS in 1h 49m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/eab52d882f4e493097a9c28e82f99f36 : FAILURE in 1h 21m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d7d4d1a93b4c42f0b2bd22ea0d2d62c0 : FAILURE in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c09bb76ccfc0c42ce36a1f624e709f2ebacf608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:47:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4273a0ec09ec2c7e8d3f3676bba011d104a58c48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:58:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64b0ea6da67c48fda953e63cf8d110d1 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa78005fcb0433c9b7f34f286b53e13 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d141a241b681445e96157815e8b2bdb4 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe6e2fdfa724b73a3e16debe64b3fcd : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc1da9fdb0c4d99a4c3db42fd1f7e5f : CANCELED in 18m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0e492b7229fa47d4a6d10f4f793e1737 : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fef46134b6fd43778bc151ee702de59d : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f6c45a5a55a46d19b06428dac3cac88 : CANCELED in 18m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb3ba5ffa1a041df91c2ae1ed27e0a2f : FAILURE in 18m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/148a9dfe89da44859a2f007e6c4ffa3f : CANCELED in 15m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4d0b4afa2cc49c99d777c3aea03b0b0 : CANCELED in 18m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/566be92ecfd444b9bc1420a900023ed2 : CANCELED in 18m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b25c3d821f0a540f58bb32d1297b18285bbee697","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-20 21:03:37.000000000","message":"Patch Set 2:\n\nbump, dependent patch merged","accounts_in_message":[],"_revision_number":2},{"id":"9c7374c98ca166fc7edfa9c1c072cfe2a1b90ee3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-20 21:03:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4f2f565ea0632fd8458fa408d5846069a82d3da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 23:19:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a421c959758c44f8a0359361c57311f2 : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf87c994ab834cf5a9cdfad87ef2a569 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecbd75310bec473184fd94c4439d689b : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f39cf1b40a404c3086eec6b011cf1eee : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5c9a1efdda34ddd9c3f865b609f659b : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b49c8ef7693406a99a2a1f05dfd9d1e : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315259b51cb34b6c9243babd590430e0 : SUCCESS in 10m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2e8c4c511dcb4cdcac88deeeb843b830 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f5fa9a8f205f47c2a0e0fd6fcbdaf693 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6d95de6dadbc4cbc93451b3d2a1f7f2c : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ced513d6dc040b7add4b8eed3ff9777 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ec4f2b838bd4f8786c4bf762ec47685 : SUCCESS in 1h 05m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/139c46617baa4371b41ad503cb9326bf : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b2a171265c4458b9b2c509f010b369f : SUCCESS in 36m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0995f955db24381baf27cf41b2d50a7 : SUCCESS in 30m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1eb2ba0946be4b1fbe12412348600a01 : SUCCESS in 48m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf2b6e8334a843f69c8400b77139702c : SUCCESS in 1h 38m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de6249444b354deb9c497161dcb227bb : FAILURE in 57m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9e2fca1d00314a30aa8fe099f6b22427 : FAILURE in 2h 13m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3bc6650395947df8b03b274323ef77f : SUCCESS in 1h 46m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7a34991b4ec048a9a7e5fc3e617f73a5 : FAILURE in 1h 46m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/45e2d70530a2457394b9eb38e2e2c583 : FAILURE in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"990f64d3bcd5de88682864022de20a018f98a095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 23:19:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"279d5cba76a54ed6864258d290727ae1c57df5a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 23:47:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b8c57f063c4ee7a5c7e3e55d1c2456 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d087e26a2648099c4c00823f970b53 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4072e4a08f34301a989974bc97bae5e : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bfdca113a8b4484b495b9c2b686f909 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b41896db75d4dda93b038fa2fbc6421 : SUCCESS in 9m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d94334affe414656a402617124b3397d : FAILURE in 27m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0bd315fce15445409de07c5b94b80c2b : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc8a072a8a034896808673289da0ebe4 : CANCELED in 27m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d93d0608efb43c4be7026415601fa8a : CANCELED in 27m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c757d4f7681047ba90c0ef8699c81cc7 : CANCELED in 27m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b9017eb6fcc4c2cb4e7d965b2246623 : CANCELED in 27m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd3d8cfb20704e298ebed57b3f681d41 : CANCELED in 27m 10s","accounts_in_message":[],"_revision_number":2},{"id":"9eff40daa83ad21b9c07621f170a4dad55efd4e5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-21 00:39:20.000000000","message":"Patch Set 2:\n\nrecheck\nnodepool host networking issue","accounts_in_message":[],"_revision_number":2},{"id":"07aa7cc7cb1bcc355e67e533c9046795173ada00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 02:27:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd287591c1c24a4b96788683272ea7f9 : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d66041eefb7a4045929212bdd06fa5b5 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b148523df842aab5166d5f1cd87276 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aecb00e0136b40c88f0ce7ec35c5f48b : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8d9b29342af4c1a965b0a3c6ea27be4 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4727f1f8aae7476cb3baa6cb365e6b6d : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/362c2cc709ca4dcfa8394a975ca2d185 : SUCCESS in 11m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a57fb5b603ce4d6babfaf47835896f59 : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2425da7e4ef144e7ab60d6966d704a35 : SUCCESS in 8m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23ab052428594a14a8d50854636ed788 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26298254b5f84284adbe8cb935e28595 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cfbdb3bf60524a2a96e903d83015130b : SUCCESS in 1h 24m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21a34dba423843debbebcee9ad4f8463 : SUCCESS in 1h 26m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cfd339d2ea434cde924acfdbb06b5118 : SUCCESS in 37m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0a6a797d03149e59633647839a6ea13 : SUCCESS in 42m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9af3a60d713d4bddbdd11ea3d3b79f38 : SUCCESS in 1h 09m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bb2560a88f48469cac093734db14614c : SUCCESS in 1h 19m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bdb9a884531f4b9cb0259f96d633fb2a : FAILURE in 42m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c60195204b074e18a9fe6d3944ad9a4a : SUCCESS in 1h 45m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64cd806316c64fb1b3a28ce394b3778a : SUCCESS in 1h 39m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ec2aec2c80ac4d2bb11375468a086ab4 : SUCCESS in 1h 22m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/970a9fc7fdb7402eb50fab74edc31087 : FAILURE in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfac5a068e67cd269a0ab884ca35a7b5035e6e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 02:27:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92491d08565959bf43e8ddfc7cc41247e80e9170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 04:27:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fa716efb90b4c7d9a1720f0e3c46c42 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14a454177e84e7da7ad876bea91364a : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4cd323910d94a6d8f8475196513d49a : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a55268b7f36e40d586f9ab07020c00e9 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d626a2ec8483471b86c8f2756d43c843 : SUCCESS in 9m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/777859b8197c45a092509966218d1060 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aef281fdf9b6464f9ef1a5243e3244a9 : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd9d079e095742bebd5d35d058e50196 : SUCCESS in 1h 08m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/291e88b46a324ab7a19adadd49145ed3 : SUCCESS in 1h 57m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/270b6e4a3eb24133856bcfb974c73aec : SUCCESS in 35m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32a899316c65487397c9a1512dec5ac8 : SUCCESS in 37m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d1838e2edc444dd4a7ae704689bc9426 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":2},{"id":"48a2352b4504b6c746129595a5d5cb876c47f4df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 04:27:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a7016148161468121a6b00fd9294921c360c1b1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 04:29:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d8aca1632344968c3f11ec95c1b49f : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b","revisions":{"2eb7cbbd5c6c52a3b396c7ea5c6b2391a9c566c1":{"kind":"REWORK","_number":1,"created":"2021-05-13 03:34:37.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/87/791087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/791087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/791087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/791087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/791087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/791087/1"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-13 03:32:52.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-13 03:32:52.000000000","tz":480},"subject":"Drop Bionic support","message":"Drop Bionic support\n\nSince victoria cycle, we have moved upstream testing to\nUbuntu Focal (20.04) and so does no Bionic distro in\nXena cycle testing runtime[1]. Grenade jobs also started\nrunning on Focal since victoria was released.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022248.html\n\nChange-Id: I42e82cdf9dd2fccc3e69689d14a686667431a9a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eb7cbbd5c6c52a3b396c7ea5c6b2391a9c566c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eb7cbbd5c6c52a3b396c7ea5c6b2391a9c566c1"}]},"branch":"refs/heads/master"},"0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-19 07:48:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/791087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/791087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/791087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/791087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/791087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/791087/2"}}},"commit":{"parents":[{"commit":"16e83eef0154a30ef91fb09842bf009c540f1c66","subject":"Merge \"Add nightly test image build job for focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e83eef0154a30ef91fb09842bf009c540f1c66"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-13 03:32:52.000000000","tz":480},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-05-19 07:47:26.000000000","tz":120},"subject":"Drop Bionic support","message":"Drop Bionic support\n\nSince victoria cycle, we have moved upstream testing to\nUbuntu Focal (20.04) and so does no Bionic distro in\nXena cycle testing runtime[1]. Grenade jobs also started\nrunning on Focal since victoria was released.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022248.html\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/788954\n\nChange-Id: I42e82cdf9dd2fccc3e69689d14a686667431a9a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
