)]}'
{"id":"openstack%2Foctavia~791643","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ic97a991437144f3a220d9b96839cec5b63565f8c","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"Ic97a991437144f3a220d9b96839cec5b63565f8c","subject":"Make /healthcheck cache results","status":"MERGED","created":"2021-05-17 07:04:21.000000000","updated":"2021-06-05 00:04:54.000000000","submitted":"2021-06-05 00:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791643","meta_rev_id":"344a8c15ac72f03c08ace7c2419339615fe3a60d","_number":791643,"virtual_id_number":791643,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-05 00:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-05-31 07:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-05-18 15:08:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-06-04 22:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-06-04 22:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:04:21.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-17 09:01:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 15:08:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-31 07:56:47.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"3e6aa31afb315a1566720579ef3664b77bac198e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80abca8f0292d540b36ce6df03b7fb1a830ed8ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/798dc31631924eeca8c9ad3cacad3b7b : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6bb456fa1b74c20ba78a77ef57cb25b : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17de664861934c009e62a104bbe68aa2 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71d2cf557b4f4b7d9a747730fb565ecd : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4272189f5de94d2aad173b23780b265a : SUCCESS in 19m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f95ba33bb1424f49901b02ffefd11685 : SUCCESS in 5m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/350351e2dac94b7581f649db412611b9 : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c9c1e4ddfcfe44269a349a833dfb33fb : SUCCESS in 8m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/196eb2e821404e5d95cd2d93105eaae4 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17c2227b11fc47c684d952e4fde149d2 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53e3fee77e864d05a801b32d3f4ea636 : SUCCESS in 1h 15m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e612361af624165b15da2b5b7b19005 : SUCCESS in 1h 33m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8511ad85366485587d6ac1cd1340f3a : SUCCESS in 37m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c0f78a8994ab4584966511c640830e95 : SUCCESS in 36m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86fd88b73b374316a266b833e8dba721 : SUCCESS in 37m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/424a4f8b1eea4128a0e13331c4c0197c : SUCCESS in 58m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/68f7391ba4d34dbf83af8cdc3ffba3ec : SUCCESS in 1h 44m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ec7eb914bc0469d87721dbd4598f09d : FAILURE in 1h 54m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98ee3de2f1ce45f1be3370208d32c7fe : SUCCESS in 1h 31m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0776d453880e4b27ba66fbecad3d1781 : SUCCESS in 1h 32m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aa8224c50e56471fb9ac7066e5a8c326 : FAILURE in 1h 48m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/027b91a644444cc29624302c2564d0ad : FAILURE in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22f23fe00d7c347470732ab33acf952f6392698d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-18 15:08:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b5ea7b0b6e4f7e3b5a80ea72e9a725c89cbabbb6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-31 07:56:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a50b487dd3f4b8f31ad22548c50ce71dbdecbf7a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:29:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"f4e0e6204c319daf846778fba31851fdc9768ed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:29:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1589f95a9e6c4f6ae996a10b15dfd985e06bc37f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:03:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8083e94578ae4c0e93f703f5c5bb8b82 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f168493dd15a4a2b87ea5b65381b29d7 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ddd314d19f4df69579f9831372e6e8 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e499281b60a47af930703ef5182a123 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3eef5e99c49448c9d035e8697a4d247 : SUCCESS in 5m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/008bab9d23484d6f9854df232145eff9 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/584d8db044ae4566bf88cba2f20d7bdd : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4fe99bbddbd413bb0324b0b37a331f9 : SUCCESS in 1h 32m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/818a48fb6a6549d3b8881036ec8854d1 : SUCCESS in 1h 27m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed72b90cf6de4c1f9015c695e8b93735 : SUCCESS in 33m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/77c8bdf5e86e4ad3939ae5cef7f83b5e : SUCCESS in 32m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4bc4c96c6df48baaf2135f7cceaf1d0 : SUCCESS in 34m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44b2ef2a259c418b9efbd0eb875b36b6 : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ca3f94a4a760d15ab5b848450025c7e6e518a76a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:03:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"344a8c15ac72f03c08ace7c2419339615fe3a60d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:04:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa278fd8e2f044f88b91bcbeed2d6581 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6fb323b8ea04fcbabe5f542ef3019f9 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"582b4bab0ed0362a0e795590c806d930414acfba","revisions":{"582b4bab0ed0362a0e795590c806d930414acfba":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:04:21.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/43/791643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/791643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/791643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/791643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/791643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/791643/1"}}},"commit":{"parents":[{"commit":"035d6d6daa86488a3a1da8a081a65ba9d21436c9","subject":"Merge \"Fix empty Batch Member Update to unlock objects\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/035d6d6daa86488a3a1da8a081a65ba9d21436c9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-27 19:32:58.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:03:54.000000000","tz":120},"subject":"Make /healthcheck cache results","message":"Make /healthcheck cache results\n\nThe healthcheck endpoint should cache results to reduce the potential load on the backend systems being tested.\nThis patch adds the caching and a configuration setting for the interval\nbetween cache refreshes.\n\nChange-Id: Ic97a991437144f3a220d9b96839cec5b63565f8c\nStory: 2008203\nTask: 40987\n(cherry picked from commit 6c54eab5b5d8fd5300cd5d62e23e6cadc1641636)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/582b4bab0ed0362a0e795590c806d930414acfba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/582b4bab0ed0362a0e795590c806d930414acfba"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
