)]}'
{"id":"openstack%2Foctavia~791644","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ic97a991437144f3a220d9b96839cec5b63565f8c","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Ic97a991437144f3a220d9b96839cec5b63565f8c","subject":"Make /healthcheck cache results","status":"MERGED","created":"2021-05-17 07:05:51.000000000","updated":"2021-06-17 19:13:29.000000000","submitted":"2021-06-17 19:13:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791644","meta_rev_id":"6d03c94053374683c5c16c3be688ed3ea25c8486","_number":791644,"virtual_id_number":791644,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 19:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-17 19:13:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-06-17 19:13:29.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-06-17 19:13:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-06-09 07:49:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:05:51.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-17 09:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 15:10:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-31 07:56:40.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"fba1834b958148198915b745b093f4da2cbb32cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:05:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2c23df37dd715b3a43755d30b9e828d9453514f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:25:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5532ea82beda42899fe017fc620209e4 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838e4e8c442a466bb19f6e7053c94a6b : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/574f386bb85548ef977da203478f8a63 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/968ac5eff9a64b62aecaa8631fa11009 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37df3b0184d40e1b402be0b59a96514 : SUCCESS in 14m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76ccf517428c4064a52c4fe2bbd50f66 : SUCCESS in 6m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5fc5aae97d8a4569a72d1eed0dfb837b : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d870f666feb146c1ab8743e925b54c6a : SUCCESS in 9m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bdb692ff50a346069b0a5e312fcfc26c : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf1775a2488341de8dd95efcedd30f1a : SUCCESS in 1h 25m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fea6e0a336d848848154bcd86c2d2ef3 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0b239d610044fb89469a35b18f89bd9 : SUCCESS in 43m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/88a0d97493574e25b41816b31818d443 : SUCCESS in 46m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c953f4de9354eb5bf46d140d2be0f87 : SUCCESS in 37m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/492a5a654c3d40329fd7a82aae08b09f : SUCCESS in 1h 06m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/381a4d6e227c467796198c9fef26c07d : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81bb8de70bd84acf80a559cde1d0246c : FAILURE in 46m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b50d285ea608437a986c583d7313ec77 : SUCCESS in 2h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f0847185996f4c9166ab522981b9b7658921e4b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-18 15:10:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c5da5eba6a4cd3857e3601a0a1713b99c5ced413","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-31 07:56:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66af55a0991549ea58a42755fbb33d9a8a874d65","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:29:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e8e9d33d6c80cdd98ab0e1f3a87f6fa513771958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:29:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2226b1ca944d9c1ec4ebc4ed43dc9be0b993d442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:03:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10927e1db1f442b8db7ff776387b2bc : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c1feb36e852445282e4e9c59c552117 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e104ea9f6f14c669fd5a9819558d920 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce93a1c521e542189c7823de62973b72 : CANCELED in 5m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60a62f2cb3c54ca1ab8196a2c7689b2a : CANCELED in 4m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b98fd07f05f94686afa1396c356709f2 : CANCELED in 6m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/33b9831d45e249abb24f7bb9ea9e9a0f : FAILURE in 6m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68c727f294014b5c8045391593123e69 : CANCELED in 3m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/None : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"127558d91ed1fe94c4e9ede47e2163f8652ac56d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 14:48:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"067ca0eaadc12e07c92fe249cb3c57a76316dd28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 16:53:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/812375f87da343b697e7d2718f660ea3 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1e413f424145d996be02eeb8105b0e : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc62675db8114151890a4f8cf39fcb18 : POST_FAILURE in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5846ff99e9224246a1ec8d7eedfbe465 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbc5a75acc04c069f07400c51216c69 : SUCCESS in 16m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ce23e0b17974bb8adeb78749b9a6515 : SUCCESS in 6m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4924eecfb89f49fc86fe3c05e6491159 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8877031510ab4e73a93b9711ea051a14 : SUCCESS in 9m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4fa615d768c64fd1b360c6823ee8e572 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba97f44bbdb74a70bd48a93569e2bb52 : SUCCESS in 1h 48m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e876cd2efb96401f9896748d7983ad95 : SUCCESS in 1h 37m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0afeef1131e4486691a2a954062c2971 : SUCCESS in 1h 13m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1a6c49dbf8b748feb7b3370c5dc5cc6c : SUCCESS in 52m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e451a5e5912e49b490846afcad10d70f : SUCCESS in 41m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c77590c979f34ccfb72271ff39f416c2 : SUCCESS in 1h 09m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a461c3d2d5ba45908a0daf251bf659d3 : SUCCESS in 2h 00m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aaaea650097f4349aa2fc21690e038e2 : RETRY_LIMIT in 8m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8e54174672f942209b724a3dd93a54a2 : SUCCESS in 1h 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"27a23ea82c5d622d478dbf7eabbfa9f636c20f3f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 17:48:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5fe656c8a5f84fa9c74338b0b276f68ab229d9fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 19:28:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b282c51affb7468d9b8a6fa0fc082a41 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd55f619c104a5a8a3ba661a4d3aed5 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/319376ff75884fcd96b324f8562b9892 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/146537f8361b4dc7899c283b03f50c71 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b0738f7cec475cb54d37987e5ac9b8 : SUCCESS in 14m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/715635517ec041c6a57ad5578e61b27d : SUCCESS in 6m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f6c6289d193540619b4722367d600bc2 : SUCCESS in 7m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/07b829f04f8e49169c0f2a7807a9a31d : SUCCESS in 9m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/015615553d294d70b11eef20cad30ad7 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/580aa8ddfe4d452fb85686e3f6104aae : SUCCESS in 1h 25m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e884965968f4d22ab664dda025d7a4d : SUCCESS in 1h 29m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a8622e3ff606416ba64cdbda913737ef : SUCCESS in 34m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be6b4b5239f44141bd95c68af8f2bf44 : SUCCESS in 34m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd974fc716ea455080e51818a360e8a9 : SUCCESS in 38m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7836ad86695244b1a0cd00d3cbdfe42e : SUCCESS in 59m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/47c25a2ffd334cb6815f85cebfd4474d : SUCCESS in 1h 28m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2458f130608447ea82c1e06bad720c90 : RETRY_LIMIT in 7m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e0758ea1aa54306b87db9040333c8ce : SUCCESS in 1h 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a42ff12befc8e0e261ea7744926880b08e4623f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-09 07:49:34.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"44ac35b5b312395e5304595906048d666f09dba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 17:28:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b782cecdbdd0a0ac48f01d2c04d976f348e99487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:13:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef0e11d36694d17a4cb9d9215d402cc : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e91c6aca58246cb8a9d9026afee94b5 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1806e6039d2f483db90b60d7ba2f2b97 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acb8bd4f69f4b7eaa491098afaf3aee : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7743da691b3441688a68de033b699587 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/723f2bde64984a4bbaa17de0efb77127 : SUCCESS in 8m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af88497a4a8741ab85e881efffb20a61 : SUCCESS in 1h 07m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7628df6a4c048aa9682ec9409b0a7ae : SUCCESS in 1h 30m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6e9c989f24ef4e80912213900801c4eb : SUCCESS in 40m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4a2d8c26f4ad4814a663293859a6d2d7 : SUCCESS in 37m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14999adaad404945827ec65851fe7853 : SUCCESS in 42m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ef4e414f8434deb8ecf7031314db830 : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":2},{"id":"713d623193036c13068ac441a2fe15965c50b61e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:13:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"170be0dd53cd342a42a9816146ebcea1d80863a3","revisions":{"577b4cfafe01546e5f96b2dc576c2dd9db25fc8a":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:05:51.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/791644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/791644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/791644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/791644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/791644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/791644/1"}}},"commit":{"parents":[{"commit":"46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf","subject":"Cap hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-27 19:32:58.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:05:39.000000000","tz":120},"subject":"Make /healthcheck cache results","message":"Make /healthcheck cache results\n\nThe healthcheck endpoint should cache results to reduce the potential load on the backend systems being tested.\nThis patch adds the caching and a configuration setting for the interval\nbetween cache refreshes.\n\nChange-Id: Ic97a991437144f3a220d9b96839cec5b63565f8c\nStory: 2008203\nTask: 40987\n(cherry picked from commit 6c54eab5b5d8fd5300cd5d62e23e6cadc1641636)\n(cherry picked from commit 582b4bab0ed0362a0e795590c806d930414acfba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/577b4cfafe01546e5f96b2dc576c2dd9db25fc8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/577b4cfafe01546e5f96b2dc576c2dd9db25fc8a"}]},"branch":"refs/heads/stable/ussuri"},"170be0dd53cd342a42a9816146ebcea1d80863a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-08 14:48:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/791644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/791644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/791644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/791644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/791644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/791644/2"}}},"commit":{"parents":[{"commit":"5b6d25ca57bc383e4a110da84b8c325eb5ec20e2","subject":"Explicitely set nodeset to Bionic-based","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b6d25ca57bc383e4a110da84b8c325eb5ec20e2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-27 19:32:58.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-06-08 14:46:35.000000000","tz":120},"subject":"Make /healthcheck cache results","message":"Make /healthcheck cache results\n\nThe healthcheck endpoint should cache results to reduce the potential load on the backend systems being tested.\nThis patch adds the caching and a configuration setting for the interval\nbetween cache refreshes.\n\nChange-Id: Ic97a991437144f3a220d9b96839cec5b63565f8c\nStory: 2008203\nTask: 40987\n(cherry picked from commit 6c54eab5b5d8fd5300cd5d62e23e6cadc1641636)\n(cherry picked from commit 582b4bab0ed0362a0e795590c806d930414acfba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/170be0dd53cd342a42a9816146ebcea1d80863a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/170be0dd53cd342a42a9816146ebcea1d80863a3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
