)]}'
{"id":"openstack%2Foctavia~791647","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I39391070cea170a6039f901093f09fc89ba06123","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I39391070cea170a6039f901093f09fc89ba06123","subject":"Fix using subnets with host_routes in amphorav2 driver","status":"MERGED","created":"2021-05-17 07:08:14.000000000","updated":"2021-06-17 19:16:47.000000000","submitted":"2021-06-17 19:13:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791647","meta_rev_id":"05070bca314c6859c36751b00f2141758c2d23ae","_number":791647,"virtual_id_number":791647,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 19:13:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-17 19:16:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-06-17 19:16:47.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-17 19:16:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-06-09 09:31:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-06-09 08:10:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:08:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 09:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 15:12:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-31 07:54:16.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-06-04 22:32:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-09 09:31:38.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"78d8bb1d03fa01dcc27bf6a08fce6cbf57a9c8dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:08:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9633866dc9bd1763af155b12b117db3bd4453133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:19:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbedf40f1f0242ff9e7605e2e07a92fa : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d092777e29c484eaf4c94d7f6f448f7 : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd65cb6ad4748d49d3d2a79cdad7edd : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0337bff5402d493cb595a799a99d2927 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3cd691f612404ba72ab0eea1da87cd : SUCCESS in 21m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b711438966641b4a4c65591702c8c73 : SUCCESS in 11m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/70ce1d19c8f744b2b1e620fda4875fe0 : SUCCESS in 6m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9f355a7b309343f885dd7cd9eab93cc6 : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9718db7dc61a4a2e9dc20b9e3caac1a2 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9089644659b2489486afc190dfb972df : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/722e0fc5d8674f3db04a754c218fcf67 : SUCCESS in 2h 00m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4767b35ea88f4a2a9534681da0599f85 : SUCCESS in 40m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e68d82cdfc3a4e7fa2aaf0cc1254f36f : SUCCESS in 39m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81b9f901c8f34c74865dd6ae6d76137e : SUCCESS in 42m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a40b9ff018cf4921804e6b277bd8b3cb : SUCCESS in 1h 33m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d3f0c31e18094173b3246b6223f94fca : SUCCESS in 2h 01m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8f2b4478f9b74d82bdc18dec8e51ec63 : FAILURE in 1h 01m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94721ceea2ac473eaa0f645f70b96e14 : SUCCESS in 2h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de03bc6f7f1c77b9e538bbe61e675dcf4920f2ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-18 15:12:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"acb8f3ecd9f6ec1e09124c5deb47c186fdccc480","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-31 07:54:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a68fbe726711270c463f38593db15a60c1d4c8b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:32:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"61adca28198139c6da6be9a15c62ccfcadf44c8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ae5472e86d287b791707efb5b18c4479af83de8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:42:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae1bc19ca1d480db8a4e25818de6343 : CANCELED in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5470ff0ab3964ca8a46d9f8aa476922e : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44709d9bc6d246509afef2bc8bd79808 : CANCELED in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01cd61d17388487980d02266c92a51b1 : CANCELED in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8dea98ef9ba43f689ddbb4942297c8f : CANCELED in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0752cd6e24c84505a7a7f68b87ab6186 : CANCELED in 5m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4305598d8d6f436f8d53fc0f1bc221f5 : CANCELED in 5m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97783211d6bc4680a32d276944eaf02a : CANCELED in 6m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb0ab7ef5e3e45559494f96c909a894e : FAILURE in 6m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7c224119e877445197924204cc349c97 : CANCELED in 6m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0dd67edb6cd4be18563ecf9f672f0ac : CANCELED in 7m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/36e3f055826341fbbd854be2b64cfb62 : CANCELED in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2939259b2a4cf8d91711d8414de971b04289a262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 14:48:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"67fa5719f27977ae9eae5913a41352a744b61408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:56:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d92336fa08a144d693d8f6e5e70d98aa : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e95c1ae733b43e6a5930acd3ab03ff0 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bab858e2b0241f887295e19f1d43e71 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a0321774e4c43758c79338d0556b14d : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cd196b21fd442cb6202ac7cc38a5b1 : SUCCESS in 14m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4624018841a04c7693121b4de81b0fd1 : SUCCESS in 6m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/91e05289fc774a85a094d01b8f574a9a : SUCCESS in 7m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/174974fa9cd8465f89d362e3977f8bfd : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f2cfcfb2331420e88a9df8fa558bb35 : SUCCESS in 8m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86d8f82802cd43a9b77e524fafc41ab4 : SUCCESS in 1h 36m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfe3c89a4659466682cb43514b5a6be3 : SUCCESS in 1h 53m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6529481cd0c4bf7a7bba8df3df08c30 : SUCCESS in 48m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d681a90e9a244175bde96ba107e9e4d4 : SUCCESS in 1h 08m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43fd29bfe0f9499a99503787655ad6e0 : SUCCESS in 55m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0084ec26779403d9e1cda1257b56041 : SUCCESS in 56m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/403216f998a94d5d96c6c3d8f8439e11 : SUCCESS in 1h 30m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57825e2d62664a21abda6d80d7871b9c : RETRY_LIMIT in 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f8b42fa486314cd6bbd4f6f5203ca458 : SUCCESS in 1h 58m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b03df85208b91430c09cbf8c7ea6c73299581ab4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-09 08:10:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3576d780c73ce081f62471469852805795030fe6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-09 09:31:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0baafedd32bb760dc4871b67265b35d70b0e301e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 17:28:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eea047cb0f88cd3dcb4df5649d044a58c4208e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:13:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b75270979f4a29adb3a22e908a569f : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0344210d14a4fdeaa9d48b7f337d7c6 : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95f8d41ebd3046f9b9543ab913fa8124 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0053032a333246908a461ae7497ebcdc : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ada2d138123a46b69ce442245cbcc68c : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c419179931364273a87db918bbece5d8 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/333453418105437e8ccc196d40c2217e : SUCCESS in 1h 07m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b2333abc8af4a71ba4bd423e9f21feb : SUCCESS in 1h 30m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e34c6ef233bb4cf4bf2d49679164a1c9 : SUCCESS in 40m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c74a599feff1471bb2c5ef878ac8cf58 : SUCCESS in 36m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0ebbda1de8e475c88401776d14ef44c : SUCCESS in 50m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/161b8d6c100c4120b3722e9e2b0fc236 : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":2},{"id":"dcf540ed33b972ed868e772f42bd6b907a28a53b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:13:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6243d576202c81f85f5a2e722d32db955b411eb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:16:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e66940df1744b9bb4d660ca0f2da1db : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d781e0a2b5984b15b1ab2879d1b38249 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7b4b875d74d55ecee1494f6c3a939023df3b420","revisions":{"34ab2bc8f88d66da2afd8c713d810b2ebafdcacd":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:08:14.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/47/791647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/791647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/791647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/791647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/791647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/791647/1"}}},"commit":{"parents":[{"commit":"46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf","subject":"Cap hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 08:15:20.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:08:06.000000000","tz":120},"subject":"Fix using subnets with host_routes in amphorav2 driver","message":"Fix using subnets with host_routes in amphorav2 driver\n\nWhen using subnets with host_routes in amphorav2, the host_routes\nattribute in the Subnet data structure was not correctly converted to a\nHostRoute data structure. It triggered exceptions and failed to\nprovision the load balancer.\n\nStory 2008738\nTask 42092\n\nChange-Id: I39391070cea170a6039f901093f09fc89ba06123\n(cherry picked from commit d0ec3aaf23eba0ffc8fe7bc86b51d4857e0cc507)\n(cherry picked from commit a139d2f97c54e88c508d9ffa85a5c6268538b84e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34ab2bc8f88d66da2afd8c713d810b2ebafdcacd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34ab2bc8f88d66da2afd8c713d810b2ebafdcacd"}]},"branch":"refs/heads/stable/ussuri"},"c7b4b875d74d55ecee1494f6c3a939023df3b420":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-08 14:48:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/47/791647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/791647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/791647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/791647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/791647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/791647/2"}}},"commit":{"parents":[{"commit":"170be0dd53cd342a42a9816146ebcea1d80863a3","subject":"Make /healthcheck cache results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/170be0dd53cd342a42a9816146ebcea1d80863a3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-22 08:15:20.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-06-08 14:46:54.000000000","tz":120},"subject":"Fix using subnets with host_routes in amphorav2 driver","message":"Fix using subnets with host_routes in amphorav2 driver\n\nWhen using subnets with host_routes in amphorav2, the host_routes\nattribute in the Subnet data structure was not correctly converted to a\nHostRoute data structure. It triggered exceptions and failed to\nprovision the load balancer.\n\nStory 2008738\nTask 42092\n\nChange-Id: I39391070cea170a6039f901093f09fc89ba06123\n(cherry picked from commit d0ec3aaf23eba0ffc8fe7bc86b51d4857e0cc507)\n(cherry picked from commit a139d2f97c54e88c508d9ffa85a5c6268538b84e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7b4b875d74d55ecee1494f6c3a939023df3b420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7b4b875d74d55ecee1494f6c3a939023df3b420"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
